)]}'
{"id":"openstack%2Ftrove~884165","triplet_id":"openstack%2Ftrove~master~I08063748e15de6767b437aa443311d41e25ed578","project":"openstack/trove","branch":"master","topic":"separate_backup_image","attention_set":{},"removed_from_attention_set":{"28691":{"account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"last_update":"2023-10-08 04:43:35.000000000","reason":"Change was submitted"},"31737":{"account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"last_update":"2023-10-08 04:43:35.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2023-09-08 07:16:34.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"I08063748e15de6767b437aa443311d41e25ed578","subject":"Separate backup docker image for each database version","status":"MERGED","created":"2023-05-24 10:25:34.000000000","updated":"2023-10-08 04:44:38.000000000","submitted":"2023-10-08 04:43:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":78,"unresolved_comment_count":7,"has_review_started":true,"submission_id":"884165-separate_backup_image","meta_rev_id":"0937f3a1928278eeadaeb399fd6a448993a0e443","_number":884165,"virtual_id_number":884165,"owner":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-08 04:43:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"recommended":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"all":[{"value":1,"date":"2023-09-21 01:21:32.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-08 02:17:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-08 02:17:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}],"CC":[{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-24 10:36:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-24 11:43:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-13 12:03:35.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2023-06-13 17:13:58.000000000","updated_by":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"reviewer":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"state":"CC"}],"messages":[{"id":"2ae2cf461c71266af5695c3bd19aa380a86cc7e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-05-24 10:25:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d4e95ce4a761fecb1a43e3ede4b70fa3e8ddd19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-05-24 10:28:27.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"85370ced2ea4a3d8bcd507c1e0199e3eb04f3cf5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 10:36:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69b8440716184d4dbb05011eea4ebd3d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8f697ec051ff40e0a68b8267693e9cd4 : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/30f1c4836c1544dd810ca44ddb264261 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0bf7d0d786f948c5a02c22ab62ecd072 : SUCCESS in 6m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"47308350213ef1adabca3406768224bfe2785fd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 11:43:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73c910337ad24030ab54b58ead51b8b4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47e7b6e2f17a42eca10ff297da5a4123 : FAILURE in 6m 14s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a53434c215a4bdf9c403e3ee3d6aa3e : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/836fde326fe943b89dd319ff68f2e209 : SUCCESS in 6m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b75fad01cf74971bc7849c1906298e5 : SUCCESS in 5m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bac3737ae1764cefa999686836ea4ec8 : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dd0d5fb42e442738f55ff9922462673 : SUCCESS in 5m 24s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b9b671de0b0f4f24bacb7b23bf38ebb0 : SUCCESS in 5m 00s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/b52575ae434a42c8ada4c79377d26cc6 : SUCCESS in 1h 12m 33s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/1cb987f19c744923bf2aa94db1159188 : SUCCESS in 1h 10m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7d1d2e636c92ba78aefc1d133212a254802fcd01","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-13 12:03:35.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3b0d8e7389d57c2a900a53acd345fc76325ef395","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-06-13 17:13:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1ae5bae718e98b270070c7cc0c2b6a3bcb4ba4cf","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-06-14 03:51:50.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d4a53ded12f33b75f8c6e397d8b0f4622fcc6f72","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-14 06:30:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"14ff33992144e9f0b7f56f33e31f3a97bc3835eb","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-14 06:41:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bd412c5f185b7098c1fef0101e49d7632b8ddfb6","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-06-28 01:29:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4ee95b1942cadbcc867e008f1d6eb7b35f2e2fcf","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-06-28 01:31:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8302eaa2118430cb06bd6d15a6c0da0828493392","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-06-29 03:47:11.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"0d45f9e03b747167d0824dc2d199510b56f4c188","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 03:53:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93a51eb590e2413f95f861c6650ac768\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/97c6fd0b7ab04242acefff096a92752b : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0f064a64a48145bf96870ec43bee81f4 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e00cde11d96f4a76b230cf4240b76324 : SUCCESS in 4m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"38faa2c0737ef721f76778e0294d901247322e43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 05:33:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fb9938f20594c6a9e4e109cc0fc0f7c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c351da26bd6c431d99dd5bac26c4c0ae : FAILURE in 7m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90b8c626743a465a860cdb6643314a46 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd9c8975b8f44bbf9826c524dcea515a : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e6f90a68f0b42c6988d656a9488afa6 : SUCCESS in 6m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8d79d4069ee54b7d984608fe07ddaac2 : SUCCESS in 14m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93c0e50bf16749acab261b53d9813d9f : SUCCESS in 6m 02s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/00b77c1992bf47e08e3721a039b0070e : SUCCESS in 6m 06s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/7519dc82462548db95fcc76806796234 : FAILURE in 1h 23m 05s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/f3c57dfe641748019a31ce0d28cd680e : FAILURE in 1h 44m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a68d0263afffe10dacacf464eae15baabb2f09a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-29 06:58:46.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"e5015fcc16ae5424fdd8d30a080ed2eb17ff998e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 07:05:53.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3ebab2fc78147149c3a5294186e5f73\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6a4de761d01542f181ddf9e60eea3dc9 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d0d36637b65e4f77bdac4e718953012e : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b4727ec11b6f4b64b46530037ff859f3 : SUCCESS in 5m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"62de4f27bd55e26b16b53a02dde9bb81271d9d42","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-29 07:45:30.000000000","message":"Patch Set 4: Code-Review-1\n\n(13 comments)","accounts_in_message":[],"_revision_number":4},{"id":"e8f56fb39fa03aeb1729339c028ac245adba38f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 08:49:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ad3b26bd1824f6dbdbc7bd50c47d76e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/264107c1a5674cea8a68731e4265aa5c : FAILURE in 5m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adfe166830fe4d5f96efafd254ed08e9 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/705795a7d5774727aac8f4a70144cc51 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/83df7a61b9e04a7bb65d2230d5d55854 : SUCCESS in 5m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6a42815ae4114fb281b764b0d6a0097e : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e578771ee094917ad5626252561412c : SUCCESS in 5m 08s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/699570458c354777b3703b7e523ab6c6 : SUCCESS in 5m 18s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/5eae89d214bf45aaa038c0591fb6e4ab : FAILURE in 1h 30m 54s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/e9a02f49f14440518f3cace165a6c932 : FAILURE in 1h 50m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e497a5804d29d456bb25b9446b4160be567fbe18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-07-07 08:00:21.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"cd5cf675872994ce868115df9d10d2b5bf16aefb","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-07-07 08:02:39.000000000","message":"Patch Set 4:\n\n(12 comments)","accounts_in_message":[],"_revision_number":4},{"id":"4da432e3bc81a5723216525f61ff5c9febd0d46a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 08:07:03.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6648cd626d7041c3a1770e6d1fd43d16\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bfece879aa2949bbb6739b682c69d5d9 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c1778814105649b29bd64e434583072b : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/49de4d5f7b9746fdad201530cb124eec : SUCCESS in 5m 18s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"87f0a76ea42f345cc8efef5048f5cd2de0109c75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 09:53:39.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e57a628ee85c4f9a9135201b09b98112\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c28b0c01344745d0b35040b22c600475 : FAILURE in 7m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2719db1caabf4080902a43808a44fcc1 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e970eb4cfd964e7b870d413b2d967af3 : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f0b82ce04f2457185dcbc2b0d96f211 : SUCCESS in 5m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e1dbbe340a274c4b9bc31472e082f0e1 : SUCCESS in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/415ab3bb61dd4c65b1eb4785b257b7d1 : FAILURE in 4m 58s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/473aa3c765f94841bd90efd924c586ed : SUCCESS in 5m 07s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/1696f23e6e5147aabc6ef2b3d6a96335 : FAILURE in 1h 26m 48s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/729cb3ae8737400eb401d5ba33ba105b : FAILURE in 1h 51m 37s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"49c119e9d40e533f66d5858646a6df917dc6b956","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-07-14 08:36:12.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"3df1e05fe1ac4adcc805b7027c7aec9d7dba03b4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-14 08:43:07.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f63ce02d8e04d98906339ae5591f438\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8baf005cb0cb4e169b5cad0299d2b0ea : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1d6610068a6e48e0a0a58672679e5e8d : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/af2ea9148b8d47fe80fdd5ab5691277e : SUCCESS in 5m 14s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c1e0aa68f9dbad4b0e162534b4fc5caa99461caf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-14 10:22:37.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/630343b03d174e83b7af80769cea5dda\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d09192364e441f1ab92b9513e4a201a : FAILURE in 7m 46s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a98e9b91bf6a4e6f9ea8dbc3888cc4d1 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9eed17f2a76c48fe80befe9a4f3acd8c : SUCCESS in 6m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/680128379b714ebcac3bb6fc2b8b18f9 : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5a449b599eab422e8698fc7b96217c6d : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/418356f0602443999e6476deff50e929 : SUCCESS in 4m 58s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/32d7d7989803402383c612e34c83d57e : SUCCESS in 4m 38s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/34fa75cbf35d4bac8f1c849b9f3b6518 : FAILURE in 1h 27m 17s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/77ed5d4cdb8243f1bb90221f9ce0b807 : FAILURE in 1h 45m 19s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0c308be20aca9148e1998459fee445b6abcf4e45","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-07-21 01:36:09.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0b4ddc004f155f33e6f41f3861ef3fa43309c1f5","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-07-21 01:36:39.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a358f044e7f900de90d7471f352f5f22a40a6ace","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-07-25 01:45:53.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"af24371b73ebe7be1a4f76a13938d19e41ae080b","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-07-25 01:59:14.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e482ed5c91f2fc78ef36614e4343c513c4b67612","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-07-25 02:11:59.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d7e1e191748832007eaf0f846eb887f18ff561b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-07-27 03:32:22.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"50ad26a22dd77f456a04171345fc8ed2c60622f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 03:39:22.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f87d5a6237b2411ba26542d732b36824\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/54d2f8c824304fa78ce815feb66a6f26 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8cc046b85d09483cb5c6bf4041dae9db : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8d0103f4214a45218a55e4c339c32b85 : SUCCESS in 6m 03s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"797ea1852943ce73000c4ccbf566e411a008b8ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 05:29:55.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f1cc3388a904777b65ed9403b7e1550\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6446b02f4a7b46a29c7ad29fef842bbf : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9c782ea4d284764a2c3f75afc90a3ed : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5025a57a72b141128693c3d080784d4a : SUCCESS in 6m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4eea68a6a1c04b68a5ceda548f1dc84f : SUCCESS in 6m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2f16d5c5ca8c4fe8b2ea73468f2d3687 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b3b40677da64d3fb313386f460b23f4 : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50f5223a75ae4246a559a81140cae383 : FAILURE in 6m 18s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0e877fc033de44b4a8ae55b2a15cfbf6 : SUCCESS in 6m 17s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/bfedc22689124ae5bdeea70625d44623 : FAILURE in 1h 31m 18s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/c27acf0384e9480590832b153ae69946 : FAILURE in 1h 56m 28s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"9f8e03797dbb49fb5bc4d95ee3112f724a64b113","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-07-27 05:49:43.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"392f68b5a9d839d607f8ac17168e3c76f16cba4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-07-27 07:13:10.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"4a6eba2d3c849003acf90b360109d8099a5e9751","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-07-27 07:16:29.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"86b1c8521cf3bd9ac09f8fa4d85227b3453684e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 07:38:41.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a37a733d080d4d3ba7722d313dc3af19\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0ddee516def94a8d8416c0762be81db9 : SUCCESS in 14m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/738ff354c48c461e966ac0ccfe25a6f2 : SUCCESS in 7m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/71b832d0c77243fba8c2f03943a308e5 : SUCCESS in 15m 58s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"b3f7fb13791e23f934f385c091d0d3aa61ed1f2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 08:49:31.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4ec6acd524054452874989a8fd6eaf20\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f23180273f4246c8bf2963c4a991d6ee : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08a5adea4ef04f4e80167f46f7f4246a : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5eebb8f7a37d45c9b8580eceafb5bc92 : SUCCESS in 7m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/79f19ccb2aeb45d1a243b27c5a91a9d3 : SUCCESS in 7m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1e096923c43e4a14a1fb4581b92cfeed : SUCCESS in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86ac691be97d4cd2b560266261aa78f4 : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7892dcdb0614396bc5f5f71ed8b0b2c : FAILURE in 5m 12s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7d9f2b3349c742e6acf13007768c32fc : SUCCESS in 5m 56s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/a754e5e8cfd646c99b409a84639ce8a1 : FAILURE in 1h 31m 57s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/a1b7bba2f2c54a729efed53e3971b595 : FAILURE in 1h 29m 21s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"572d83fcc4375365c994197b7731abc5ceb67b20","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-07-27 09:04:06.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"1957d704ddc9fe950a4bc5ed3d8f6f2b37fe52b3","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-07-27 09:05:51.000000000","message":"Patch Set 9: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"d7568368227a82d5dd193758e6e477fb45d22947","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-07-27 09:09:57.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"feeb9326c20ba68f48a89ed1789064769e762a5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-07-27 09:16:59.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"56a9b6f4eb79b85dc0b888ffe1f812ea90726865","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 09:23:45.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a765c2fa0e1943c3acf3ff3078cbf206\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d83e37dab96f4f7ab11844398f884d9c : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7a4346b09cb14f0bba471b3742661088 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/55a23acfe8df4cb5a431cb49fb12a1da : SUCCESS in 4m 46s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"e64d1260605f60f679a292bf7a04ed6bf75099e9","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-07-27 09:49:32.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"1aff05cd9c291a17b3ac9df11b1f2ca509bbf749","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-07-27 10:06:01.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"2925966cea0d93a47ea4cf108f1fe0292c214725","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-07-27 10:07:25.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"0dcdd2ff5cd4a576475e528a08e8661eed66e651","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 11:39:54.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d15e2e05e1bf4d3c975e5234cde74220\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4064d5bde6a74391aff940bded7ba46d : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd7e010d8e9e461cb94efa5a5ab5736e : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46600a77f62a4bf7a105acca0e235810 : SUCCESS in 7m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7e1769183d948b6a0c2b0c773f68360 : SUCCESS in 5m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6d571ccf97464fc5a1e5036ca4893b73 : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dacc7ff493404cd7855fe3b7a72f4da0 : SUCCESS in 5m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af998e94dde447a9a4712a5317207c25 : FAILURE in 5m 01s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/026170eb4cc0484f99bc04c37d838c5f : SUCCESS in 4m 52s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/85921e2a2967463a931d61e2475472b5 : FAILURE in 2h 21m 28s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/efb76d8988aa4d14acf7efe2f0efd698 : SUCCESS in 2h 10m 55s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"7f8d33573f523d0f60cf29d53d6887149cfe6c22","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-07-28 01:29:16.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"1b7d3055b31eae1b122f3079cb5bc0dab0de38e5","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-07-28 01:51:05.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"ca36f6b65510bd8a9dc6145129ed579df1983d35","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-07-28 01:56:29.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"2c003ab3ad05006fb0cc6987e91f5c764c4c6c0a","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-07-28 02:15:40.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"92a9973b4b13f756b789d49aa469edddb7942dc6","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-07-28 05:44:55.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"2840da7745adf979f2a14612e0e38a6ac9ca21f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-07-28 05:50:55.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"d60b176852775e95ed55b580d9ff2c9e383f84f8","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-07-28 05:57:14.000000000","message":"Patch Set 11: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"610e6605f79e0ef21432c8a2f9706cd6ad2d15d3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-28 06:10:52.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e4eff686b4f46648d03688e3f3d556d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/39f74374414b4367a1ea87a63c9eae85 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/856db0d718c14b688e339d1547c23ab8 : SUCCESS in 18m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7a2d4a8d860143218003994f389be4e3 : SUCCESS in 6m 24s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"7d177d09fabae6d8b3e7f352eeac06552c8f94dd","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-07-28 07:06:34.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"a953a4a392f45a85b249e571616a188f53310d0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-07-28 07:43:07.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":12},{"id":"63ad8a0e42f928f4d4e1d23434baef350d015626","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-07-28 07:46:14.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"ccb3d1710acfb44250de1bdab2933639031e4d63","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-07-28 07:47:13.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"8a27660bc98ac756daf773af17eceb919eff80e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-28 08:03:57.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b37fd7d4aa304b32b7beff3220e1c12a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4dd8d20459d84717bafae70118d678b2 : SUCCESS in 12m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/be4ea46c60754bcaac300d0c91be4698 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/86d39305524b48aeb309dba1741a3c71 : SUCCESS in 15m 17s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"0e154353607eb8168cd883715e01b633d27cbde3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-28 09:51:11.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9e04b4b0aa9742b7ad7497b3419cefaf\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/354844928bd747e18b93e18e9687e8bd : SUCCESS in 7m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5eef69772e34f6fa6ab4df83eef16db : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a582ad745734058870648374284e404 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0a7357596304751b51f291d9c674964 : SUCCESS in 6m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a93a4956038445891490af5f4cba7af : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/287c818ab7954a88a71c8bbdb28d7347 : SUCCESS in 5m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c9681ad88a34441afabca51435dd1d2 : FAILURE in 5m 49s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fcbd08ff94e84a2d9260e3b8c137c191 : SUCCESS in 5m 48s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/a1d418ea75444b89bb6851f65417fff3 : FAILURE in 1h 28m 59s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/bf1031265fea460a9fef1aa0b5d7f74f : SUCCESS in 2h 03m 49s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"f8dff801d4a85048e73e265e64b6d43624246c91","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-07-28 09:54:58.000000000","message":"Patch Set 13: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":13},{"id":"5accaa0f1f26218727a1365621c4472e56743651","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-07-31 01:33:51.000000000","message":"Patch Set 13:\n\n(3 comments)","accounts_in_message":[],"_revision_number":13},{"id":"622665e6b68645f6493cfaed6af63d508610c34a","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-07-31 02:37:34.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"f621baa8f02788714e452df6f13e453b6b0cf8e1","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-07-31 02:39:44.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"8e9ae1c1a11a195998c137b8bd85799ecb73ede4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-08-10 03:15:25.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"d317b899ad1e9dafb8e4aef24a02cc8748ddaeb3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 03:21:33.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ecbd2e4fd31468e89405812bc062d4f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1b1ffa7c473c4b3cb1c58e7af948ab48 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/24f2f1144b454f25a0e7cdfb21a24cc4 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/14fadd475de74f65b70a2eb96560fcd2 : SUCCESS in 4m 53s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"f642328306e009aac3c88c511f649a990bbeb397","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 05:25:39.000000000","message":"Patch Set 14: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f9545ca2565844ee83deb89ccb8ddb73\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5260108baac47a0b6f43d7a99593f60 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46b14a95d68c4f04854ee877dd375067 : FAILURE in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f0b4e25aafd4fdfb0366b3bf3e29897 : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f39e8aac3bd94fbe82a1374b058339eb : SUCCESS in 6m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/574cb9f1b98d46d0b616009b26c321af : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfcb955cfb24403c9fe64e5f3cb58b80 : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6cbd63f9ca8347b69680abf4c20a6da4 : FAILURE in 7m 51s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/94ba4ac3f47240819f23d5e9df1c251a : SUCCESS in 5m 58s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/88119f5b35444afda46f9fc1f03d7c47 : FAILURE in 1h 32m 25s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/bbdf6fe846e947f2a7cdb406f5901044 : FAILURE in 51m 02s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/07f8db3e815942ecac4984c34f7a92d0 : SUCCESS in 2h 09m 02s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"f02d1893d50639d5cf1fbf81b659b0a5f8f8d3b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-08-31 06:53:42.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"2168d2109911c4e424ba8cc5c45469c42838ed66","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 07:08:40.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d75deb0770e04746b83d820e2a95ff1a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7ab26504dd0c4415b47052c8afd53a09 : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d3244bc9de6a4aa5a253e0c777e4f560 : SUCCESS in 12m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9df7f6215029423ca85d77fd41de0f30 : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/27358db7714d4a2e877923688678530d : RETRY_LIMIT in 3m 18s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"57e55baed7830fc394af1f85236cdeca6c7585a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-08-31 07:10:34.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"87970f014feaaff4e6f2b208cbba637f8d43f330","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 07:25:39.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7b1a083efad4160a459808d99ac2ad7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6e5cd89a59724b8e91ba7f4d9668ac48 : SUCCESS in 11m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d03de095a02249998c738081c61b71a6 : SUCCESS in 12m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/81d0f87182e74d69b4a8370d4d5e5f68 : SUCCESS in 13m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6e3de12aa88f454abb4ccaa0beae5c86 : RETRY_LIMIT in 1m 31s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"cabbf3dc8cb57c2a2241c0e504b4755c4a5d937e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-08-31 07:33:31.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"774b905c2a211c78a03fea8655c248f7b4c2b681","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-08-31 07:37:05.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"5b73df2ed6caaccc8569e39dae69d03348a7793f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 07:55:16.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e064e70bb72d4a13b1ef68d81f1d8e2d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2f396c05b980437a99b4fb06225ca9cd : SUCCESS in 16m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e7591caf79824748a17e14b92fb81e0f : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/90c5263bbf7f4eff8fecee5a72cf8143 : SUCCESS in 12m 23s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ad53ec79c7cc47e7bc3b5a6dca319274 : RETRY_LIMIT in 1m 49s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"94b22e3e54b04d23b6e24f4053e7a1d438f2895b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 10:42:47.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2ee377e92ea4490b179fa5892f55e8b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d97a73344357402cbeed27e9ddc16873 : SUCCESS in 7m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50a7a3d088d844899656860e380019d2 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13049c357d06479f9b06f094dcc97109 : SUCCESS in 7m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8dcebb17e281419ca84043962bd2a9a3 : SUCCESS in 6m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/560c5cb1cab3491daf1b5a08b6807679 : SUCCESS in 7m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f34396b7b08646e3b3967e72157da04d : RETRY_LIMIT in 1m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67c3ea4ea0f04da1a63a4ed91dd2c90e : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3bf84fd22e9e4563be9000ce005ed9f4 : SUCCESS in 4m 34s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/91fbf42aed934df9a455b145cd799407 : SUCCESS in 5m 17s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/6a6c92b3f22f4146953191c973945b75 : TIMED_OUT in 3h 02m 54s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/8cc677d66775411c94d6c162ee80b6f8 : SUCCESS in 1h 34m 01s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/2828638426a14d37a774391a291ba8d3 : TIMED_OUT in 3h 04m 04s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"3a2bfbfdb3d134093dec5a0d269379dd9a07e77c","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-09-01 02:03:27.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"841c043170c88a607fd90655ab8ca126cfe7d03d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 02:36:43.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/246630529b8d4bff847cbaf2bef65a50\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/01287907d79c4e2d9f24ea900fcde5cd : SUCCESS in 13m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cf29cc4886ac499f95aa964d82590932 : SUCCESS in 13m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0ba622df8e9b410a89403a1256395944 : SUCCESS in 9m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3b2e36af7d6e408abf68860cb0fa1594 : RETRY_LIMIT in 4m 01s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"f2a539d2f719d254d05eddc8be32e71f022e5cd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 04:30:54.000000000","message":"Patch Set 18:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71f7c1a495ab4077a2b8e993d8ff7097\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5af6db55a544722911b9ce0083f4070 : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d4866571ab048a68f6c22427452bbd5 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/87f289c5a08e43718ac8e8b3f092e793 : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81412f98e2d9420ab311f901ed1ad445 : SUCCESS in 5m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a9e69c4a0204fb987158a1201747a12 : SUCCESS in 5m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5d0ffbc8216d499e843018243bc31088 : RETRY_LIMIT in 1m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/441486edbdeb450c9af7faefe397f852 : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25ec0b71bb194cb8a7d899d7bab7f319 : SUCCESS in 5m 37s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2db24217ef914dfca38a7ab47c7de7b2 : SUCCESS in 5m 37s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/f475b66454104d9dafb9081861fd53af : SUCCESS in 2h 15m 30s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/1224b76819d14f489e1eecf25e4f0af3 : SUCCESS in 1h 31m 31s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/86823ccc56444929a1aac9d6812dd7b8 : SUCCESS in 2h 12m 54s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"632a2a608f18915598b5361fbc24f810c8840153","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-09-01 05:59:56.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"7d4b9af849d4c8d762e7832d240ed36b0973f503","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-09-04 07:06:43.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"55f6388a162f1567f4e8cbc6f20099e429b0c957","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-09-06 02:54:13.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"b595b410ccdd9e80836fad800767ac64057b1458","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-09-06 02:55:06.000000000","message":"Patch Set 19:\n\n(2 comments)","accounts_in_message":[],"_revision_number":19},{"id":"1930bc0739bc5242f97cebb9bb9babd7fbbfed9a","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-09-06 02:56:52.000000000","message":"Patch Set 19: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"a589291e80c14460552f41e042273feea9bdfc9c","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-09-06 02:59:08.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"3381c4004d6c020d7b8df74dc775a2fb53c5b749","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 03:04:17.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb5a2a18607a4b52a38b68ba10a631db\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/25a5d3523f6f441796c5453d0bd31f94 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2cb90755701e4afd910c06f7d1106100 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cd21e4a5e01045aeb7f7084ccaba972c : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/daf650615cc34917896bf15e8d9460bf : RETRY_LIMIT in 1m 38s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"755d85b4aa1ce4f4b2174c4deed073d35c03151d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 03:07:37.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f8a597a572814a8bb2fd4e94d9078c80\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5268af59b74c44a28ddbadc9fe7d6ccc : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/909af5f38edb42ceade7438e8ae1cb10 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/817c62099682478a8d102d4b65ab781a : SUCCESS in 7m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82d1e4cd7e154d608e95b6acbd1eef97 : SUCCESS in 5m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/10855e403d2c46209b64816b82b0d121 : SUCCESS in 6m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5c3918b53546461996e652a0ac56af43 : RETRY_LIMIT in 2m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf1ea69b056b4816afd83a6746ca1ce0 : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/251761f6a2b541c4a5532305c4863b47 : SUCCESS in 6m 16s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/36e97dea50a14f5b980678aa4140b277 : SUCCESS in 7m 22s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/61f11276405749f488780a63ef8063fb : FAILURE in 8m 11s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/6d34a0ee6dd64d5081551651e0aad504 : FAILURE in 8m 11s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/1ff360cb85a64d81a1f8bdc936ec3dde : FAILURE in 8m 05s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"052672c26c41fdae8db1d11659e13a029dab3662","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-09-07 05:42:01.000000000","message":"Patch Set 20: Patch Set 19 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"340911133719ec278679037008b596ac3a5dd46e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 05:57:23.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52b6272ff7934fb6926fcd67b01af0d9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/39e56e6575a345e8bba755de22bea8fc : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f7d20d91fef7423abf38491b89b5cda3 : SUCCESS in 13m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b06d669921364f06aa9befdbd8f52bf4 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/da5de764a0cb4a4fad6d9c7ee0502b1c : RETRY_LIMIT in 1m 31s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"53b0917229992d80cbfbc2ed909d7ad8b68cd2e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 08:48:57.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dff1d849151143fdba7238d04f6d6278\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d09cf3c4b8394d7bb054385fc0404ef2 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d099c10ca197421ab797f448c933a25c : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/115b789486f64ab780492bb11c365f57 : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b15d77c0476b424cbf94dfc307a35ed2 : SUCCESS in 7m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7a4ead8249bc44c98c5fcadf76d6697e : SUCCESS in 5m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b55d5b62502947b1a92ac630c5ef223d : RETRY_LIMIT in 1m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85887db699494106b78ef640166166fa : SUCCESS in 5m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/547733d1e3c74e60a5de4d7c65694084 : SUCCESS in 5m 02s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3e3e78a0e12e492598e98300c214856c : SUCCESS in 5m 00s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/399d5cdff8f64320be0e1068e2a07dcc : TIMED_OUT in 3h 04m 31s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/3956895f154b4748b9760a8c642b055c : SUCCESS in 2h 15m 25s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/a67a62b84b104a5a80e8e8fb62ddfeed : FAILURE in 3h 03m 53s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"53732edd9901c53a9a8be64f31cc3016e8b767f8","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-09-08 07:16:34.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"b1a9381ffc7a49d12d59a2ee8fb89e405ef1b155","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 07:23:05.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78799654cd3c4deb87c96cc9d33230b7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a6c2760b1c884f2188d9cd95a7cde9aa : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8e84f9ac56644c17b4e200ed1883ff2b : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/943ae03bf89c4c43b66154d5b5bc867a : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/07310b43e5d74efeb0aad312fbc915a7 : RETRY_LIMIT in 1m 39s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"fc5c241d008d37bcd220ca115c2ba6be13439d3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 09:37:17.000000000","message":"Patch Set 20:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d9c8ec898ab454c9d24916c4cb77f28\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2478a45193354bd5b8fcb996dadcae4c : SUCCESS in 7m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c25a7fda89384eb3979e1b972a95f55b : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3333a61343f3449db4c76040cdb9324b : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f1e7b3cdcc1487aacd71f02fdab138e : SUCCESS in 6m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6599c92460104f3388b5afbd2f8c30bd : SUCCESS in 5m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ffa28c01d2414e269e77dbcf93a2679b : RETRY_LIMIT in 1m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed300ec15a884da193ef2c24a779144d : SUCCESS in 5m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07556c82b2394a69be70afeaec1c7880 : SUCCESS in 5m 37s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ccfe086cd4e14d8e8288f1e611432bdf : SUCCESS in 5m 27s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/82158b2760d04293985f12708df50ec0 : SUCCESS in 2h 17m 17s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/19c9a8789080456f8988242c52a42eb1 : SUCCESS in 1h 29m 54s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/5790e659daa2446d804f3026f187c653 : FAILURE in 30m 58s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"0b49c04e8b34c20cec631c173a2826db1adedf30","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-09-13 06:17:42.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"3fef1ba139d3c3f0683f102fcded2966244ec442","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 06:30:10.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c672857c41746789eb6320b4d7f566e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5924a8f2ecae46fcab1da3c4b43a55af : SUCCESS in 9m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3e9036a2de74437e8f203226980e7ee7 : SUCCESS in 10m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f9b9f0c619f24a2b9921e06599f3b517 : SUCCESS in 10m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b75cbe4a1288408eb318ea8343476d6a : RETRY_LIMIT in 2m 20s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"dc3e9095ebb357a402762fddea957fa6d362a716","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-09-13 06:35:44.000000000","message":"Patch Set 20: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"24adc0a84dabb98ac3b1c95f5ebe728965699d9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 09:22:07.000000000","message":"Patch Set 20:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6be368ab4b74b398705b8ffe8b3c55e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/150b026dd5a14c55bcff770d6597677a : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/275bd2111461433298bbb64a22cabce3 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91ea44a156f547f48d7a84c08114be56 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3aee6dddb1e34e239929addcce07c56b : SUCCESS in 6m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7565fba99f374b6baac8304bb25ca45e : SUCCESS in 5m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4150b1467ba040b180ebfede3c881f11 : RETRY_LIMIT in 2m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2aaad59d785a4c5b85caf28732daad3c : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76361d3f8f67416cae75858d696ebf55 : SUCCESS in 6m 01s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/779aaad8b8174cbe82775cdf47fe84a3 : SUCCESS in 5m 13s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/eb48bc1c7ea64f3398f1f26d1711defd : SUCCESS in 2h 58m 52s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/72b7886fb891480fbaceacdccaf180e9 : SUCCESS in 2h 00m 38s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/4e7458511c644de9bb68d72a76a9512b : TIMED_OUT in 3h 02m 54s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"753f5c48e6db98473ef28de757706df66cc7627d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-09-21 01:21:32.000000000","message":"Patch Set 21: Patch Set 20 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"ebad8c32bbb9b1c037f9118948c26ac3fb7c669c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 01:36:06.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44a782dd28c04d058fa37caee9409d04\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e78116f11de84c03bc3a8be6cf6fd1bc : SUCCESS in 11m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6ea8bf057f7146d2a63670fb06f62818 : SUCCESS in 11m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a59dda2ce9254e219e27a318b6d3a8f6 : SUCCESS in 11m 51s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4d6a18f674eb4dd0a31a82e8d67dc898 : RETRY_LIMIT in 2m 20s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"3777288aa13cfd3ba3d529b709f9942597d618fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 04:31:57.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2e8abaa0e6d47509c02b6b2f55988ae\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88d446b42b91498d87ceb09103ef86a9 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/485bd489976b4808b6ae02caa8eec045 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21cb1ba2c6fe4515bd645b50c882a599 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5cc906ea10a48b1b8c57540ca29c768 : SUCCESS in 5m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a469d4bd62141b88cde588e1cf08169 : SUCCESS in 6m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9a2c96c148f74acc8e97dd23dbf5954c : RETRY_LIMIT in 2m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dbcd2441df24e57ba54f68023e4090c : SUCCESS in 4m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e38207f1ae54b3b8654da78b0710527 : SUCCESS in 5m 58s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/88ac91ff7eb74d8599f926733c29af8b : SUCCESS in 5m 16s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/8c7a6cc9c8fc417085b40a5f68d344bc : TIMED_OUT in 3h 04m 38s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/0beb9b072e5f45d7a2149f97480f1165 : SUCCESS in 2h 16m 17s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/82bbaa5f98c148b38dbb842b7ac8f07a : SUCCESS in 2h 20m 32s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"879df86b5fab2317c9c7602fd30af8cd7f8d7737","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-10-08 02:17:19.000000000","message":"Patch Set 21: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":21},{"id":"c32137d20533fc8273862310cd1c7dd2ceb8cd6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-08 02:18:14.000000000","message":"Patch Set 21: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":21},{"id":"4b4d7de3be04a4d0bbfbb5e3c25784bdedaaa621","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-08 04:43:35.000000000","message":"Patch Set 21: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe0a694b567d4f44b4b4a9764ba29266\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8577962cbee48269515a9fab79f4643 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec7f6705dfad4caeb350a6ff27184d93 : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/211650343b004632ba4981a5ee44e75f : SUCCESS in 7m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4cfabb3e2dc24e3aa56850b478ac83e1 : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4c699a95ba84392847dad917b340d43 : SUCCESS in 5m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a50362eec2b2434d896c127406d33376 : SUCCESS in 5m 36s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/96adb355c5494d5a983f8b588be8b629 : SUCCESS in 2h 23m 31s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"c47465179b379844a73a2700b998baec6f2f196d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-08 04:43:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":21},{"id":"0937f3a1928278eeadaeb399fd6a448993a0e443","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-08 04:44:38.000000000","message":"Patch Set 21:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5b0832b7a4d488196a482e3f3225162\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b81ecbc30c849c3846e52c07b3394b0 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8ef2955fc6346e089526d81f315a2a8 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":21}],"current_revision_number":21,"current_revision":"4c83bb8862c987cb0f88ee35c5f4b2cd20d86844","revisions":{"a8003897a9878382efebb9ea2727be5bffac67d0":{"kind":"REWORK","_number":1,"created":"2023-05-24 10:25:34.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/65/884165/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/884165/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/884165/1"}}},"commit":{"parents":[{"commit":"ee3ba2375bdeaf41d39e569ec3551aacf1118ec8","subject":"Remove tempest ipv6 job to experimental pipeline","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ee3ba2375bdeaf41d39e569ec3551aacf1118ec8"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2023-05-24 10:23:55.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2023-05-24 10:23:55.000000000","tz":420},"subject":"Separate backup docker image for each database version","message":"Separate backup docker image for each database version\n\nStory: #2010770\nTask: #48087\nChange-Id: I08063748e15de6767b437aa443311d41e25ed578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a8003897a9878382efebb9ea2727be5bffac67d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a8003897a9878382efebb9ea2727be5bffac67d0"}]},"branch":"refs/heads/master"},"c03e590066fac2dddb68487b925fe371b90f76b6":{"kind":"NO_CHANGE","_number":2,"created":"2023-05-24 10:28:27.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/65/884165/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/884165/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/884165/2"}}},"commit":{"parents":[{"commit":"ee3ba2375bdeaf41d39e569ec3551aacf1118ec8","subject":"Remove tempest ipv6 job to experimental pipeline","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ee3ba2375bdeaf41d39e569ec3551aacf1118ec8"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2023-05-24 10:23:55.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2023-05-24 10:28:08.000000000","tz":420},"subject":"Separate backup docker image for each database version","message":"Separate backup docker image for each database version\n\nStory: #2010770\nTask: #48087\nChange-Id: I08063748e15de6767b437aa443311d41e25ed578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c03e590066fac2dddb68487b925fe371b90f76b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c03e590066fac2dddb68487b925fe371b90f76b6"}]},"branch":"refs/heads/master"},"71c5c0f9f79708436993df2f88ba35d249f6c244":{"kind":"REWORK","_number":3,"created":"2023-06-29 03:47:11.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/65/884165/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/884165/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/884165/3"}}},"commit":{"parents":[{"commit":"ee3ba2375bdeaf41d39e569ec3551aacf1118ec8","subject":"Remove tempest ipv6 job to experimental pipeline","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ee3ba2375bdeaf41d39e569ec3551aacf1118ec8"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2023-05-24 10:23:55.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2023-06-29 03:46:48.000000000","tz":420},"subject":"Separate backup docker image for each database version","message":"Separate backup docker image for each database version\n\nStory: #2010770\nTask: #48087\nChange-Id: I08063748e15de6767b437aa443311d41e25ed578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/71c5c0f9f79708436993df2f88ba35d249f6c244"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/71c5c0f9f79708436993df2f88ba35d249f6c244"}]},"branch":"refs/heads/master"},"1e849cf29306182942dff9b721f496c8a56be663":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-06-29 06:58:46.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/65/884165/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/884165/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/884165/4"}}},"commit":{"parents":[{"commit":"1fe5dcbf75963a0d4dd5a929440f4d7f204f8809","subject":"Merge \"CI: fix trove tempest postgresql job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1fe5dcbf75963a0d4dd5a929440f4d7f204f8809"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2023-05-24 10:23:55.000000000","tz":420},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-06-29 06:58:46.000000000","tz":0},"subject":"Separate backup docker image for each database version","message":"Separate backup docker image for each database version\n\nStory: #2010770\nTask: #48087\nChange-Id: I08063748e15de6767b437aa443311d41e25ed578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1e849cf29306182942dff9b721f496c8a56be663"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1e849cf29306182942dff9b721f496c8a56be663"}]},"branch":"refs/heads/master","description":"Rebase"},"7a24ccdcbb8db3f8070137345de138a2cad374d5":{"kind":"REWORK","_number":5,"created":"2023-07-07 08:00:21.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/65/884165/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/884165/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/884165/5"}}},"commit":{"parents":[{"commit":"ee3ba2375bdeaf41d39e569ec3551aacf1118ec8","subject":"Remove tempest ipv6 job to experimental pipeline","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ee3ba2375bdeaf41d39e569ec3551aacf1118ec8"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2023-05-24 10:23:55.000000000","tz":420},"committer":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-07 08:00:01.000000000","tz":420},"subject":"Separate backup docker image for each database version","message":"Separate backup docker image for each database version\n\nStory: #2010770\nTask: #48087\nChange-Id: I08063748e15de6767b437aa443311d41e25ed578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7a24ccdcbb8db3f8070137345de138a2cad374d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7a24ccdcbb8db3f8070137345de138a2cad374d5"}]},"branch":"refs/heads/master"},"dd811628e6f359debe2f1725c679c46be4639bd0":{"kind":"REWORK","_number":6,"created":"2023-07-14 08:36:12.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/65/884165/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/884165/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/884165/6"}}},"commit":{"parents":[{"commit":"ee3ba2375bdeaf41d39e569ec3551aacf1118ec8","subject":"Remove tempest ipv6 job to experimental pipeline","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ee3ba2375bdeaf41d39e569ec3551aacf1118ec8"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2023-05-24 10:23:55.000000000","tz":420},"committer":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-14 08:35:58.000000000","tz":420},"subject":"Separate backup docker image for each database version","message":"Separate backup docker image for each database version\n\nStory: #2010770\nTask: #48087\nChange-Id: I08063748e15de6767b437aa443311d41e25ed578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dd811628e6f359debe2f1725c679c46be4639bd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dd811628e6f359debe2f1725c679c46be4639bd0"}]},"branch":"refs/heads/master"},"9b99623923a4b9e16dfa0da562a141061f7c9dcd":{"kind":"REWORK","_number":7,"created":"2023-07-27 03:32:22.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/65/884165/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/884165/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/884165/7"}}},"commit":{"parents":[{"commit":"ee3ba2375bdeaf41d39e569ec3551aacf1118ec8","subject":"Remove tempest ipv6 job to experimental pipeline","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ee3ba2375bdeaf41d39e569ec3551aacf1118ec8"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2023-05-24 10:23:55.000000000","tz":420},"committer":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-27 03:32:06.000000000","tz":420},"subject":"Separate backup docker image for each database version","message":"Separate backup docker image for each database version\n\nStory: #2010770\nTask: #48087\nChange-Id: I08063748e15de6767b437aa443311d41e25ed578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9b99623923a4b9e16dfa0da562a141061f7c9dcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9b99623923a4b9e16dfa0da562a141061f7c9dcd"}]},"branch":"refs/heads/master"},"e9e4654182279bba135fbf500c0b1f0f4f9fdbc8":{"kind":"REWORK","_number":8,"created":"2023-07-27 07:13:10.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/65/884165/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/884165/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/884165/8"}}},"commit":{"parents":[{"commit":"ee3ba2375bdeaf41d39e569ec3551aacf1118ec8","subject":"Remove tempest ipv6 job to experimental pipeline","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ee3ba2375bdeaf41d39e569ec3551aacf1118ec8"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2023-05-24 10:23:55.000000000","tz":420},"committer":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-27 07:12:57.000000000","tz":420},"subject":"Separate backup docker image for each database version","message":"Separate backup docker image for each database version\n\nStory: #2010770\nTask: #48087\nChange-Id: I08063748e15de6767b437aa443311d41e25ed578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9e4654182279bba135fbf500c0b1f0f4f9fdbc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9e4654182279bba135fbf500c0b1f0f4f9fdbc8"}]},"branch":"refs/heads/master"},"35cbd150544bf455aa185b9d8705d765208acbb0":{"kind":"REWORK","_number":9,"created":"2023-07-27 07:16:29.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/65/884165/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/884165/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/884165/9"}}},"commit":{"parents":[{"commit":"ee3ba2375bdeaf41d39e569ec3551aacf1118ec8","subject":"Remove tempest ipv6 job to experimental pipeline","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ee3ba2375bdeaf41d39e569ec3551aacf1118ec8"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2023-05-24 10:23:55.000000000","tz":420},"committer":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-27 07:16:15.000000000","tz":420},"subject":"Separate backup docker image for each database version","message":"Separate backup docker image for each database version\n\nStory: #2010770\nTask: #48087\nChange-Id: I08063748e15de6767b437aa443311d41e25ed578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/35cbd150544bf455aa185b9d8705d765208acbb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/35cbd150544bf455aa185b9d8705d765208acbb0"}]},"branch":"refs/heads/master"},"6456f638852f75c53b25b136bcfa1279ab876a02":{"kind":"REWORK","_number":10,"created":"2023-07-27 09:16:59.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/65/884165/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/884165/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/884165/10"}}},"commit":{"parents":[{"commit":"ee3ba2375bdeaf41d39e569ec3551aacf1118ec8","subject":"Remove tempest ipv6 job to experimental pipeline","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ee3ba2375bdeaf41d39e569ec3551aacf1118ec8"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2023-05-24 10:23:55.000000000","tz":420},"committer":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-27 09:16:42.000000000","tz":420},"subject":"Separate backup docker image for each database version","message":"Separate backup docker image for each database version\n\nStory: #2010770\nTask: #48087\nChange-Id: I08063748e15de6767b437aa443311d41e25ed578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6456f638852f75c53b25b136bcfa1279ab876a02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6456f638852f75c53b25b136bcfa1279ab876a02"}]},"branch":"refs/heads/master"},"59ae1ff538ec79bfc8cf1eb533763b5cc2f5fec0":{"kind":"REWORK","_number":11,"created":"2023-07-28 05:50:55.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/65/884165/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/884165/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/884165/11"}}},"commit":{"parents":[{"commit":"ee3ba2375bdeaf41d39e569ec3551aacf1118ec8","subject":"Remove tempest ipv6 job to experimental pipeline","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ee3ba2375bdeaf41d39e569ec3551aacf1118ec8"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2023-05-24 10:23:55.000000000","tz":420},"committer":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-28 05:50:39.000000000","tz":420},"subject":"Separate backup docker image for each database version","message":"Separate backup docker image for each database version\n\nStory: #2010770\nTask: #48087\nChange-Id: I08063748e15de6767b437aa443311d41e25ed578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/59ae1ff538ec79bfc8cf1eb533763b5cc2f5fec0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/59ae1ff538ec79bfc8cf1eb533763b5cc2f5fec0"}]},"branch":"refs/heads/master"},"ac6a396c3393600d1b2a87381026a492d52b6d38":{"kind":"REWORK","_number":12,"created":"2023-07-28 07:43:07.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/65/884165/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/884165/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/884165/12"}}},"commit":{"parents":[{"commit":"ee3ba2375bdeaf41d39e569ec3551aacf1118ec8","subject":"Remove tempest ipv6 job to experimental pipeline","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ee3ba2375bdeaf41d39e569ec3551aacf1118ec8"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2023-05-24 10:23:55.000000000","tz":420},"committer":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-28 07:42:45.000000000","tz":420},"subject":"Separate backup docker image for each database version","message":"Separate backup docker image for each database version\n\nStory: #2010770\nTask: #48087\nChange-Id: I08063748e15de6767b437aa443311d41e25ed578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ac6a396c3393600d1b2a87381026a492d52b6d38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ac6a396c3393600d1b2a87381026a492d52b6d38"}]},"branch":"refs/heads/master"},"e8130b60b4441cba62b7c35409da49590e74c2ee":{"kind":"REWORK","_number":13,"created":"2023-07-28 07:46:14.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/65/884165/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/884165/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/884165/13"}}},"commit":{"parents":[{"commit":"ee3ba2375bdeaf41d39e569ec3551aacf1118ec8","subject":"Remove tempest ipv6 job to experimental pipeline","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ee3ba2375bdeaf41d39e569ec3551aacf1118ec8"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2023-05-24 10:23:55.000000000","tz":420},"committer":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-28 07:45:48.000000000","tz":420},"subject":"Separate backup docker image for each database version","message":"Separate backup docker image for each database version\n\nStory: #2010770\nTask: #48087\nChange-Id: I08063748e15de6767b437aa443311d41e25ed578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e8130b60b4441cba62b7c35409da49590e74c2ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e8130b60b4441cba62b7c35409da49590e74c2ee"}]},"branch":"refs/heads/master"},"03642c6a35fd78a21394478b35da832b3c5918f9":{"kind":"REWORK","_number":14,"created":"2023-08-10 03:15:25.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/65/884165/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/884165/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/884165/14"}}},"commit":{"parents":[{"commit":"94ec942eaab0496e5ffbf448fa7f1ad2bff7888d","subject":"Merge \"Adds postgres guestagent test driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/94ec942eaab0496e5ffbf448fa7f1ad2bff7888d"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2023-05-24 10:23:55.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2023-08-10 03:14:32.000000000","tz":420},"subject":"Separate backup docker image for each database version","message":"Separate backup docker image for each database version\n\nStory: #2010770\nTask: #48087\nChange-Id: I08063748e15de6767b437aa443311d41e25ed578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/03642c6a35fd78a21394478b35da832b3c5918f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/03642c6a35fd78a21394478b35da832b3c5918f9"}]},"branch":"refs/heads/master"},"403c7161c958cfda43a091124bdfc990ed54f89b":{"kind":"REWORK","_number":15,"created":"2023-08-31 06:53:42.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/65/884165/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/884165/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/884165/15"}}},"commit":{"parents":[{"commit":"05c30cbcf6436713dce71e16894a2eb53eac5f12","subject":"Merge \"Fix: failed to create the replicas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/05c30cbcf6436713dce71e16894a2eb53eac5f12"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2023-05-24 10:23:55.000000000","tz":420},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-08-31 06:51:59.000000000","tz":480},"subject":"Separate backup docker image for each database version","message":"Separate backup docker image for each database version\n\nCo-Authored-By: wu.chunyang \u003cwchy1001@gmail.com\u003e\n\nStory: #2010770\nTask: #48087\nChange-Id: I08063748e15de6767b437aa443311d41e25ed578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/403c7161c958cfda43a091124bdfc990ed54f89b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/403c7161c958cfda43a091124bdfc990ed54f89b"}]},"branch":"refs/heads/master"},"f78986a8a37a2e42dafc40fd203a4a0cfe89d0d3":{"kind":"REWORK","_number":16,"created":"2023-08-31 07:10:34.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/65/884165/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/884165/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/884165/16"}}},"commit":{"parents":[{"commit":"05c30cbcf6436713dce71e16894a2eb53eac5f12","subject":"Merge \"Fix: failed to create the replicas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/05c30cbcf6436713dce71e16894a2eb53eac5f12"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2023-05-24 10:23:55.000000000","tz":420},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-08-31 07:10:00.000000000","tz":480},"subject":"Separate backup docker image for each database version","message":"Separate backup docker image for each database version\n\nCo-Authored-By: wu.chunyang \u003cwchy1001@gmail.com\u003e\n\nStory: #2010770\nTask: #48087\nChange-Id: I08063748e15de6767b437aa443311d41e25ed578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f78986a8a37a2e42dafc40fd203a4a0cfe89d0d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f78986a8a37a2e42dafc40fd203a4a0cfe89d0d3"}]},"branch":"refs/heads/master"},"548a2ae06527bf91007a8ddb84dcdeb77e94add0":{"kind":"REWORK","_number":17,"created":"2023-08-31 07:33:31.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/65/884165/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/884165/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/884165/17"}}},"commit":{"parents":[{"commit":"05c30cbcf6436713dce71e16894a2eb53eac5f12","subject":"Merge \"Fix: failed to create the replicas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/05c30cbcf6436713dce71e16894a2eb53eac5f12"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2023-05-24 10:23:55.000000000","tz":420},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-08-31 07:33:10.000000000","tz":480},"subject":"Separate backup docker image for each database version","message":"Separate backup docker image for each database version\n\nCo-Authored-By: wu.chunyang \u003cwchy1001@gmail.com\u003e\n\nStory: #2010770\nTask: #48087\nChange-Id: I08063748e15de6767b437aa443311d41e25ed578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/548a2ae06527bf91007a8ddb84dcdeb77e94add0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/548a2ae06527bf91007a8ddb84dcdeb77e94add0"}]},"branch":"refs/heads/master"},"1c38b03564b5757084d9b0a6859165da830fea35":{"kind":"REWORK","_number":18,"created":"2023-08-31 07:37:05.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/65/884165/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/884165/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/884165/18"}}},"commit":{"parents":[{"commit":"05c30cbcf6436713dce71e16894a2eb53eac5f12","subject":"Merge \"Fix: failed to create the replicas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/05c30cbcf6436713dce71e16894a2eb53eac5f12"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2023-05-24 10:23:55.000000000","tz":420},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-08-31 07:36:53.000000000","tz":480},"subject":"Separate backup docker image for each database version","message":"Separate backup docker image for each database version\n\nCo-Authored-By: wu.chunyang \u003cwchy1001@gmail.com\u003e\n\nStory: #2010770\nTask: #48087\nChange-Id: I08063748e15de6767b437aa443311d41e25ed578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1c38b03564b5757084d9b0a6859165da830fea35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1c38b03564b5757084d9b0a6859165da830fea35"}]},"branch":"refs/heads/master"},"697001f6d4355426e99f37b1832e413f7e9d6183":{"kind":"REWORK","_number":19,"created":"2023-09-06 02:54:13.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/65/884165/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/884165/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/884165/19"}}},"commit":{"parents":[{"commit":"05c30cbcf6436713dce71e16894a2eb53eac5f12","subject":"Merge \"Fix: failed to create the replicas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/05c30cbcf6436713dce71e16894a2eb53eac5f12"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2023-05-24 10:23:55.000000000","tz":420},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-09-06 02:53:49.000000000","tz":480},"subject":"Separate backup docker image for each database version","message":"Separate backup docker image for each database version\n\nCo-Authored-By: wu.chunyang \u003cwchy1001@gmail.com\u003e\n\nStory: #2010770\nTask: #48087\nChange-Id: I08063748e15de6767b437aa443311d41e25ed578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/697001f6d4355426e99f37b1832e413f7e9d6183"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/697001f6d4355426e99f37b1832e413f7e9d6183"}]},"branch":"refs/heads/master"},"94cbbea34fcb6f0f36e49abe80bb06c99ec33969":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2023-09-07 05:42:01.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/65/884165/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/884165/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/884165/20"}}},"commit":{"parents":[{"commit":"0a6211645a12a7ae64259ebc47be4995f532e433","subject":"Merge \"Fix Zuul warnings with regexps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0a6211645a12a7ae64259ebc47be4995f532e433"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2023-05-24 10:23:55.000000000","tz":420},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-09-07 05:42:01.000000000","tz":0},"subject":"Separate backup docker image for each database version","message":"Separate backup docker image for each database version\n\nCo-Authored-By: wu.chunyang \u003cwchy1001@gmail.com\u003e\n\nStory: #2010770\nTask: #48087\nChange-Id: I08063748e15de6767b437aa443311d41e25ed578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/94cbbea34fcb6f0f36e49abe80bb06c99ec33969"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/94cbbea34fcb6f0f36e49abe80bb06c99ec33969"}]},"branch":"refs/heads/master","description":"Rebase"},"4c83bb8862c987cb0f88ee35c5f4b2cd20d86844":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2023-09-21 01:21:32.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/65/884165/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/884165/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/884165/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/884165/21"}}},"commit":{"parents":[{"commit":"14ea484284d95c6bfd2a44c051b0a71ce022fd24","subject":"Merge \"Update master for stable/2023.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/14ea484284d95c6bfd2a44c051b0a71ce022fd24"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2023-05-24 10:23:55.000000000","tz":420},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-09-21 01:21:32.000000000","tz":0},"subject":"Separate backup docker image for each database version","message":"Separate backup docker image for each database version\n\nCo-Authored-By: wu.chunyang \u003cwchy1001@gmail.com\u003e\n\nStory: #2010770\nTask: #48087\nChange-Id: I08063748e15de6767b437aa443311d41e25ed578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4c83bb8862c987cb0f88ee35c5f4b2cd20d86844"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4c83bb8862c987cb0f88ee35c5f4b2cd20d86844"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
