)]}'
{"id":"openstack%2Ftrove~884672","triplet_id":"openstack%2Ftrove~master~I2c188d27f0595ac3b2d71f4612d93829915e389c","project":"openstack/trove","branch":"master","topic":"ci","hashtags":[],"change_id":"I2c188d27f0595ac3b2d71f4612d93829915e389c","subject":"Don\u0027t ignore H306 pep8 check","status":"MERGED","created":"2023-05-30 09:57:29.000000000","updated":"2023-06-06 14:44:41.000000000","submitted":"2023-06-06 14:43:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"884672-ci","meta_rev_id":"88bea8bfc5f12cac7b8b7933794339c2093a1d4d","_number":884672,"virtual_id_number":884672,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-06 14:43:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-06 12:49:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-06-06 12:49:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-30 10:04:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-30 11:47:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6938a1edd20f7aee251a17d41db685f2b710a072","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-05-30 09:57:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8da4b073534dc88cb5a2abc8e6fc8047433c3b32","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 10:04:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/402d6af584dc43e59f90b92397d766d7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7c1722dac421432786d57621c59bd648 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/386d4c1a0c314e73b38076ff2d87de21 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e0008faee97f4013ac2030ae983d9c8b : SUCCESS in 5m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5a73121094f64ef92c792d3652328d24b03130dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 11:47:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1528b2a50d0545d88bb545cd9891eac0\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/921969006d074fb1974e86124ef66baf : SUCCESS in 5m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/064c8fee3f9149b4968c96ac7f408d26 : FAILURE in 6m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20d411e519794f1fae5f3e7775aeb27b : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6596fb7979cd4858911c156c2ac5ccac : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ad1777e28c747c1954168692e0fdf4b : SUCCESS in 6m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9004afa6b54d46d2b40382b4bc7e513b : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6077dac4ee42422f8185ec04cb294686 : SUCCESS in 6m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/edf0a7dc45c34e25aa8e788e7561f1ab : SUCCESS in 5m 06s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1e4957bf81154799bf65768c8b0a0056 : SUCCESS in 4m 51s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/0d469daf648b44b8b4fad1d31e364e3d : SUCCESS in 1h 48m 23s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/accb9ee0220643f5ac9e3a549614885e : SUCCESS in 1h 10m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e571e3d3fae268bd8852e36144913fa511dbd441","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-06 12:49:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"90f732ec08dad8955ea5a66e49eeb0902f006205","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 12:50:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b9c81cae26b09f0761208711104fe3d80dfba9a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 14:43:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2a7eab4082b458fac104ac9abe91e59\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a97937d126444abfb1ce81865227b186 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa91df93180b45539ab57e735369d896 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4837d37d2a5e4af89b92528583c152e6 : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5bf5f11f52244305b9320c4d1953307b : SUCCESS in 5m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/696a84fe78ef48e49901679ea58f3f4b : SUCCESS in 7m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87cd4fc21fa74d46b9cc799a4e96fba7 : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/783bed75fe08465ab4e35f842de77ee6 : SUCCESS in 7m 19s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/293ecefd57c34bbab11bb09d01cf3aa5 : SUCCESS in 1h 52m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"06d5969eaabf28819114fa917629d7004a3593a4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 14:43:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"88bea8bfc5f12cac7b8b7933794339c2093a1d4d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 14:44:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4072d2fc585400e9ffdce87b47839f4\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/6e1ae5f3bfbf49adbe132c47d60d39f1 : SUCCESS in 45s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5a5d688dcec4d168fb5393422cadf4e : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5bba21072398491086a23a919c5a5f1c : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c1761147ba7f22cbb25a634b151bca7cdc4f43cf","revisions":{"c1761147ba7f22cbb25a634b151bca7cdc4f43cf":{"kind":"REWORK","_number":1,"created":"2023-05-30 09:57:29.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/72/884672/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/72/884672/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/884672/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/884672/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/72/884672/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/72/884672/1"}}},"commit":{"parents":[{"commit":"e854cb58538f83d7a4538218b27d8ad25d865f35","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e854cb58538f83d7a4538218b27d8ad25d865f35"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-05-30 09:51:16.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-05-30 09:51:16.000000000","tz":480},"subject":"Don\u0027t ignore H306 pep8 check","message":"Don\u0027t ignore H306 pep8 check\n\nH306: import in alphabeical order\n\nImports should be grouped in the following order:\nstandard library imports\nrelated third party imports\nlocal application/library specific imports\nput a blank line between each group of imports.\nhttps://docs.openstack.org/hacking/latest/user/hacking.html#imports\n\nChange-Id: I2c188d27f0595ac3b2d71f4612d93829915e389c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c1761147ba7f22cbb25a634b151bca7cdc4f43cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c1761147ba7f22cbb25a634b151bca7cdc4f43cf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
