)]}'
{"id":"openstack%2Ftrove~885326","triplet_id":"openstack%2Ftrove~master~I24695238075f9b1354ee4c54673d021de481a868","project":"openstack/trove","branch":"master","topic":"add-guest-agent-logs","attention_set":{},"removed_from_attention_set":{"31737":{"account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"last_update":"2023-06-06 12:41:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I24695238075f9b1354ee4c54673d021de481a868","subject":"Add debug log for CI tests","status":"MERGED","created":"2023-06-06 01:50:46.000000000","updated":"2023-06-07 06:25:10.000000000","submitted":"2023-06-06 12:41:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"885326-add-guest-agent-logs","meta_rev_id":"046fbe6be1c01cae41aba607a6c29e217ccbab4b","_number":885326,"virtual_id_number":885326,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-06 12:41:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-06 10:19:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-06-06 10:19:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-06 02:02:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-06 02:20:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-06 05:54:25.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"state":"REVIEWER"}],"messages":[{"id":"bc441efee720e54f8b9424f17c5ddd08e2169932","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-06 01:50:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d57e962ee71459efef4624e0b2ef0609c9530c1d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 02:02:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e016ece94711443f9fbb8b77a87db31a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a7f858f8483042a8b60e0dd5214448a6 : SUCCESS in 9m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6e64c1aab6404a93ae17726500e3d1ba : SUCCESS in 7m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f994bf4117cb4a8bb761433432360efb : SUCCESS in 9m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6a8097b06ee81358824acc63693b7d910cb19894","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 02:20:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e874d30f786642f79140f91d6329a441\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/645d1df3117a47b692e85aa25313cf53 : FAILURE in 5m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a101bd9186a7401aacb4a74cc12dcdb6 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52967df1d6ed499389e9cbcac4ff3b40 : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f67935aae4e402ab2d54c359bce583d : SUCCESS in 5m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0f205f0a00ec4c4baca31769fdfd994f : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef54c2c154df45038556d538b63c7c37 : SUCCESS in 5m 17s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e0b0954f72374d5a97a50a9f10cc96af : SUCCESS in 5m 16s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/025479bf25e54d8196eea5a3ff2a09e0 : FAILURE in 28m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9b34450391c9f85b99cd60d8602a983eedc00035","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-06 02:52:33.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ee4b9f9fe9d689cb0653f9c8a6db37075c4322ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 03:04:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0f831ee50104ebd8659116fb227bf56\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/52a02796241c46df8d5fd370826e35e9 : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/26c30e66f15f49a78a8b124961baf6c1 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ecbdb169775e4d1387cbb57bb1a7fba6 : SUCCESS in 5m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8cc13b904f98573915464d79d7b56107bf9bc61e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 04:04:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ebaa7dd95284c6eab618c8f3416e726\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b269150502d450690b4076f1135df70 : FAILURE in 7m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/577084d4d2734799b9caa62861f8cb8f : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ac8ed173c64486882b71d7673038188 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/30cd4a409ffe48518a15444d41507260 : SUCCESS in 6m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e1538b6460bd459faab4b67cf2ef1f23 : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4243e0125bde4f2da54e21752f8e0af8 : SUCCESS in 5m 32s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/85d6e7e944b54d019afa8a2fc2935751 : SUCCESS in 5m 51s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/56f1bc04883c44f881dc9f57499922f1 : SUCCESS in 1h 10m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3dcef39fca39beed55cf880b2f8f18f135316061","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-06 05:54:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1f759ff632d254219fd1a489cf6dc0d5495a45f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-06 05:55:32.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"040323edd9dd1450024a0b1b68698caa1f043796","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 06:11:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67572e0b401746ccbb36d8cd360d8f4a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0b75921152ac4ba3bb8f0dcefd34e531 : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/174d254c95a740ef8765514d1847256d : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e4516a0e553d406bad2bee4c98dabd02 : SUCCESS in 12m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c82d79082a673d920da40d832b209df57c38bec4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 07:08:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5cae9c4261f4974a17034c6b30845e9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/465b2bf9014440e8982292fbcf63924c : FAILURE in 6m 19s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d611ab867d164565af181c48b2182cd2 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d23abc211f8e4e71bc71865142e90120 : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f3d8de9ae5a4243b1d0e0fd2e12ea64 : SUCCESS in 5m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cd97edff069c43e69599759689116cd7 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f7f0050f3eb41658bb0c2eac63ff0b8 : SUCCESS in 4m 57s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3b7417ff988640868bfe63481f03e1fd : SUCCESS in 4m 39s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/0ea5b169ab4f402b812d9867533a7044 : SUCCESS in 1h 11m 00s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/b32873052397419788ded3a16623775c : SUCCESS in 8m 11s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/7100e40f3e8049c89593822e2574ae7a : SUCCESS in 8m 48s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/57703e54f31c4df6b897f6e29ad6d776 : SUCCESS in 1h 09m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e203af23fa04e355548c7e9dc063bd2b26cf5cbd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-06 07:43:59.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"ba512e7d9dab401dfb7885b903175174efe6c839","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 07:58:29.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bf97e9ebe024525b46979735db90444\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3de8f87769204d62b83b6f88d1045492 : SUCCESS in 12m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/869d6b02e1b04bd5a1393f285951843a : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2322477606b8415d87cc6e5d68bb6949 : SUCCESS in 11m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"809d8f9fcfc43c4119fe22b82dda7d6fa5285c03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 09:27:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac584ad4a605442f87e1ab32ed39d072\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14f22177466540fe9ccde9a8b1d25069 : FAILURE in 7m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f798e2b2456b447f9c0f9716f6667550 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ee3ad9f45454775a25fb153f53e42e1 : SUCCESS in 6m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5aaa6ae26d12402b87de63483173921e : SUCCESS in 5m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d31f129fc3074525a3634a8901facd04 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32270515376a48c4bcd8453d0f123423 : SUCCESS in 5m 45s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/13bcdaa4ce004f64bb6875f7f9bd0c36 : SUCCESS in 5m 08s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/4497c4dca2cc4796bd9259cdc5d14b92 : SUCCESS in 1h 41m 45s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/4029f7a5f76a428b8bf331f427dc59dd : SUCCESS in 8m 13s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/39362cf87bf04c99ae40b49eccaaabbc : SUCCESS in 11m 41s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/68385a36119e44a78f5efc0ab913e634 : SUCCESS in 1h 39m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e37d42fd3c4be4902a1da404c8070ce9b3eb6b5e","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-06 10:19:50.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"1d3779b46be54f203924fbbe6c55f9095f800545","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 10:20:25.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"cb410c52bbf86c387b00e48ddb4f2922d681ae5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 12:41:29.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43640a0d93444688a1e291cd16e20445\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0889f0d9ac78449a834535d8fbadb341 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/423ac747216e418495b2049c48bb0591 : SUCCESS in 6m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f74eecf32924bcea6bd1de919ee01a4 : SUCCESS in 7m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de3421ba1886489c9168b446382bf824 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ded7a444bc9b43e5bf609a10ed2dcd47 : SUCCESS in 6m 15s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/ad273ba1f5dd4b5d9678f599da49b28c : SUCCESS in 1h 45m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ef6bd3b7adefe8da75875c627a8d1b10e474e77e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 12:41:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"ab227282240a3fb53697310ac62ba89d9d5653b1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 12:43:23.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/024da1fbf9874d9f903d36bf0f1a2f9c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ef16715f31d496a8b1ee57e3a9096ed : SUCCESS in 38s","accounts_in_message":[],"_revision_number":4},{"id":"046fbe6be1c01cae41aba607a6c29e217ccbab4b","tag":"autogenerated:gerrit:revert","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-07 06:25:10.000000000","message":"Created a revert of this change as I398dd7ac8c628d19b75693dcac0521346e8f5d64","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"49c16b889b68039b068eb420f345efdfaafcc1cc","revisions":{"5fca2f5bc5fd5edd49bbeec489e51b437be2b74b":{"kind":"REWORK","_number":1,"created":"2023-06-06 01:50:46.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/26/885326/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/26/885326/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/885326/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/885326/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/885326/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/26/885326/1"}}},"commit":{"parents":[{"commit":"e854cb58538f83d7a4538218b27d8ad25d865f35","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e854cb58538f83d7a4538218b27d8ad25d865f35"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-06 01:49:05.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-06 01:49:05.000000000","tz":480},"subject":"Add debug log for CI tests","message":"Add debug log for CI tests\n\nWhen the CI fails, specifically related to the Trove guest agent,\nwe are unable to find a way to debug it.\nthis patch synchronizes the guest agent logs to controller nodes.\n\nChange-Id: I24695238075f9b1354ee4c54673d021de481a868\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5fca2f5bc5fd5edd49bbeec489e51b437be2b74b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5fca2f5bc5fd5edd49bbeec489e51b437be2b74b"}]},"branch":"refs/heads/master"},"1b75b7f151b4dfa7ad38212743c30942f922c2d5":{"kind":"REWORK","_number":2,"created":"2023-06-06 02:52:33.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/26/885326/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/26/885326/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/885326/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/885326/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/885326/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/26/885326/2"}}},"commit":{"parents":[{"commit":"e854cb58538f83d7a4538218b27d8ad25d865f35","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e854cb58538f83d7a4538218b27d8ad25d865f35"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-06 01:49:05.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-06 02:52:16.000000000","tz":480},"subject":"Add debug log for CI tests","message":"Add debug log for CI tests\n\nWhen the CI fails, specifically related to the Trove guest agent,\nwe are unable to find a way to debug it.\nthis patch synchronizes the guest agent logs to controller nodes.\n\nChange-Id: I24695238075f9b1354ee4c54673d021de481a868\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1b75b7f151b4dfa7ad38212743c30942f922c2d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1b75b7f151b4dfa7ad38212743c30942f922c2d5"}]},"branch":"refs/heads/master"},"878daaa45037e4900f5ea0019f72280bf7c5fe37":{"kind":"REWORK","_number":3,"created":"2023-06-06 05:55:32.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/26/885326/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/26/885326/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/885326/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/885326/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/885326/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/26/885326/3"}}},"commit":{"parents":[{"commit":"e854cb58538f83d7a4538218b27d8ad25d865f35","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e854cb58538f83d7a4538218b27d8ad25d865f35"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-06 01:49:05.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-06 05:55:18.000000000","tz":480},"subject":"Add debug log for CI tests","message":"Add debug log for CI tests\n\nWhen the CI fails, specifically related to the Trove guest agent,\nwe are unable to find a way to debug it.\nthis patch synchronizes the guest agent logs to controller nodes.\n\nChange-Id: I24695238075f9b1354ee4c54673d021de481a868\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/878daaa45037e4900f5ea0019f72280bf7c5fe37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/878daaa45037e4900f5ea0019f72280bf7c5fe37"}]},"branch":"refs/heads/master"},"49c16b889b68039b068eb420f345efdfaafcc1cc":{"kind":"REWORK","_number":4,"created":"2023-06-06 07:43:59.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/26/885326/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/26/885326/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/885326/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/885326/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/885326/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/26/885326/4"}}},"commit":{"parents":[{"commit":"e854cb58538f83d7a4538218b27d8ad25d865f35","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e854cb58538f83d7a4538218b27d8ad25d865f35"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-06 01:49:05.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-06 07:43:44.000000000","tz":480},"subject":"Add debug log for CI tests","message":"Add debug log for CI tests\n\nWhen the CI fails, specifically related to the Trove guest agent,\nwe are unable to find a way to debug it.\nthis patch synchronizes the guest agent logs to controller nodes.\n\nChange-Id: I24695238075f9b1354ee4c54673d021de481a868\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/49c16b889b68039b068eb420f345efdfaafcc1cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/49c16b889b68039b068eb420f345efdfaafcc1cc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
