)]}'
{"id":"openstack%2Ftrove~885373","triplet_id":"openstack%2Ftrove~master~I3595d6860da69ddf3c34f715ea9f6c6cae876264","project":"openstack/trove","branch":"master","hashtags":[],"change_id":"I3595d6860da69ddf3c34f715ea9f6c6cae876264","subject":"[CI] Add --backup flag for rsync","status":"ABANDONED","created":"2023-06-06 16:14:42.000000000","updated":"2023-06-07 02:30:21.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6a8545890fc3cba5f3e049c8d983fd1c62b078e2","_number":885373,"virtual_id_number":885373,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-06-06 17:57:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-06 16:34:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-06 17:57:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4d47b1ec2dc4fff6d584daef2b35363eb0631703","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-06 16:14:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87e6d58e232c16123165114a4061c820732de99e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 16:34:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af84412db84141c687323db5916be8ef\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/11c7876fa79c478f9bda44b7d1ab313b : SUCCESS in 12m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/46cd91bb265742c6a355b937ec598b33 : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/282ee239ed4049efa1857fb783a5a052 : SUCCESS in 6m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9588357fe81f4c44d70319fee96c12e7ca008758","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 17:57:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de5d5e53a32940c89140627aa2149553\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c6808027d8c4f5ca0bd302d5e45c8e4 : FAILURE in 6m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f82419d3b2e149e0bb2eee1506617ce7 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b188c4e713b4f33b29fcd8b3a1bc386 : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d92903f4a48c44bfb6694ce7e69a9976 : SUCCESS in 5m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a213dec909ad4102aa0ffd78abdd587c : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13da63f95f2c4669ab4843e9b747f7ca : SUCCESS in 5m 42s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f72ac5dc31a94598bc6375e159d8aed2 : SUCCESS in 4m 45s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/d9dfa6cbb7884d3e8b269c71a3ccf899 : SUCCESS in 1h 39m 20s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/62121cc981ed40419bcfb419dc0c3a3c : SUCCESS in 9m 08s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/86b3a3fa6dfa4a92a33e0d44dd443bce : SUCCESS in 10m 34s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/043fafeff97842ef942b5867ff3a44cb : SUCCESS in 1h 17m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6a8545890fc3cba5f3e049c8d983fd1c62b078e2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-07 02:30:21.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e78aa78b9130547e003ceb3f3e4ad749f0fb8d7c","revisions":{"e78aa78b9130547e003ceb3f3e4ad749f0fb8d7c":{"kind":"REWORK","_number":1,"created":"2023-06-06 16:14:42.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/73/885373/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/73/885373/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/73/885373/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/73/885373/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/73/885373/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/73/885373/1"}}},"commit":{"parents":[{"commit":"79c8d0caa0f3de2da002ed57c7c23af84cf91ff1","subject":"Merge \"Don\u0027t ignore H306 pep8 check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/79c8d0caa0f3de2da002ed57c7c23af84cf91ff1"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-06 16:13:58.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-06 16:13:58.000000000","tz":480},"subject":"[CI] Add --backup flag for rsync","message":"[CI] Add --backup flag for rsync\n\nChange-Id: I3595d6860da69ddf3c34f715ea9f6c6cae876264\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e78aa78b9130547e003ceb3f3e4ad749f0fb8d7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e78aa78b9130547e003ceb3f3e4ad749f0fb8d7c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
