)]}'
{"id":"openstack%2Ftrove~885407","triplet_id":"openstack%2Ftrove~master~I398dd7ac8c628d19b75693dcac0521346e8f5d64","project":"openstack/trove","branch":"master","topic":"add-guest-agent-logs","attention_set":{},"removed_from_attention_set":{"31737":{"account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"last_update":"2023-06-07 10:38:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I398dd7ac8c628d19b75693dcac0521346e8f5d64","subject":"Revert \"Add debug log for CI tests\"","status":"MERGED","created":"2023-06-07 06:25:10.000000000","updated":"2023-06-07 10:40:03.000000000","submitted":"2023-06-07 10:38:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"revert_of":885326,"submission_id":"885407-add-guest-agent-logs","meta_rev_id":"f45e0d8f68d0961f504ae94802499c04df13408c","_number":885407,"virtual_id_number":885407,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-07 10:38:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-07 08:31:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-06-07 08:31:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-07 06:25:10.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"state":"REVIEWER"},{"updated":"2023-06-07 06:25:10.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"690e250f86d3ae05633e8784918562fa953c2507","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-07 06:25:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"45ab305145f2f79b6581845d0dbea54b195b0f2c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 06:43:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8a3635d02164890a8b2877e0d0c3a49\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/df18bc1798834cff8a4d7985cf0e5680 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2a696b644af94443aaad730b398a9b5e : SUCCESS in 16m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6d1272c4233e4392bb188bfbedea5e83 : SUCCESS in 4m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f7c2abaec11dc59603a33abe2b38bc7c769f993c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 07:47:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34b2a79227d14b99a9112106a6fba915\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d889962d2bed463eba9099c2a8d72fba : FAILURE in 7m 32s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b03d75ca9414e8f86e919ba0b7b8651 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98d2c27b709b47bb8792693e693c29a2 : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/465cd54ac1984b02b4ab67b174a3e6d3 : SUCCESS in 6m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f5a0ba6b5b78463eabaa93238cfa7257 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/228e962433eb4100952e65d140c69bd7 : SUCCESS in 5m 25s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9e06bf903f444814aa9f450b6915e397 : SUCCESS in 5m 17s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/568f1d3122724847ad9923c91e56b81e : SUCCESS in 1h 20m 49s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/31fc24a5e7f842b48cf407202ff4c53d : SUCCESS in 10m 35s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/2b6f31767f5d4fdc961a374e6f9ed81a : SUCCESS in 14m 57s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/cf0fb9ab508745649cd423882d94b74b : SUCCESS in 1h 13m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"27ff29fae54c98f8db227b2a95ef0046aeab9f22","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-07 08:31:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bf4b4c252a98f57d545c28bc1756ea8287062f28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 08:31:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6a828a6a42c98d6021da436d07b9339a6aa802ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 10:38:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7e2aa6b48894effae3a91973cf3497a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cc763efa01c48dfae33a1c585743738 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bbf9f231812b45748d27da73711903d3 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ecd0373e9ab4fd388e98c036abb0964 : SUCCESS in 5m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7b2cb5856c224e71af7e654db38f8047 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/796446ca7a7d4974badcd4d86c4e8350 : SUCCESS in 5m 32s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/7550d96bc91c466095398f307b0a229f : SUCCESS in 2h 02m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d6509776d221eb9bd6c89672edc806775e1c2c29","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 10:38:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f45e0d8f68d0961f504ae94802499c04df13408c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 10:40:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07135a606d864b5f92182270faeee473\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f80099456924727adcd212af5e9e182 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"06c8ed94ab708c8f7a962178e8bc746d4b36762f","revisions":{"06c8ed94ab708c8f7a962178e8bc746d4b36762f":{"kind":"REWORK","_number":1,"created":"2023-06-07 06:25:10.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/07/885407/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/07/885407/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/07/885407/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/07/885407/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/07/885407/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/07/885407/1"}}},"commit":{"parents":[{"commit":"49c16b889b68039b068eb420f345efdfaafcc1cc","subject":"Add debug log for CI tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/49c16b889b68039b068eb420f345efdfaafcc1cc"}]}],"author":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-06-07 06:25:10.000000000","tz":0},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-06-07 06:25:10.000000000","tz":0},"subject":"Revert \"Add debug log for CI tests\"","message":"Revert \"Add debug log for CI tests\"\n\nThis reverts commit 49c16b889b68039b068eb420f345efdfaafcc1cc.\n\nReason for revert: this patch doesn\u0027t work as expected\n\nChange-Id: I398dd7ac8c628d19b75693dcac0521346e8f5d64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/06c8ed94ab708c8f7a962178e8bc746d4b36762f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/06c8ed94ab708c8f7a962178e8bc746d4b36762f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
