)]}'
{"id":"openstack%2Ftrove~885452","triplet_id":"openstack%2Ftrove~master~I264235ce7b7df585358654c46bdac4dcd743fb3d","project":"openstack/trove","branch":"master","hashtags":[],"change_id":"I264235ce7b7df585358654c46bdac4dcd743fb3d","subject":"synchronize guest logs to controller node","status":"MERGED","created":"2023-06-07 09:53:51.000000000","updated":"2023-06-13 13:14:39.000000000","submitted":"2023-06-13 13:12:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"885452","meta_rev_id":"d025e69e24526abf83bcf9195a1c5d06ecfa0f24","_number":885452,"virtual_id_number":885452,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-13 13:12:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-13 11:12:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-06-13 11:12:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-07 10:01:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-07 12:42:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6813fe44528b54ff4aced8f1c2ee94e8221ef1d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-07 09:53:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3bc38fb4cbfc6369d4995e7a994e1ccbc72143f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 10:01:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/743caf5f96ad4584a16944d4176710ea\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/24cfb9e02088408cb1adbb2883da49aa : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a249b1d61a6f44448d6a10fdfb801102 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3a05d9190c26423f925538dae81d6a4e : SUCCESS in 5m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"afb593d121efd447fce3c233d40c9f2cca01e34a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-07 11:25:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7395c270adfd69f5de1388163d3cc80e2bb15fbc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 11:32:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34943b23ec884643ac1d45f9375d6412\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b6cf9063893b432ab1dc3057098cad32 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fb8c6747e0b7433a9c07caa1ddca0e81 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ad1197a4544040dbbd63d23bb652f63d : SUCCESS in 5m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7cd2b611d4255ae190ed7cae1911ac93e491d92d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 12:42:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6df6fcef59dd4e1cba1c6d8e7b330f66\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35781a5952544a069faf99254c1bd18b : FAILURE in 7m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3e3fc0866aa43ccb7d7f91bd078d790 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f3f78cd304e48688b4b86cd6d4d190e : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/951bda9c5a684eff80edf48332b7dcdd : SUCCESS in 6m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eed8cece80e74323bbed3545ba50c0f0 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aad45b3623d14a1a96f58bb2ce775216 : SUCCESS in 4m 56s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ab7fc99988b34eeeaa50e3578d906fe3 : SUCCESS in 5m 16s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/8cc2ef8d6f8548f2beb2acf1837ad5c3 : SUCCESS in 1h 14m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"db80e6ac439549d84536b3ac95f31d207b085a1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-07 13:17:20.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"4e6dc6d1b4f0d7e3748f9a9adb18eadbeaab8de0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 13:25:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d09a045e59ff416f92463b42e299779b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4dd2e49b49a947d08a75f1e9c61857d8 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a961eacf40ce4e7bb807d7e717f2b076 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d42a715bf3dd4731b19ed941c3cd6374 : SUCCESS in 6m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7e1f351b1a2e61de706665e2461dfdcb876b8e5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 15:03:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99231cc3611d452793027b0db892b527\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01b82d1be0464691a4dfc7f2fc48ecd6 : FAILURE in 7m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0672ded93bf64b0bbe17419cab781491 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5ab8c4eabbf476591fd6c6fbc81ad4f : SUCCESS in 6m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01c095cf1d504e5cad01185f2173cfe9 : SUCCESS in 5m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2e049b1888eb46d485898062a07e385f : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebbeb55ba8d34343b667fe8f138f612c : SUCCESS in 5m 51s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cd5909a7afa143e89ce875e29b4dd70c : SUCCESS in 4m 25s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/039b843523924225a9f3c11cffe4083c : SUCCESS in 1h 44m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2dab0232a781b579e618fe007c971a615f7c88f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-07 15:34:50.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"9ed9e604cfb5ec41729254e15f8bef4f055b7cc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 17:08:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc1bd6363df74e43b3a75f8bafb71d20\n\n- trove-tempest https://zuul.opendev.org/t/openstack/build/7714bfd565554f1ab324bcdd00026816 : SUCCESS in 1h 32m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e2504a2e213db724c6cf728da504eff416b7c8c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-08 02:44:16.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"d0d9728013446cd5a5de7732336bf90a40d712c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-08 03:54:29.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9476c87227d94c888e04b28d134d216f\n\n- trove-tempest https://zuul.opendev.org/t/openstack/build/ebc4e20bda1f48a19f6f5971328bedcd : SUCCESS in 1h 08m 15s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"603cdeb75804afa789b51603a50a33f14eda17ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-08 07:06:26.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"960c8feb48fcd45cb1d6c9f05316e712176e4d41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-08 09:16:01.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fcb1235e1edd42559cdc95a344e14b17\n\n- trove-tempest https://zuul.opendev.org/t/openstack/build/5b593bba1357483eb7f13146218a1fd5 : FAILURE in 2h 02m 14s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d5aaa698038707a48da88a262e12e389fe4f8b62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-08 15:44:55.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"e2eb079eb3f1973189896ffcba38a7c19a695f40","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-08 16:02:42.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88c0333cef7448229e915a19d63512c0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/276b76d8ec8b440d83ce7d6b19ad703c : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b2bf6148a6634377860a5839e6153a84 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/892f6553d746484da13f0c701646b8ec : SUCCESS in 17m 13s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"07c713139475ff218f12868e572723a3ff83eb2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-08 17:01:28.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17dd156f99a743a2a06e894badc582e0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc13973ed6774fd49e4e30315968824e : FAILURE in 7m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5c174e2198a4f2d834e5b7698c8ddbc : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b7b571655804c6c95a2e11a93d760d4 : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce1642136b744865acd67b7b383b79e2 : SUCCESS in 5m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4c1a53fddbc744369b73cfa3ed2d3597 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1415af6710cf4d998925b546664810f1 : SUCCESS in 5m 05s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a4fdf4ac724f41d49fb03fd1aa96c103 : SUCCESS in 5m 25s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/922c79517c764b19b78124e6739a2c33 : SUCCESS in 1h 13m 50s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/22f8939800814917b7841afbefda35d6 : SUCCESS in 8m 54s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/7a4d7fe1325f4d238a41a0cb6eeadd5f : SUCCESS in 9m 33s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/79b8e4fc98ac45a597644cbabb69590a : SUCCESS in 1h 08m 26s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"fa9ba6d167f2a02277501b2f6f9704a68bb9c2f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-09 02:09:04.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"bb4cdd5036be11bfe94d12215d3ccca46ea69bac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-09 02:20:07.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3480bdc09a2e4c1f950c95f6ddd035e0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d6730eaba4e848748bc42682afde70aa : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/38d897ea95484bd180d0d87c86a6586e : NODE_FAILURE Node request 300-0021427858 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ac744cd8784d4811a7911404eaeb677e : SUCCESS in 6m 16s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"91b838af63b57d87abb90897080f14e4bc4c2aa3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-09 04:44:26.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2592fb10842b41f7a40666a3666041d6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/085283d0f2ed428aa96e9c75cfbc0a55 : FAILURE in 6m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c69cf3a0206146f8bce592bdbf988440 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3dedc3c2ad6f4a8dab3afdf5591b6ed8 : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/14a11ea906bb412a922b67f255a667f9 : SUCCESS in 7m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2944a2b2fc564925966393ee3fd41ade : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a42e08cdd52747aca5697e9310771f51 : SUCCESS in 5m 33s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5be762cdf4244196b9f1253c2b3062a8 : SUCCESS in 8m 27s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/74bbd052c1a548a28b83b0cedb3a87db : SUCCESS in 1h 23m 08s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/8e5857a4557c4a12b46f3be3a4d001bf : SUCCESS in 9m 38s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/11e8b78241eb4304ba8d8af140964e99 : SUCCESS in 11m 01s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/a2b8348a8f464ce2be0de97d51b0a5ae : SUCCESS in 1h 56m 29s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"5770ff603022edd5c7cb13b13342be0bd46486d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-09 15:00:48.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"cca9d38b65c720f3a418e506c446fa5d52d8a73f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-09 15:17:43.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61fd172be43d4ba584d4c7bc3c85e4af\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1baed283163741af83d908736f61671a : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dafcaa3f81e64c6eb673046b12ed8faa : SUCCESS in 14m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/79c2efa0e3dc42768245daee507d0d5b : SUCCESS in 5m 32s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"7639cd644c091c59c45af203ace8deb39efe9b34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-09 17:16:21.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f538787d43f04fd590fa2fbf6b590e85\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c14f5bfb01f4988a652ee93e3f451fd : FAILURE in 7m 32s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2a2486a4556433eb251cf100478d820 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ec7be1fae5041878f2565ab363e3709 : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/117af08cab16414a9ec7a195e3aade3d : SUCCESS in 6m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/711e454ca1d847efb0b9eb4624a1ba08 : SUCCESS in 7m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13072ef862b84238b4b79cc90d97e3b1 : SUCCESS in 5m 40s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2549b6f84fa04e4594145c94473f48ac : SUCCESS in 4m 55s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/c402a43e9d7f45c7b21345e790f40a0b : FAILURE in 2h 12m 42s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/9e2e54098fa34d888803c2c3f6946b21 : SUCCESS in 9m 26s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/831d8917a25b40ae84a70c00dbaca7ce : SUCCESS in 11m 09s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/6db01b2eba79459bb0794ff46565a7e7 : SUCCESS in 1h 40m 07s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"52170a5271bc72d91977c6c5c63a593a7838a421","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-10 00:52:57.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"3bb08d6c7222a08d988c8f2165fa7072039ed620","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-10 01:02:34.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57f3d5bd383c48598acce6e7ec4e2bc3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9d8c76ab404249a8a104ba486506387d : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d420aa2e0ea04ca781d983aa6a04a1be : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2a15b99c069749bc8740cc76a3104cce : SUCCESS in 6m 58s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"71a5297c8bbd687bc78555a89c047f3b5ec007f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-10 02:07:28.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f6229af5c494bd9b2ca03f4acec01d9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f172b398d2d47eb9db3fda7bbc03e20 : FAILURE in 7m 01s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bd7fc4886374150af09e63bf08dc4fe : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9bb0eabf5cc4fc7b37d69bc8f1332e4 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b95b91fc61be467d864392c59aa08e56 : SUCCESS in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0028a1f4f0e94111881538398ec1d81c : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/086b46cb4f224495bcff9b6a82ecfaf9 : SUCCESS in 4m 42s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/af4e195e195d4def8161b3424328b823 : SUCCESS in 4m 13s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/75746d8dd0ae4e08bb058e16914aec28 : SUCCESS in 1h 13m 08s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/9feaf3d0d10141ae8a9ca9bb3cb84730 : SUCCESS in 8m 34s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/ab93b55f6e3a4723834216446580f1cd : SUCCESS in 10m 37s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/3cb744b9755a49f59d377bd0b51232df : SUCCESS in 1h 09m 48s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"6c4c4b5930c9d5015dbba079d13931fb1a6f1ba9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-13 02:39:44.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"5cb26bc88741713762e7db6112ee948943c0d9cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 03:02:40.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbce32d5a5294c4bb302d14a1392ee12\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/45eca6c4ecf94d0baec3197da259057c : SUCCESS in 19m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d0dfc3e8daff44a3b0e33e1b8028630a : SUCCESS in 18m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/13330747ac7a44aea041ba23242d6069 : SUCCESS in 7m 52s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"801d59aa6cc5b2e9bb2bd8b7afbafbe1a6d8f26a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 03:56:00.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/219e8c0296494d96abdc151e963c975f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87e0d50d4e434f6eb0bd9aa029e23a74 : FAILURE in 6m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7e2f86125a14160bc77f1f7f8e47b72 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8499b52765f440b9686ec0496eee41b : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8733ef065b7b4f6b820860367c3a0176 : SUCCESS in 5m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bfb3d5319cd54f01803af13241dd49f6 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f47f5c06b9349f18b822d08999de390 : SUCCESS in 5m 18s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5078d7db33e144fcb4e4158763adcc57 : SUCCESS in 5m 28s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/9505f390f82e406db338e9075ea0261f : SUCCESS in 1h 14m 11s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/9b9fe44d24264052903d7b31ab1d968d : SUCCESS in 9m 42s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/afc043d3f9ff489c8c9322a56e20f7e4 : SUCCESS in 10m 55s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/663aad027b7b4cc4b0917855f60179c9 : SUCCESS in 1h 10m 22s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"04734fbb5b39eb862eb5c4f52bcea52d37046247","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-13 11:12:56.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"10ab37b94c879646cd189dbe61e76f2d0c34336b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 11:13:31.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"a40b2d88b7d17989d99b341ed6c46e8ebe63eebd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 13:12:53.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a98e11749d5472e86dbf40e02084ce5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85100de07d084edf9d229be2f5ed864f : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05afc1af898842efaa5da6a11ab597be : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a3a25d8ac3b4f3ca864863d918991fd : SUCCESS in 5m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3c124521a1a243bca10a59f0299e436f : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/421d8005de5c4714b69c1babd80992ef : SUCCESS in 6m 05s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/2a5f00a6d05741d792b92b8c6446fa57 : FAILURE in 1h 54m 04s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"4a6859e75d913098beb78078e7a22b4fe9978c1a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 13:12:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"d025e69e24526abf83bcf9195a1c5d06ecfa0f24","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 13:14:39.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fabb52aad8c3436d89b169ecba7b781f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45e7e32690d84a1b9ce20fc50ec951ab : SUCCESS in 36s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"09fd606163017aaa4e09e577da7080aff5a1d774","revisions":{"f53d2456e0fc78e13cf7ac397a824d21be3d8c2c":{"kind":"REWORK","_number":1,"created":"2023-06-07 09:53:51.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/52/885452/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/52/885452/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/885452/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/885452/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/885452/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/52/885452/1"}}},"commit":{"parents":[{"commit":"79c8d0caa0f3de2da002ed57c7c23af84cf91ff1","subject":"Merge \"Don\u0027t ignore H306 pep8 check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/79c8d0caa0f3de2da002ed57c7c23af84cf91ff1"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-07 09:44:21.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-07 09:44:21.000000000","tz":480},"subject":"synchronize guest logs to controller node","message":"synchronize guest logs to controller node\n\nuse rsync to synchronize guest logs to the controller node so that we can\nfind the logs even after the instance is deleted\n\nAdd a guest-log-collection timer, this timers triggers\nguest-log-collection service every 10 secondes\n\nChange-Id: I264235ce7b7df585358654c46bdac4dcd743fb3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f53d2456e0fc78e13cf7ac397a824d21be3d8c2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f53d2456e0fc78e13cf7ac397a824d21be3d8c2c"}]},"branch":"refs/heads/master"},"18c26cf083f902cc9e766a3a532e68ee159e28a4":{"kind":"REWORK","_number":2,"created":"2023-06-07 11:25:46.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/52/885452/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/52/885452/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/885452/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/885452/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/885452/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/52/885452/2"}}},"commit":{"parents":[{"commit":"58b24ad44fd24cb5481575040e23cfd7d45a3eee","subject":"Merge \"Revert \"Add debug log for CI tests\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/58b24ad44fd24cb5481575040e23cfd7d45a3eee"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-07 09:44:21.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-07 11:24:42.000000000","tz":480},"subject":"synchronize guest logs to controller node","message":"synchronize guest logs to controller node\n\nuse rsync to synchronize guest logs to the controller node so that we can\nfind the logs even after the instance is deleted\n\nAdd a guest-log-collection timer, this timers triggers\nguest-log-collection service every 10 secondes\n\nChange-Id: I264235ce7b7df585358654c46bdac4dcd743fb3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/18c26cf083f902cc9e766a3a532e68ee159e28a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/18c26cf083f902cc9e766a3a532e68ee159e28a4"}]},"branch":"refs/heads/master"},"4aeabf043d9fa15cae167355e36ba867c3132558":{"kind":"REWORK","_number":3,"created":"2023-06-07 13:17:20.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/52/885452/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/52/885452/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/885452/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/885452/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/885452/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/52/885452/3"}}},"commit":{"parents":[{"commit":"58b24ad44fd24cb5481575040e23cfd7d45a3eee","subject":"Merge \"Revert \"Add debug log for CI tests\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/58b24ad44fd24cb5481575040e23cfd7d45a3eee"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-07 09:44:21.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-07 13:16:51.000000000","tz":480},"subject":"synchronize guest logs to controller node","message":"synchronize guest logs to controller node\n\nuse rsync to synchronize guest logs to the controller node so that we can\nfind the logs even after the instance is deleted\n\nAdd a guest-log-collection timer, this timer triggers\nguest-log-collection service every 10 seconds\n\nChange-Id: I264235ce7b7df585358654c46bdac4dcd743fb3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4aeabf043d9fa15cae167355e36ba867c3132558"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4aeabf043d9fa15cae167355e36ba867c3132558"}]},"branch":"refs/heads/master"},"c484c54ce99c444ed44acc2eb879082af77a896f":{"kind":"REWORK","_number":4,"created":"2023-06-07 15:34:50.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/52/885452/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/52/885452/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/885452/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/885452/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/885452/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/52/885452/4"}}},"commit":{"parents":[{"commit":"58b24ad44fd24cb5481575040e23cfd7d45a3eee","subject":"Merge \"Revert \"Add debug log for CI tests\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/58b24ad44fd24cb5481575040e23cfd7d45a3eee"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-07 09:44:21.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-07 15:34:34.000000000","tz":480},"subject":"synchronize guest logs to controller node","message":"synchronize guest logs to controller node\n\nuse rsync to synchronize guest logs to the controller node so that we can\nfind the logs even after the instance is deleted\n\nAdd a guest-log-collection timer, this timer triggers\nguest-log-collection service every 10 seconds\n\nChange-Id: I264235ce7b7df585358654c46bdac4dcd743fb3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c484c54ce99c444ed44acc2eb879082af77a896f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c484c54ce99c444ed44acc2eb879082af77a896f"}]},"branch":"refs/heads/master"},"00f6e326d89fb9ee3769b4be991f1402349951ed":{"kind":"REWORK","_number":5,"created":"2023-06-08 02:44:16.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/52/885452/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/52/885452/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/885452/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/885452/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/885452/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/52/885452/5"}}},"commit":{"parents":[{"commit":"58b24ad44fd24cb5481575040e23cfd7d45a3eee","subject":"Merge \"Revert \"Add debug log for CI tests\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/58b24ad44fd24cb5481575040e23cfd7d45a3eee"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-07 09:44:21.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-08 02:44:00.000000000","tz":480},"subject":"synchronize guest logs to controller node","message":"synchronize guest logs to controller node\n\nuse rsync to synchronize guest logs to the controller node so that we can\nfind the logs even after the instance is deleted\n\nAdd a guest-log-collection timer, this timer triggers\nguest-log-collection service every 10 seconds\n\nChange-Id: I264235ce7b7df585358654c46bdac4dcd743fb3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/00f6e326d89fb9ee3769b4be991f1402349951ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/00f6e326d89fb9ee3769b4be991f1402349951ed"}]},"branch":"refs/heads/master"},"e6a7f714165b015a8cda6c128bf69887a83954ab":{"kind":"REWORK","_number":6,"created":"2023-06-08 07:06:26.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/52/885452/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/52/885452/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/885452/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/885452/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/885452/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/52/885452/6"}}},"commit":{"parents":[{"commit":"58b24ad44fd24cb5481575040e23cfd7d45a3eee","subject":"Merge \"Revert \"Add debug log for CI tests\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/58b24ad44fd24cb5481575040e23cfd7d45a3eee"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-07 09:44:21.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-08 07:06:07.000000000","tz":480},"subject":"synchronize guest logs to controller node","message":"synchronize guest logs to controller node\n\nuse rsync to synchronize guest logs to the controller node so that we can\nfind the logs even after the instance is deleted\n\nAdd a guest-log-collection timer, this timer triggers\nguest-log-collection service every 10 seconds\n\nChange-Id: I264235ce7b7df585358654c46bdac4dcd743fb3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e6a7f714165b015a8cda6c128bf69887a83954ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e6a7f714165b015a8cda6c128bf69887a83954ab"}]},"branch":"refs/heads/master"},"aa728d5d5e555ff1e41fe60139a31d5b784cda4f":{"kind":"REWORK","_number":7,"created":"2023-06-08 15:44:55.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/52/885452/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/52/885452/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/885452/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/885452/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/885452/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/52/885452/7"}}},"commit":{"parents":[{"commit":"58b24ad44fd24cb5481575040e23cfd7d45a3eee","subject":"Merge \"Revert \"Add debug log for CI tests\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/58b24ad44fd24cb5481575040e23cfd7d45a3eee"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-07 09:44:21.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-08 15:43:55.000000000","tz":480},"subject":"synchronize guest logs to controller node","message":"synchronize guest logs to controller node\n\nuse rsync to synchronize guest logs to the controller node so that we can\nfind the logs even after the instance is deleted\n\nAdd a guest-log-collection timer, this timer triggers\nguest-log-collection service every 10 seconds\n\nChange-Id: I264235ce7b7df585358654c46bdac4dcd743fb3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/aa728d5d5e555ff1e41fe60139a31d5b784cda4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/aa728d5d5e555ff1e41fe60139a31d5b784cda4f"}]},"branch":"refs/heads/master"},"d75ee097558d088f9ef0f6132d8343ab85bfd029":{"kind":"REWORK","_number":8,"created":"2023-06-09 02:09:04.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/52/885452/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/52/885452/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/885452/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/885452/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/885452/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/52/885452/8"}}},"commit":{"parents":[{"commit":"58b24ad44fd24cb5481575040e23cfd7d45a3eee","subject":"Merge \"Revert \"Add debug log for CI tests\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/58b24ad44fd24cb5481575040e23cfd7d45a3eee"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-07 09:44:21.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-09 02:08:47.000000000","tz":480},"subject":"synchronize guest logs to controller node","message":"synchronize guest logs to controller node\n\nuse rsync to synchronize guest logs to the controller node so that we can\nfind the logs even after the instance is deleted\n\nAdd a guest-log-collection timer, this timer triggers\nguest-log-collection service every 10 seconds\n\nChange-Id: I264235ce7b7df585358654c46bdac4dcd743fb3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d75ee097558d088f9ef0f6132d8343ab85bfd029"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d75ee097558d088f9ef0f6132d8343ab85bfd029"}]},"branch":"refs/heads/master"},"6d4dc2dee90d6ce33003540daad108f92a21fb31":{"kind":"REWORK","_number":9,"created":"2023-06-09 15:00:48.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/52/885452/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/52/885452/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/885452/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/885452/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/885452/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/52/885452/9"}}},"commit":{"parents":[{"commit":"58b24ad44fd24cb5481575040e23cfd7d45a3eee","subject":"Merge \"Revert \"Add debug log for CI tests\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/58b24ad44fd24cb5481575040e23cfd7d45a3eee"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-07 09:44:21.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-09 15:00:29.000000000","tz":480},"subject":"synchronize guest logs to controller node","message":"synchronize guest logs to controller node\n\nuse rsync to synchronize guest logs to the controller node so that we can\nfind the logs even after the instance is deleted\n\nAdd a guest-log-collection timer, this timer triggers\nguest-log-collection service every 10 seconds\n\nChange-Id: I264235ce7b7df585358654c46bdac4dcd743fb3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6d4dc2dee90d6ce33003540daad108f92a21fb31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6d4dc2dee90d6ce33003540daad108f92a21fb31"}]},"branch":"refs/heads/master"},"09fd606163017aaa4e09e577da7080aff5a1d774":{"kind":"REWORK","_number":10,"created":"2023-06-13 02:39:44.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/52/885452/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/52/885452/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/885452/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/885452/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/885452/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/52/885452/10"}}},"commit":{"parents":[{"commit":"58b24ad44fd24cb5481575040e23cfd7d45a3eee","subject":"Merge \"Revert \"Add debug log for CI tests\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/58b24ad44fd24cb5481575040e23cfd7d45a3eee"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-07 09:44:21.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-13 02:39:25.000000000","tz":480},"subject":"synchronize guest logs to controller node","message":"synchronize guest logs to controller node\n\nuse rsync to synchronize guest logs to the controller node so that we can\nfind the logs even after the instance is deleted\n\nAdd a guest-log-collection timer, this timer triggers\nguest-log-collection service every 10 seconds\n\nChange-Id: I264235ce7b7df585358654c46bdac4dcd743fb3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/09fd606163017aaa4e09e577da7080aff5a1d774"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/09fd606163017aaa4e09e577da7080aff5a1d774"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
