)]}'
{"id":"openstack%2Ftrove~886044","triplet_id":"openstack%2Ftrove~master~I8156bb87ce3e4d1a19d2df9e2c3f67f789bffa46","project":"openstack/trove","branch":"master","topic":"enable-e125-extension","attention_set":{},"removed_from_attention_set":{"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2023-06-15 20:48:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8156bb87ce3e4d1a19d2df9e2c3f67f789bffa46","subject":"Don\u0027t ignore E125 check","status":"MERGED","created":"2023-06-14 06:36:11.000000000","updated":"2023-06-15 20:49:22.000000000","submitted":"2023-06-15 20:48:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886044-enable-e125-extension","meta_rev_id":"3fd287865674cc89b506137222a92168beb2e38a","_number":886044,"virtual_id_number":886044,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-15 20:48:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"all":[{"value":2,"date":"2023-06-15 18:39:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"all":[{"value":1,"date":"2023-06-15 18:39:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-14 06:51:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-14 09:46:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-15 18:39:53.000000000","updated_by":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"reviewer":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"state":"REVIEWER"}],"messages":[{"id":"838508fca94cb4fdb1f377d2d0f72e6d26adc696","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-14 06:36:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd35e63e651ae1a1d9f756ca1efa7fdd22f2541a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 06:51:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2fafacc741345808a239cef1ea4bebc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3618f5f2edec4f7ba9d3095207c8f04a : SUCCESS in 12m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6a748539fe6d4ba5886dc0a9a07bc025 : SUCCESS in 12m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2a7b623873094b47a7d6368b87a90ad7 : SUCCESS in 12m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"428ed72bc4e4a9bfdefff2922cf52c6637953fc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 09:46:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44012a612c004b5598dc8b9374335028\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/506566295fe94018aa382bc9171bf733 : SUCCESS in 5m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/288e6f1a7cf64ca1b817820d5f9e40bb : FAILURE in 7m 01s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e77b185aabdc4e3ab04f74b0056b394a : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c35534e6ad614c9a9a2fa440253e1b4f : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/23a34e94e7ae4396878029d34f3e0df2 : SUCCESS in 6m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6e84825ed6d0440ab167f2f17d5c0d7e : SUCCESS in 6m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18f0bb6fdd7848b4bc0a81cbb37ed779 : SUCCESS in 5m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92d952a8b8054b54b7874fd88d51f571 : SUCCESS in 5m 29s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a01700daeeb24e629f32a06e758f1d1d : SUCCESS in 4m 53s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/d083505054204fffa4d3384731e597c0 : TIMED_OUT in 3h 03m 38s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/a8efd08200be410a85a571a741909a7c : FAILURE in 1h 36m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1a05bb0f68979c96ba7ad0ab89b6b8cdf68c0bf7","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2023-06-15 18:39:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3aed889b0cec909fd0b0a90b1611e0a115526b0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 18:40:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"56558cd4c7d190ea54d49e57ba127d516982328f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 20:48:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4caaf2cc8ca45498fc278b398135cf1\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/11468088a6834f6bb3f21907013f644a : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10f9b4349812490290af8c8272afc1fb : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5dd91d4980d9445d89fd5bdcc5f8ab40 : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84c5b267f0de4dcfb75af43b5506a1f3 : SUCCESS in 5m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9216b19e5ecd4b068bfcc9840d4a4d51 : SUCCESS in 7m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b25cb3d58a14cb7bd7955a3748701fe : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23bc1027eb774e7381b8cb4a4d67cb81 : SUCCESS in 5m 21s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/02b47b55c5374721bf8763f8d4799981 : SUCCESS in 2h 07m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"486e7bf07045ac742ee02b4962a734cbbdce5166","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 20:48:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3fd287865674cc89b506137222a92168beb2e38a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 20:49:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3a3972b9af64ecbb2884d18867a847a\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/32582b0396c247708ad83d35752e4a7b : SUCCESS in 37s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dcc6ab78fee4bcbba3f8f9e7d8c1014 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff36eac3289344d08b115115b3a7fbc2 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"89514c0d5b3e125496c55d1b361693f7ecaa2972","revisions":{"89514c0d5b3e125496c55d1b361693f7ecaa2972":{"kind":"REWORK","_number":1,"created":"2023-06-14 06:36:11.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/44/886044/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/44/886044/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/44/886044/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/44/886044/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/44/886044/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/44/886044/1"}}},"commit":{"parents":[{"commit":"09fd606163017aaa4e09e577da7080aff5a1d774","subject":"synchronize guest logs to controller node","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/09fd606163017aaa4e09e577da7080aff5a1d774"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-14 06:35:07.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-14 06:35:07.000000000","tz":480},"subject":"Don\u0027t ignore E125 check","message":"Don\u0027t ignore E125 check\n\nE125: https://www.flake8rules.com/rules/E125.html\nChange-Id: I8156bb87ce3e4d1a19d2df9e2c3f67f789bffa46\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/89514c0d5b3e125496c55d1b361693f7ecaa2972"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/89514c0d5b3e125496c55d1b361693f7ecaa2972"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
