)]}'
{"id":"openstack%2Ftrove~886241","triplet_id":"openstack%2Ftrove~master~I73ec6c659d8ad7216460055077737429c878df33","project":"openstack/trove","branch":"master","topic":"fix-postgress-create-database","attention_set":{},"removed_from_attention_set":{"31737":{"account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"last_update":"2023-06-26 04:15:26.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2023-06-26 02:03:08.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"I73ec6c659d8ad7216460055077737429c878df33","subject":"Fix \"create database\" failed for postgress","status":"MERGED","created":"2023-06-16 03:40:14.000000000","updated":"2023-06-26 04:16:41.000000000","submitted":"2023-06-26 04:15:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886241-fix-postgress-create-database","meta_rev_id":"89cc57049e79ec77e5d7dddbc60c3e7eebc8a668","_number":886241,"virtual_id_number":886241,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-26 04:15:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30941,"name":"Hung, Nguyen Trong","display_name":"hungnt1","email":"sudo.nguyenhung@gmail.com","username":"hungnt1"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"recommended":{"_account_id":30941,"name":"Hung, Nguyen Trong","display_name":"hungnt1","email":"sudo.nguyenhung@gmail.com","username":"hungnt1"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-06-25 10:38:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30941,"name":"Hung, Nguyen Trong","display_name":"hungnt1","email":"sudo.nguyenhung@gmail.com","username":"hungnt1"},{"value":2,"date":"2023-06-26 02:03:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30941,"name":"Hung, Nguyen Trong","display_name":"hungnt1","email":"sudo.nguyenhung@gmail.com","username":"hungnt1"},{"value":1,"date":"2023-06-26 02:03:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30941,"name":"Hung, Nguyen Trong","display_name":"hungnt1","email":"sudo.nguyenhung@gmail.com","username":"hungnt1"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":30941,"name":"Hung, Nguyen Trong","display_name":"hungnt1","email":"sudo.nguyenhung@gmail.com","username":"hungnt1"},{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-16 03:46:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-16 05:45:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-16 09:22:31.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"state":"REVIEWER"},{"updated":"2023-06-25 10:38:37.000000000","updated_by":{"_account_id":30941,"name":"Hung, Nguyen Trong","display_name":"hungnt1","email":"sudo.nguyenhung@gmail.com","username":"hungnt1"},"reviewer":{"_account_id":30941,"name":"Hung, Nguyen Trong","display_name":"hungnt1","email":"sudo.nguyenhung@gmail.com","username":"hungnt1"},"state":"REVIEWER"}],"messages":[{"id":"6f444fd39e5312ec676b54d944a96123e6b71fd4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-16 03:40:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d6e6fde68cfdad304966d45ea60d8f2736b5bec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 03:46:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc07b15bbcb040a79574f9d05fc24368\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/db2c02a6fbdf40f5af769c7bb6e69b54 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/38c246546a314af8b6cc8d7b0b8f8f01 : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7804231a2fc1462fa1ca0ebebef1e24d : SUCCESS in 5m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ed194616d132872c2f8df81e641cff1652e51260","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 05:45:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a680b66ffa24020a173d5f75c0d92a3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0162d85a05f347bcb72566151286b5bb : FAILURE in 6m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecdd602ef03e4299b2d17221e7042bcb : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/587549f362e7467ebdedaf369bcee6fd : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f4872df7ca640ffb4d65337a24fa92b : SUCCESS in 6m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf55b7b52fb8447f819f3d40b22266d3 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3611d425b37c4762b152a1f6c1e7f2c8 : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/221d8bebb5c74717936dc576c00007ed : SUCCESS in 5m 52s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/78e0f4b72ddc4239b26cec4a56c2100e : SUCCESS in 5m 41s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/e70e900a73794082a5aca1767e155974 : SUCCESS in 2h 04m 41s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/702ed074eef34f2da35f87ed6974782e : FAILURE in 1h 51m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ed22f4844f879d263479407a6adbebbd2dcb0b87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-16 05:56:56.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"caae04a30e3c4c9484139dda3a9397c84de0958b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 06:11:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc1bfd4a86f1402fa3b16e760e66c1b9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/88f4658e6398499e8435caf61afeaa1f : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/398a036083704b98ad7a1c4ea34bb152 : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6db2caec69c944c5970488dcad3d36ae : SUCCESS in 12m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8b2854b02343100e97be1451e2dcf8ff377f06b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 08:03:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f329bcf2c2c0495184f47c427cbadb48\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ea9586d4bee4eadb46becd73ba299f1 : FAILURE in 5m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a32227e901c4320b3b5a33568e10f44 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19624cb61a1e46f3b1d75e7784e6454e : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d8554d1db9c4f998c625b3fc98de5a6 : SUCCESS in 5m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/64accde6e92a47e2acc76803fd28adaa : SUCCESS in 7m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f0e5cfe012340b4be2ff10a97cfa40e : SUCCESS in 4m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/700ec75a110d4663939d17413f9d847f : SUCCESS in 4m 34s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4626fe4acb8749b7b06ddee1e5df052b : SUCCESS in 4m 33s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/68c4499618ff47818846f7835f6fba8e : SUCCESS in 2h 05m 04s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/ed269421296f4b8aa86cf80ea0bba179 : FAILURE in 1h 22m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ad01051f7ae1abca23ead37f81b904115d04f679","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-16 09:22:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"797660677ea96b637b17cde77812c52aaa4525d0","author":{"_account_id":30941,"name":"Hung, Nguyen Trong","display_name":"hungnt1","email":"sudo.nguyenhung@gmail.com","username":"hungnt1"},"date":"2023-06-25 10:38:37.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"295de38b3f31a14db4004abcc98815dd6208b9ad","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-26 02:03:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a30f7577043ec11fc6ea623d7eaff938f526bcb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 02:06:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9e2fecf6f8859c77795ef4a79997c071932556b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 04:15:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe40639b61d245dd876d04adb9500d62\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7466757f45d4a44be7560c5c8f11f77 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/193f8e979704420194d1c11cb71c628b : SUCCESS in 6m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8782f02e18d6472d9ad9d0ddf1cb1db1 : SUCCESS in 6m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/71ab0a181fc14c6cbcffc3aedf378c33 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43960150cb6140e0a4e7ef43931f83f2 : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c6b6fcfa55b4019bdcecd4816eca463 : SUCCESS in 5m 15s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/963ff067e8e5448a880325667028e3ac : SUCCESS in 2h 07m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b2645dfb1cc84b4e8dccdd924dd48a486ca2fef3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 04:15:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"89cc57049e79ec77e5d7dddbc60c3e7eebc8a668","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 04:16:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b13536fd68449c7aed1bc2108aba0e9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64798522b2414634864f1499f21b9c2c : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d66082161494c7c91ab24b57549c987 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cd11d7677e80826d15d4a79935eb127e5129946c","revisions":{"da0b6a70ccba726544b79eb24ff5509761975edf":{"kind":"REWORK","_number":1,"created":"2023-06-16 03:40:14.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/41/886241/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/41/886241/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/41/886241/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/41/886241/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/41/886241/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/41/886241/1"}}},"commit":{"parents":[{"commit":"09fd606163017aaa4e09e577da7080aff5a1d774","subject":"synchronize guest logs to controller node","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/09fd606163017aaa4e09e577da7080aff5a1d774"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-16 03:35:22.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-16 03:35:22.000000000","tz":480},"subject":"Fix \"create database\" failed for postgress","message":"Fix \"create database\" failed for postgress\n\nThis PR stops using the psycopg2’s connections context manager\nbecause \"CREATE DATABASE\" command must be run outside any\ntransaction[1].\n[1]: https://www.psycopg.org/docs/usage.html#transactions-control\n\nStory: 2010761\nTask: 48059\nChange-Id: I73ec6c659d8ad7216460055077737429c878df33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/da0b6a70ccba726544b79eb24ff5509761975edf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/da0b6a70ccba726544b79eb24ff5509761975edf"}]},"branch":"refs/heads/master"},"cd11d7677e80826d15d4a79935eb127e5129946c":{"kind":"REWORK","_number":2,"created":"2023-06-16 05:56:56.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/41/886241/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/41/886241/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/41/886241/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/41/886241/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/41/886241/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/41/886241/2"}}},"commit":{"parents":[{"commit":"09fd606163017aaa4e09e577da7080aff5a1d774","subject":"synchronize guest logs to controller node","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/09fd606163017aaa4e09e577da7080aff5a1d774"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-16 03:35:22.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-16 05:54:08.000000000","tz":480},"subject":"Fix \"create database\" failed for postgress","message":"Fix \"create database\" failed for postgress\n\nThis PR stops using the psycopg2’s connections context manager\nbecause \"CREATE DATABASE\" command must be run outside any\ntransaction[1].\n[1]: https://www.psycopg.org/docs/usage.html#transactions-control\n\nCo-Authored-By: hungnt1 \u003csudo.nguyenhung@gmail.com\u003e\nCo-Authored-By: Hirotaka Wakabayashi \u003chiwkby@yahoo.com\u003e\n\nStory: 2010761\nTask: 48059\nChange-Id: I73ec6c659d8ad7216460055077737429c878df33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cd11d7677e80826d15d4a79935eb127e5129946c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cd11d7677e80826d15d4a79935eb127e5129946c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
