)]}'
{"id":"openstack%2Ftrove~886326","triplet_id":"openstack%2Ftrove~stable%2Fxena~I217033ab7266e9760ee92eb46401a11b82990030","project":"openstack/trove","branch":"stable/xena","topic":"fix-zuul-errors","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-06-20 11:09:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I217033ab7266e9760ee92eb46401a11b82990030","subject":"[CI] Move queue setting to project level","status":"MERGED","created":"2023-06-20 07:46:33.000000000","updated":"2023-06-20 11:11:09.000000000","submitted":"2023-06-20 11:09:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886326-fix-zuul-errors","cherry_pick_of_change":886325,"cherry_pick_of_patch_set":2,"meta_rev_id":"05aa3452e5ac7395b54850966bad928b8bd9cf5d","_number":886326,"virtual_id_number":886326,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-20 11:09:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-20 09:30:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-06-20 09:30:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-20 07:46:33.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"CC"},{"updated":"2023-06-20 09:23:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-20 09:30:41.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"39db6c639d0647c8b7317ce7897ec7707e08db03","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-20 07:46:33.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"5d606402c1bf41aa2e7fe4b641e93e8880d29061","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-20 07:46:43.000000000","message":"Patch Set 1:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"558670174ef45daef7a8d0a48141936f74552e07","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-20 07:51:40.000000000","message":"Topic set to fix-zuul-errors","accounts_in_message":[],"_revision_number":1},{"id":"0ac857c15d714dbd9a0e4ed89c0314b34817684d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 09:23:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bab10e123ce54c878ac8cb27668799ab\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f5a5c7321bee41ba95c2c5762fb58217 : SUCCESS in 3m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2dec7d72e77c494285feb74fb8d75c72 : FAILURE in 6m 19s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf9ec3d84f0f40338bec6240ad78ee50 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eee13aaa547046a3bd90926daea89ccc : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0518655125df4063a472fcd48ad20bd0 : SUCCESS in 6m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/946653d854e84034a0b7c1ae0ef45048 : SUCCESS in 8m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a30e82e1b126479ba9f6fd8cb579c2cf : SUCCESS in 9m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3911f816442e442594fad32fd5da3bcb : SUCCESS in 6m 59s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c5dc330e45b84877bb412ebad855d0f2 : SUCCESS in 8m 12s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/78a76eb5785845f39488a54c930ff507 : FAILURE in 1h 03m 22s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/ed76edfb877a4a54ad8d391eaab117a7 : FAILURE in 1h 04m 06s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d99df28d9be402da0ed5258a08b1f08 : FAILURE in 1h 33m 56s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/20cfdec4ca194636901839ee345429c9 : FAILURE in 28m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4d637a4672608155314620a8e6e48a15b11b4a13","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-20 09:30:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4b37c18a847bc83ad35da9b787d7c14c58df4c02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 09:31:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"59094001a8bed575a5317acf59edda9b7f5da6b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 11:09:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d376d552e0ec40fbaee90ea7a64343e5\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/4f15ae261b7b49e68c080ea354d337f1 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b2dde41457641f9a280060d92353e70 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a237dc33c354970af55ac6c1e2271ba : SUCCESS in 8m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07dfa62a124f4b07ac4fb52d2c4ab35b : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/001bcdb06ea6448280cff20f4b1ef215 : SUCCESS in 7m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d54dadb20cc8445a8513338837f4ad95 : SUCCESS in 5m 21s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/1feeedaeacf94d2bb9439c749a40fcb4 : FAILURE in 1h 36m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a3d588a7f636cb0490cbabcb51fdaf2757e6b17b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 11:09:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"05aa3452e5ac7395b54850966bad928b8bd9cf5d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 11:11:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/937c77c9de474ac3a6ac1bad81f47f11\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f0af3c846514130849c2cab3ad4e4aa : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b0b92ef4848749b78aa48e3d9f2038ae9999f45c","revisions":{"b0b92ef4848749b78aa48e3d9f2038ae9999f45c":{"kind":"REWORK","_number":1,"created":"2023-06-20 07:46:33.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/26/886326/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/26/886326/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/886326/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/886326/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/886326/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/26/886326/1"}}},"commit":{"parents":[{"commit":"71d8400db5403e37f17340bc05b2a22324d334cd","subject":"Merge \"Fix illegal shell characters\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/71d8400db5403e37f17340bc05b2a22324d334cd"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-09-20 03:28:13.000000000","tz":480},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2023-06-20 07:46:33.000000000","tz":0},"subject":"[CI] Move queue setting to project level","message":"[CI] Move queue setting to project level\n\nPer [1].\n\n[1] http://lists.zuul-ci.org/pipermail/zuul-discuss/2022-May/001801.html\n\nManually fixed cherry-pick to apply to the right file.\n\nChange-Id: I217033ab7266e9760ee92eb46401a11b82990030\n(cherry picked from commit 8ef65e0e49a6eacea37df893826a9e7f2964373a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b0b92ef4848749b78aa48e3d9f2038ae9999f45c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b0b92ef4848749b78aa48e3d9f2038ae9999f45c"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
