)]}'
{"id":"openstack%2Ftrove~886329","triplet_id":"openstack%2Ftrove~stable%2Ftrain~I217033ab7266e9760ee92eb46401a11b82990030","project":"openstack/trove","branch":"stable/train","topic":"fix-zuul-errors","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-06-26 04:47:34.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"I217033ab7266e9760ee92eb46401a11b82990030","subject":"[CI] Move queue setting to project level","status":"MERGED","created":"2023-06-20 07:50:44.000000000","updated":"2023-06-26 04:52:42.000000000","submitted":"2023-06-26 04:47:42.000000000","submitter":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886329-fix-zuul-errors","cherry_pick_of_change":886451,"cherry_pick_of_patch_set":1,"meta_rev_id":"5a93f48f475934425c3116535e017717904a67c1","_number":886329,"virtual_id_number":886329,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-06-26 04:47:34.000000000","_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-06-20 08:46:17.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-06-26 04:47:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-20 09:30:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2023-06-26 04:47:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-06-20 09:30:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-20 07:50:44.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"CC"},{"updated":"2023-06-20 08:46:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-20 09:30:48.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"41daeeb18073884b299aa4500cb32d0df58b15b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-20 07:50:44.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"9f58a4622b6e979456b8c8cf9410475caa9273f6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-20 07:51:40.000000000","message":"Topic changed from fix-zuul-errors-stable/train to fix-zuul-errors","accounts_in_message":[],"_revision_number":1},{"id":"f961cf1102c2a48ca8f7715be3d0d0ad8a2903f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 08:46:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/86bb4d63c11f409081a797b37c92553a\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b92beed8747641648f27068fa6269833 : FAILURE in 3m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73d1107be7734271b29bdee1dbfb6803 : FAILURE in 3m 48s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df22fe3488ac4bbc84cff6fa172f4db5 : FAILURE in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd93ac97f7f9491fa237fe422bcaf1ad : FAILURE in 4m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b1eeb69e641b4fdb99feca9b7e1f409b : FAILURE in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/776a2871ee6245329e927c7bc3699d12 : FAILURE in 3m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4fedb1c772b441cda657f3c9aac3ff60 : FAILURE in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7956e6b54c5c4c0f944dc7f5beeb2813 : FAILURE in 3m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0714ac94f37946eca48d826917c0e655 : SUCCESS in 5m 16s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1807225a6a9d4d17a53f23cfe1bb70fe : FAILURE in 3m 32s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b18ea1c17e2f420e91ef31ccea1e3d71 : FAILURE in 3m 56s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/453ef45f50264bc797110791dba18d49 : FAILURE in 47m 06s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/e2b98bb4e96640ec80500473f0b26833 : FAILURE in 43m 57s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/1ccfcb7c989e48a1adffef10cd95e157 : FAILURE in 43m 58s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/527d7de3860d431bb976830a285d6bdc : FAILURE in 44m 46s\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/00f2a0ee69714226873d8c4d58ac972e : FAILURE in 43m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7fc07b160fc73207bba37d1a350751de42faa4fa","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-20 09:30:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c2e65e28c78c12043003e96234b4e5475d0310b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 10:24:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/51aa70173caa44d2858e9342727b1db8\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/2edb261f40d84b79980feb5244c2c4aa : FAILURE in 3m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/11e3c4a843e6482d8ad09d2d973f66a7 : FAILURE in 3m 48s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e7f11489191e408d91aa78654c1270b7 : FAILURE in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1fde3e7d3ce451f86325ee18ba09bae : FAILURE in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8cb097c0a09647e183d39df92295e414 : FAILURE in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/760c9a161f4148e6a7be1e642c099fa7 : FAILURE in 6m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/01aed4e783b342cfbb5c51c8771fc4c2 : FAILURE in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44f5b67af871419f99e6f60ad6fc6825 : FAILURE in 4m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ef1f2d056d94e1f8071d192183d8bde : SUCCESS in 5m 19s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8010916d4bc541a385195f8b51b660fe : FAILURE in 4m 06s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/720e5cf444884f39bc4cf41b65cb3a0f : FAILURE in 4m 01s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/0c46584e10ce4f41aa7f6434dcfab493 : FAILURE in 47m 14s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/5684556386c04b0893232118e56052d5 : FAILURE in 48m 14s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/453ddf1f62264805a625d9c8713704ae : FAILURE in 30m 54s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/e5cd649c961340d08128aabc1e3c085f : FAILURE in 31m 25s\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/b065e0d1132c4fd4bcab1e8ad2130549 : FAILURE in 33m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6814c2ad6c0305b68392281e66364b44b21b9422","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-26 04:47:34.000000000","message":"Patch Set 1: Verified+2 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9561c9d286083998874d5ceea12a55a19857d8c1","tag":"autogenerated:gerrit:merged","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-26 04:47:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5a93f48f475934425c3116535e017717904a67c1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 04:52:42.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/69bb1b52b6d04271b3882d304e2a04c3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d1f0de96ee44c34ad564a5f56b16b9f : FAILURE in 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0683f1c597ca79949de4160753051238d9c0d1ac","revisions":{"0683f1c597ca79949de4160753051238d9c0d1ac":{"kind":"REWORK","_number":1,"created":"2023-06-20 07:50:44.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/29/886329/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/29/886329/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/29/886329/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/29/886329/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/29/886329/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/29/886329/1"}}},"commit":{"parents":[{"commit":"d8b5525fd2a544c264a86edcd0d3493709f262af","subject":"Fix CI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d8b5525fd2a544c264a86edcd0d3493709f262af"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-09-20 03:28:13.000000000","tz":480},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2023-06-20 07:50:44.000000000","tz":0},"subject":"[CI] Move queue setting to project level","message":"[CI] Move queue setting to project level\n\nPer [1].\n\n[1] http://lists.zuul-ci.org/pipermail/zuul-discuss/2022-May/001801.html\n\nManually fixed cherry-pick to apply to the right file.\n\nChange-Id: I217033ab7266e9760ee92eb46401a11b82990030\n(cherry picked from commit 8ef65e0e49a6eacea37df893826a9e7f2964373a)\n(cherry picked from commit ce9dbe696b4b3b516e6cea041ac7123242956f28)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0683f1c597ca79949de4160753051238d9c0d1ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0683f1c597ca79949de4160753051238d9c0d1ac"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
