)]}'
{"id":"openstack%2Ftrove~886846","triplet_id":"openstack%2Ftrove~stable%2F2023.1~I73ec6c659d8ad7216460055077737429c878df33","project":"openstack/trove","branch":"stable/2023.1","topic":"fix-postgress-create-database-stable/2023.1","hashtags":[],"change_id":"I73ec6c659d8ad7216460055077737429c878df33","subject":"Fix \"create database\" failed for postgress","status":"MERGED","created":"2023-06-26 05:50:14.000000000","updated":"2023-06-28 14:45:18.000000000","submitted":"2023-06-28 14:43:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886846-fix-postgress-create-database-stable/2023.1","cherry_pick_of_change":886241,"cherry_pick_of_patch_set":2,"meta_rev_id":"c948230d625ff7dd5d9095772a0d1ce167201834","_number":886846,"virtual_id_number":886846,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-28 14:43:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-28 12:06:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-06-28 12:06:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-26 05:57:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-26 07:39:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7b35b8f0d30f1d154fe1e7b11d41cabd4409fe8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-26 05:50:14.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"555dfb9168c68f6784de1209eff296120e9ce2d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 05:57:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab8de4df892b4a4482eb6c459c74bbce\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bcb379cb32094ebfbb3b75aa2914029d : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/606db145743e43f5a3acda744898ad4b : SUCCESS in 6m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b3218af464b7fc4bb49b948ed6b1d3ef2e806414","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 07:39:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f700a6e5c4f4421d82e6fef4b60e6a19\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ea393f407144f5e83612145a2463c65 : FAILURE in 5m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c64cc2bcb7254d59aca4075de27e8f51 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d04e82857bf44cc495e87be0469e87a6 : SUCCESS in 5m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7fbc5f2e936a4d84b8fcfc8ce087769f : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b156b81f0ae7475eb5d35c515911fcad : SUCCESS in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/431f23417255402b91a3613606ed8338 : SUCCESS in 5m 08s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bbaf8368fa16433f924476895af8fe01 : SUCCESS in 5m 33s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/7ea8a293490842769116d1b93de11fb8 : FAILURE in 1h 25m 08s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/5a2fceae8aaa49c7a87c87726513761c : FAILURE in 1h 48m 00s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/405719556a554b9fbb98019ec9cb2285 : FAILURE in 1h 25m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"245aa60524c4985868b2234decfa3481680efb85","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-28 12:06:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ace52b785e731bcbc88fb83f2b59147644aa2ca2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 12:07:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d07110d5deac204b313ce402dae0f11afe378519","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 14:43:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75f8e3ce92f24368a6faf0440cfcee8a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29455873d3184a4bbb2d58e5c37f6528 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6168ad529574466d825cff5faed908e0 : SUCCESS in 5m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c002e07be8a14364b765d790a9438c8a : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/695f6f50e163470298cc6aca2644a978 : SUCCESS in 5m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33ad2a43b23841b8b9ae8c39a619933a : SUCCESS in 5m 13s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/d36242d421ec4675a9375cafad98be25 : FAILURE in 1h 19m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7495280997e3436ae8c403d3fea996afef854093","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 14:43:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c948230d625ff7dd5d9095772a0d1ce167201834","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 14:45:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bf376e7f435420e8c670444d9f8e1a5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b2053c4e87c4486af7fab7385ec7ef5 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b4c2158950a437799344ac9862a5cd8 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f62243a941d3f0f8eda35099361b20e706de357c","revisions":{"f62243a941d3f0f8eda35099361b20e706de357c":{"kind":"REWORK","_number":1,"created":"2023-06-26 05:50:14.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/46/886846/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/46/886846/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/46/886846/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/46/886846/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/46/886846/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/46/886846/1"}}},"commit":{"parents":[{"commit":"fee50052d1b9e9800406aa1037aef3cb04c8fbba","subject":"[CI] Don\u0027t run public trove images jobs in periodic pipeline","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fee50052d1b9e9800406aa1037aef3cb04c8fbba"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-16 03:35:22.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-06-26 05:50:14.000000000","tz":0},"subject":"Fix \"create database\" failed for postgress","message":"Fix \"create database\" failed for postgress\n\nThis PR stops using the psycopg2’s connections context manager\nbecause \"CREATE DATABASE\" command must be run outside any\ntransaction[1].\n[1]: https://www.psycopg.org/docs/usage.html#transactions-control\n\nCo-Authored-By: hungnt1 \u003csudo.nguyenhung@gmail.com\u003e\nCo-Authored-By: Hirotaka Wakabayashi \u003chiwkby@yahoo.com\u003e\n\nStory: 2010761\nTask: 48059\nChange-Id: I73ec6c659d8ad7216460055077737429c878df33\n(cherry picked from commit cd11d7677e80826d15d4a79935eb127e5129946c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f62243a941d3f0f8eda35099361b20e706de357c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f62243a941d3f0f8eda35099361b20e706de357c"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
