)]}'
{"id":"openstack%2Ftrove~886847","triplet_id":"openstack%2Ftrove~stable%2Fzed~I73ec6c659d8ad7216460055077737429c878df33","project":"openstack/trove","branch":"stable/zed","topic":"fix-postgress-create-database-stable/zed","hashtags":[],"change_id":"I73ec6c659d8ad7216460055077737429c878df33","subject":"Fix \"create database\" failed for postgress","status":"MERGED","created":"2023-06-26 05:50:42.000000000","updated":"2023-06-28 14:45:15.000000000","submitted":"2023-06-28 14:43:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886847-fix-postgress-create-database-stable/zed","cherry_pick_of_change":886241,"cherry_pick_of_patch_set":2,"meta_rev_id":"bf5b3e813cd4e54f1ce0423f8377b843a19be9a9","_number":886847,"virtual_id_number":886847,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-28 14:43:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-28 12:07:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-06-28 12:07:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-26 07:22:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3f8a7d88d4661cfc9d14fda735315891eaeece13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-26 05:50:42.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9e3164e3d0372a2407624720527b58607092b20c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 07:22:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e6944b3b17a4299bdf655fc4326f86c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b663828b4894e4cb1a93089b6c9776e : FAILURE in 9m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7729e4d3250e4664a5847ff854c7950c : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8644a9d3d1f84534a273e2c1fbb6ce5c : SUCCESS in 6m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e26c14d590db4727922003805fd5fe9a : SUCCESS in 6m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c9d0744e91184aa2bce021e3ab2373d8 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a94ed4f0e8c14cf3a6f904551aa6fcf8 : SUCCESS in 6m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd8e52ebe1314efe8e087d3f3d7702fc : SUCCESS in 5m 40s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/564da80aeb3845498e8f9ad5ecc92c02 : SUCCESS in 5m 47s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/1d9d1693f4204869809167c72db1f257 : FAILURE in 1h 30m 03s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/04dd887de0544d1b963357ab9f318025 : FAILURE in 1h 30m 43s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/844a9eb931694caa94910548b5b705ec : FAILURE in 41m 20s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/3ec57f46d1eb4378b62bb17a02a88a07 : FAILURE in 31m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"680e5461d4eb5b87388b966d0499f31cb9fcad44","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-28 12:07:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ebf42747190f6ca724fd7ec8980fbcdbce963435","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 12:07:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a37e0949405db2c71c863b39bd66029344b34557","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 14:43:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84ac58c92a08495eb706275dab5e762c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da73b579045147c682662030ad170679 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ae1f57897814ae8a97ee4d417fd1c42 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/446a629cd9ba4a26a88ce82a2ed51462 : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86b1f5ce05e1489ab368f8fe346e5a78 : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d43a5ad464a469ab1a843932669f8fd : SUCCESS in 6m 33s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/4c565add34cf43dfabd3348b853ecc49 : FAILURE in 1h 26m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0a394ea3f71bdd890032ed0b510c706825c7712f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 14:43:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bf5b3e813cd4e54f1ce0423f8377b843a19be9a9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 14:45:15.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb34909d97e14758b8d355eadac98dbe\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/570770c07dea4e229aa4e13cab6556c8 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/18509b19e93b449898a7b597564d7912 : FAILURE in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b195a6ab06c15316be4cf4045550aefcbe4016d4","revisions":{"b195a6ab06c15316be4cf4045550aefcbe4016d4":{"kind":"REWORK","_number":1,"created":"2023-06-26 05:50:42.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/47/886847/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/47/886847/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/886847/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/886847/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/886847/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/47/886847/1"}}},"commit":{"parents":[{"commit":"7c8b90531402e8cd6ae879ab636fd636c123ebe5","subject":"Fixes permission problem when restoring backup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7c8b90531402e8cd6ae879ab636fd636c123ebe5"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-16 03:35:22.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-06-26 05:50:42.000000000","tz":0},"subject":"Fix \"create database\" failed for postgress","message":"Fix \"create database\" failed for postgress\n\nThis PR stops using the psycopg2’s connections context manager\nbecause \"CREATE DATABASE\" command must be run outside any\ntransaction[1].\n[1]: https://www.psycopg.org/docs/usage.html#transactions-control\n\nCo-Authored-By: hungnt1 \u003csudo.nguyenhung@gmail.com\u003e\nCo-Authored-By: Hirotaka Wakabayashi \u003chiwkby@yahoo.com\u003e\n\nStory: 2010761\nTask: 48059\nChange-Id: I73ec6c659d8ad7216460055077737429c878df33\n(cherry picked from commit cd11d7677e80826d15d4a79935eb127e5129946c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b195a6ab06c15316be4cf4045550aefcbe4016d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b195a6ab06c15316be4cf4045550aefcbe4016d4"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
