)]}'
{"id":"openstack%2Ftrove~886848","triplet_id":"openstack%2Ftrove~stable%2Fyoga~I73ec6c659d8ad7216460055077737429c878df33","project":"openstack/trove","branch":"stable/yoga","topic":"fix-postgress-create-database-stable/zed-stable/yoga","hashtags":[],"change_id":"I73ec6c659d8ad7216460055077737429c878df33","subject":"Fix \"create database\" failed for postgress","status":"MERGED","created":"2023-06-26 05:51:03.000000000","updated":"2023-06-28 14:44:44.000000000","submitted":"2023-06-28 14:43:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886848-fix-postgress-create-database-stable/zed-stable/yoga","cherry_pick_of_change":886847,"cherry_pick_of_patch_set":1,"meta_rev_id":"14001b28ac627669a99ea0004697095bffaf99b0","_number":886848,"virtual_id_number":886848,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-28 14:43:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-28 12:06:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-06-28 12:06:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-26 08:31:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"45870f5d5a0a38e39115719489313b55a35b70d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-26 05:51:03.000000000","message":"Patch Set 1: Cherry Picked from branch stable/zed.","accounts_in_message":[],"_revision_number":1},{"id":"c6dae813204f8af82c27f64a393cff6e5d0d4ef0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 08:31:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a59ec16d1d8d4d5fb53ca6de28e90d98\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fdc88c204475489abf8f2ad0a7afbbef : FAILURE in 11m 13s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ba8305d21dc4c1bb9e74e805507e72e : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78b80bda6dc34e839dd9f9a0edb130d6 : SUCCESS in 6m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3292f1ef58f4a5fae2932cd956f235b : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59450b9c9b8046479873d66fd718a020 : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ca5d8b7bac54cda9b9f8c37c274189c : SUCCESS in 6m 03s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c26f8d085e554f8dbcd992f098e2eabb : SUCCESS in 6m 03s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/d6c144990e294093841c4819e579a40b : FAILURE in 2h 35m 58s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/c2ec6fcbe95843fe94dbe465c706f10d : FAILURE in 2h 39m 20s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/39dc6a1d7cee4d6fa1dc9f6a4e384220 : FAILURE in 41m 30s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/8f919780381340198b9714d731888c0b : FAILURE in 30m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e3e8bf8bfc8fc26cf4e41bdc0e65fd387231fcd3","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-28 12:06:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c6adb4d74be52ad82f30718dc0a89b6d577cf1cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 12:06:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a280a691c0ac971ab598e9f282c0566ef6f5274b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 14:43:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f920cacad2b84a079d60571f8ec2e3c9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78c5839327114d95834a64bcb83f62c3 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83bd2547eabd4ac88d065e3359edef21 : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e413be84f78a4a9ca139448d0383fbf1 : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d855ba4a12be43ad83e2c38360072b9c : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44f86784c7234c58aae05e8ab9a53353 : SUCCESS in 5m 08s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/19e9e1671e6444f7829be541183de022 : FAILURE in 2h 35m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6acb546030e7d44a2808b7bdc9978f93868512f9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 14:43:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"14001b28ac627669a99ea0004697095bffaf99b0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 14:44:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6df00bf5c1cc4774a1c434a03556a8f0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51249d9cfce5451aa2b3141bad6043cc : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9d0ada69e844d2791d8da163323910f : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f07e85ded52abe31eaad6340d07f0d16c0a40c33","revisions":{"f07e85ded52abe31eaad6340d07f0d16c0a40c33":{"kind":"REWORK","_number":1,"created":"2023-06-26 05:51:03.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/48/886848/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/48/886848/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/886848/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/886848/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/886848/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/48/886848/1"}}},"commit":{"parents":[{"commit":"25e67af50f0dcf716887a64d79807bc2acae55b2","subject":"[CI] Move queue setting to project level","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/25e67af50f0dcf716887a64d79807bc2acae55b2"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-16 03:35:22.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-06-26 05:51:03.000000000","tz":0},"subject":"Fix \"create database\" failed for postgress","message":"Fix \"create database\" failed for postgress\n\nThis PR stops using the psycopg2’s connections context manager\nbecause \"CREATE DATABASE\" command must be run outside any\ntransaction[1].\n[1]: https://www.psycopg.org/docs/usage.html#transactions-control\n\nCo-Authored-By: hungnt1 \u003csudo.nguyenhung@gmail.com\u003e\nCo-Authored-By: Hirotaka Wakabayashi \u003chiwkby@yahoo.com\u003e\n\nStory: 2010761\nTask: 48059\nChange-Id: I73ec6c659d8ad7216460055077737429c878df33\n(cherry picked from commit cd11d7677e80826d15d4a79935eb127e5129946c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f07e85ded52abe31eaad6340d07f0d16c0a40c33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f07e85ded52abe31eaad6340d07f0d16c0a40c33"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
