)]}'
{"id":"openstack%2Ftrove~886950","triplet_id":"openstack%2Ftrove~stable%2Fwallaby~I73ec6c659d8ad7216460055077737429c878df33","project":"openstack/trove","branch":"stable/wallaby","topic":"fix-postgress-create-database-stable/zed-stable/yoga-stable/xena-stable/wallaby","attention_set":{},"removed_from_attention_set":{"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2023-07-04 01:50:00.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"I73ec6c659d8ad7216460055077737429c878df33","subject":"Fix \"create database\" failed for postgress","status":"MERGED","created":"2023-06-26 05:52:53.000000000","updated":"2023-07-04 02:21:49.000000000","submitted":"2023-07-04 02:20:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886950-fix-postgress-create-database-stable/zed-stable/yoga-stable/xena-stable/wallaby","cherry_pick_of_change":886849,"cherry_pick_of_patch_set":1,"meta_rev_id":"7ab66bf13ab8aa08041e119a4b6c81c987183abd","_number":886950,"virtual_id_number":886950,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-04 02:20:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-07-04 01:50:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-07-04 01:50:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-26 07:42:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e48e6bac37a38a5690ca1f9d6164012f97d163ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-26 05:52:53.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"6787dccc8f0854109af0b4a04131dedf77f8300c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 07:42:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/41a04db58ac6447e955b54579c26166a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8402a60882c7468a92ebce40b04b3d95 : FAILURE in 5m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9244c2aa139e49218fd047b790c52279 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07901fc35bbc403cbad636a4db31a830 : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3e18853a58e4f1db16a1a5b0daa695a : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4833789c8a743099abf2e93bbbe8f2a : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1aa1036433e2456da76d91c762bdb12b : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/748106dc52e1444aa7a37cdd22e5b507 : SUCCESS in 5m 29s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/eb9bb64892734b4fbed00f98e9a8cf79 : SUCCESS in 5m 38s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/9fe7378fe71a4603999a9a5e42779a8e : FAILURE in 34m 24s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4bbbc596a8084a67b1737c64bb1e114b : FAILURE in 1h 46m 40s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/abe650bb8be643c69974de28a58f34ea : FAILURE in 34m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7f0d5cc26db99ddbf60e32eae6f27efeaff4e755","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-30 01:35:55.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e503ba827a47a3ae83151a34dc0f5e189b95220f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 03:19:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acc8398a90974652a85f9f5aa84f36de\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49173c7e11a94c0794064658d708d1fc : FAILURE in 6m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/440d8e538b9840faabe59575c2bb1b27 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8551b59600f44276abfb44c4fcd1d4d2 : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b71667395b194dd99505fd629faa256d : SUCCESS in 6m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0434e862a5e04a9aa206a87de0a3b5bf : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd96bee1445042da88bcd9a3a0f94ea7 : SUCCESS in 4m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2fa9b901d41e4876b19910ddf8ed2134 : SUCCESS in 5m 51s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3870b0b787e54cfab3fd9e94d3593e86 : SUCCESS in 4m 32s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/9091f44394d14f9c960ebd216eeec590 : FAILURE in 34m 30s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b16af84aee024d83a77e7d0af1148552 : FAILURE in 1h 38m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"aadbea9d5ab37043129639c4b02d818fc5cd17d2","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-07-04 01:50:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7efeb9720dc14243c283e3d3bc5b9205def06b6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-04 01:50:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"76d6fdd2ac8eea965fff59cb5c3b646a7cd3bf0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-04 02:20:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d45b6d27f8f84af9a5ec942da0e3b33f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ab8c97ca0894f8ea7f325566e65ad5b : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e48833fb9ca4743a1f1f8966d9728af : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b746c58978dd407a97f76509a5b00e2d : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da401de86cb44eb0af216e886399b184 : SUCCESS in 5m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c4faa6bef1f487db7bb3bfa5f281f4d : SUCCESS in 5m 07s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/bb5a4d368d5347759b262aa35099759d : FAILURE in 29m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"30a51c7722db85efae4f09f38f9ec6d19c4b604a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-04 02:20:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7ab66bf13ab8aa08041e119a4b6c81c987183abd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-04 02:21:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70fad9446352440aa9eaca93209aa997\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4720508ce2174e92902d1c36faf01dfc : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8d53bd4ca4a4b6d831614e6648c693e : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bd4e97b4d71d18e76e25ad58b31219ebf1236e50","revisions":{"a8a9342c7b0f6a08722c879c3487614fd2e1612d":{"kind":"REWORK","_number":1,"created":"2023-06-26 05:52:53.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/50/886950/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/50/886950/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/50/886950/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/50/886950/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/50/886950/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/50/886950/1"}}},"commit":{"parents":[{"commit":"d249e5e35d0d2850d273643a193de8220d74737f","subject":"[CI] Move queue setting to project level","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d249e5e35d0d2850d273643a193de8220d74737f"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-16 03:35:22.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-06-26 05:52:53.000000000","tz":0},"subject":"Fix \"create database\" failed for postgress","message":"Fix \"create database\" failed for postgress\n\nThis PR stops using the psycopg2’s connections context manager\nbecause \"CREATE DATABASE\" command must be run outside any\ntransaction[1].\n[1]: https://www.psycopg.org/docs/usage.html#transactions-control\n\nCo-Authored-By: hungnt1 \u003csudo.nguyenhung@gmail.com\u003e\nCo-Authored-By: Hirotaka Wakabayashi \u003chiwkby@yahoo.com\u003e\n\nStory: 2010761\nTask: 48059\nChange-Id: I73ec6c659d8ad7216460055077737429c878df33\n(cherry picked from commit cd11d7677e80826d15d4a79935eb127e5129946c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a8a9342c7b0f6a08722c879c3487614fd2e1612d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a8a9342c7b0f6a08722c879c3487614fd2e1612d"}]},"branch":"refs/heads/stable/wallaby"},"bd4e97b4d71d18e76e25ad58b31219ebf1236e50":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-06-30 01:35:55.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/50/886950/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/50/886950/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/50/886950/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/50/886950/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/50/886950/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/50/886950/2"}}},"commit":{"parents":[{"commit":"40d7df7d62446b5984c9cf0654e500276fa98c7a","subject":"[CI]: set voting to false for trove-tempest jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/40d7df7d62446b5984c9cf0654e500276fa98c7a"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-16 03:35:22.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-06-30 01:35:55.000000000","tz":0},"subject":"Fix \"create database\" failed for postgress","message":"Fix \"create database\" failed for postgress\n\nThis PR stops using the psycopg2’s connections context manager\nbecause \"CREATE DATABASE\" command must be run outside any\ntransaction[1].\n[1]: https://www.psycopg.org/docs/usage.html#transactions-control\n\nCo-Authored-By: hungnt1 \u003csudo.nguyenhung@gmail.com\u003e\nCo-Authored-By: Hirotaka Wakabayashi \u003chiwkby@yahoo.com\u003e\n\nStory: 2010761\nTask: 48059\nChange-Id: I73ec6c659d8ad7216460055077737429c878df33\n(cherry picked from commit cd11d7677e80826d15d4a79935eb127e5129946c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bd4e97b4d71d18e76e25ad58b31219ebf1236e50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bd4e97b4d71d18e76e25ad58b31219ebf1236e50"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
