)]}'
{"id":"openstack%2Ftrove~887175","triplet_id":"openstack%2Ftrove~stable%2F2023.1~I72e68d9e7df6a82b4635abac995299be770546ce","project":"openstack/trove","branch":"stable/2023.1","hashtags":[],"change_id":"I72e68d9e7df6a82b4635abac995299be770546ce","subject":"[CI]: Don\u0027t run periodic jobs on stable branchs","status":"MERGED","created":"2023-06-29 07:46:48.000000000","updated":"2023-07-02 13:50:40.000000000","submitted":"2023-07-02 13:48:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887175","cherry_pick_of_change":887243,"cherry_pick_of_patch_set":2,"meta_rev_id":"6e470fd49402fd667e2a7575a2515d5744010087","_number":887175,"virtual_id_number":887175,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-02 13:48:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-07-02 10:58:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-07-02 10:58:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-29 07:47:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f0755344b0e2734aa9276a59a7f2994ec6042d51","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-29 07:46:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* zuul.d/projects.yaml","accounts_in_message":[],"_revision_number":1},{"id":"1cb92af80c01140efa9fa75295caee937ecbfa73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 07:47:11.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/trove on branch stable/2023.1.  The error was:\n\n  \n  Zuul encountered a syntax error while parsing its configuration in the\n  repo openstack/trove on branch stable/2023.1.  The error was:\n  \n    while scanning a simple key\n    in \"openstack/trove/zuul.d/projects.yaml@stable/2023.1\", line 41, column 1\n  could not find expected \u0027:\u0027\n    in \"openstack/trove/zuul.d/projects.yaml@stable/2023.1\", line 44, column 9","accounts_in_message":[],"_revision_number":1},{"id":"a1596ff5f0d9eb2cec2599327c40946025c4d139","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 07:47:17.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/trove on branch stable/2023.1.  The error was:\n\n  \n  Zuul encountered a syntax error while parsing its configuration in the\n  repo openstack/trove on branch stable/2023.1.  The error was:\n  \n    while scanning a simple key\n    in \"openstack/trove/zuul.d/projects.yaml@stable/2023.1\", line 41, column 1\n  could not find expected \u0027:\u0027\n    in \"openstack/trove/zuul.d/projects.yaml@stable/2023.1\", line 44, column 9","accounts_in_message":[],"_revision_number":1},{"id":"fda379d65c3115520ffb83299ec299548fd1e449","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-29 07:47:29.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"051512e842f6a9faf7a6f0f8bd035a21b9fe6053","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-29 07:48:07.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"32c1effc8bd1b55f4cf8c65c6e566e3a9ad685dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 07:59:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3a94813b1274db68ba9537a22d8c311\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1a78f5a3464d43a4b01a3e30cbc0ec27 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3a8077983739477387da57fe5c09b8a1 : SUCCESS in 6m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ffbf0ae7127e7a6f45f87e6d6420330c7396bd88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 09:14:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83a30d3ba74a4bd4a602ac4f07b588fd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/914a17fba28e4ecb8260c17ed50b5390 : FAILURE in 4m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0ea8aa51cbe44cc97ca4fd5a6193c38 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8bf43bd175d74f9288f5a699caf56e37 : SUCCESS in 5m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ef0ad89cb0524dbbaf523bc6ed3bb403 : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bad45b2c6149413a8d0fa888166da5f5 : SUCCESS in 5m 38s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c53a0da7e3ca4aa2b2d2c7e1dfa01c51 : SUCCESS in 4m 03s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/455807d6ca004dd0914bc5fb63206177 : FAILURE in 1h 01m 09s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/8c7a06d110ea400a963ed3702ae5ca6d : FAILURE in 1h 25m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"db0a0b2a2f232007ba6f81763279637f9eea86e3","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-07-02 10:58:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cf31be459f46f0ff41781f27257eeea904ba2a83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-02 10:59:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ba6ef7fa34674c0fcabb6b97810f646167402567","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-02 13:48:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63210364d152472aa7b0cd969d05bcdd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a2661793368498a9bbe184505f24889 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2850c4076ca4ce7a160b953ff028bb9 : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/216c23ec50fe435fb0e7674bb6aadf36 : SUCCESS in 8m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72dfb2c348e242139d1976230de98901 : SUCCESS in 5m 23s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/5135229c0f51494bb492f565fa213ffd : SUCCESS in 2h 47m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bbc1cde5ce83f9b8527680e9ac0e62c9250e177f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-02 13:48:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"6e470fd49402fd667e2a7575a2515d5744010087","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-02 13:50:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f89e8a2229e649caa596805d2f561667\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d929016814e4aa1bc1e9da60a08406c : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"959e9cd517e0806e5a7ec2c7d3b65c78eeab1d42","revisions":{"080ccf9d5904def8de73499b9b5d3f3a0b948907":{"kind":"REWORK","_number":1,"created":"2023-06-29 07:46:48.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/75/887175/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/887175/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/887175/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/887175/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/887175/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/887175/1"}}},"commit":{"parents":[{"commit":"f62243a941d3f0f8eda35099361b20e706de357c","subject":"Fix \"create database\" failed for postgress","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f62243a941d3f0f8eda35099361b20e706de357c"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-29 03:20:34.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-06-29 07:46:48.000000000","tz":0},"subject":"[CI]: Don\u0027t run periodic jobs on stable branchs","message":"[CI]: Don\u0027t run periodic jobs on stable branchs\n\nwe don\u0027t need to pulish trove image with stable branches everyday.\n\nChange-Id: I72e68d9e7df6a82b4635abac995299be770546ce\n(cherry picked from commit 7a6666f4f301cd4a9a4d2b7380546893f0c1c6f8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/080ccf9d5904def8de73499b9b5d3f3a0b948907"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/080ccf9d5904def8de73499b9b5d3f3a0b948907"}]},"branch":"refs/heads/stable/2023.1"},"959e9cd517e0806e5a7ec2c7d3b65c78eeab1d42":{"kind":"REWORK","_number":2,"created":"2023-06-29 07:47:29.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/75/887175/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/887175/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/887175/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/887175/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/887175/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/887175/2"}}},"commit":{"parents":[{"commit":"f62243a941d3f0f8eda35099361b20e706de357c","subject":"Fix \"create database\" failed for postgress","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f62243a941d3f0f8eda35099361b20e706de357c"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-29 03:20:34.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-06-29 07:47:28.000000000","tz":0},"subject":"[CI]: Don\u0027t run periodic jobs on stable branchs","message":"[CI]: Don\u0027t run periodic jobs on stable branchs\n\nwe don\u0027t need to pulish trove image with stable branches everyday.\n\nChange-Id: I72e68d9e7df6a82b4635abac995299be770546ce\n(cherry picked from commit 7a6666f4f301cd4a9a4d2b7380546893f0c1c6f8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/959e9cd517e0806e5a7ec2c7d3b65c78eeab1d42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/959e9cd517e0806e5a7ec2c7d3b65c78eeab1d42"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
