)]}'
{"id":"openstack%2Ftrove~887176","triplet_id":"openstack%2Ftrove~stable%2Fzed~I72e68d9e7df6a82b4635abac995299be770546ce","project":"openstack/trove","branch":"stable/zed","hashtags":[],"change_id":"I72e68d9e7df6a82b4635abac995299be770546ce","subject":"[CI]: Don\u0027t run periodic jobs on stable branchs","status":"MERGED","created":"2023-06-29 07:48:52.000000000","updated":"2023-06-30 03:43:27.000000000","submitted":"2023-06-30 03:41:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887176","cherry_pick_of_change":887175,"cherry_pick_of_patch_set":2,"meta_rev_id":"56a96f4142c6ce3b9277223d662bdd20b3d5dd06","_number":887176,"virtual_id_number":887176,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-30 03:41:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-30 02:17:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-06-30 02:17:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-29 07:49:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5431b1c14f3e7cf14f82fc9ddfca069ea70fe926","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-29 07:48:52.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.1.\n\nThe following files contain Git conflicts:\n* zuul.d/projects.yaml","accounts_in_message":[],"_revision_number":1},{"id":"c7f33cba3a0e5a6885277b57c157e2fd646b7ffd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 07:49:21.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/trove on branch stable/zed.  The error was:\n\n  \n  Zuul encountered a syntax error while parsing its configuration in the\n  repo openstack/trove on branch stable/zed.  The error was:\n  \n    while scanning a simple key\n    in \"openstack/trove/zuul.d/projects.yaml@stable/zed\", line 43, column 1\n  could not find expected \u0027:\u0027\n    in \"openstack/trove/zuul.d/projects.yaml@stable/zed\", line 46, column 1","accounts_in_message":[],"_revision_number":1},{"id":"ae8a558919672e018e0de3a73fa5e4338f6e4d13","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-29 07:49:23.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2b470a8fc9820a94e6598f5e432a67af1a56a0d3","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-29 08:03:27.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"1981d929a905bc31cc6d98ecdc6b9413333c04af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 09:29:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20f0d7ed8f324edda61cfd3504abd0df\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8102c6b9d60d4e9ba7badc0d73aa8fa8 : FAILURE in 7m 19s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88044febedb44aa49e8632a9bb269fa5 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/541791957f3e4dab8e6f1b409b1c2157 : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3adc4c04e2c64c55877fa01410749f38 : SUCCESS in 6m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1a0f49e2570149f7bfc2b3e1d19febe6 : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae6af2c78bac46eca9c416f5756c1aad : SUCCESS in 5m 36s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3a621c25390240c38025e6bcfa54b022 : SUCCESS in 5m 50s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/87b9d2397b3e410989f24baf217672f0 : FAILURE in 1h 39m 02s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/dca2d03074d5462a83f05bb9273ad1ec : FAILURE in 1h 21m 47s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/47aa3417c5b149f9803998a3adaec55e : FAILURE in 30m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f91936c54f429c23f849c6459ed9a611ae4d5527","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-30 02:17:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5db208419176a97319709cd36ed02b83ac6f4e7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 02:18:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"acafc1b13c938ca68e12de29a6e44a29f51c05be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 03:41:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/291814a0c2384777863311107fa76a4c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abdfb5ad9a484c8e90f0ffe90e96eaeb : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b173450dc9e424b9951d705f1f95fb6 : SUCCESS in 6m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1945320798f54868b3846ff84fe9771b : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c501e5952c514800b2572015154bd59c : SUCCESS in 5m 40s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/4123e035d28849b7aa8d43ca0aacedfe : FAILURE in 1h 21m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e7bf0309500215183c81a8242b6074aa202bc706","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 03:41:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"56a96f4142c6ce3b9277223d662bdd20b3d5dd06","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 03:43:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85077e9b49d542cf84693bcb251ca608\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9faeff87ecb44209d1b897c9e42163a : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9f8ed20e812942337936410d13fb0ef1fd90970f","revisions":{"f5ed5154db6ffb3c7a20152d8b21c17699c94ceb":{"kind":"REWORK","_number":1,"created":"2023-06-29 07:48:52.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/76/887176/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/76/887176/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/76/887176/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/76/887176/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/76/887176/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/76/887176/1"}}},"commit":{"parents":[{"commit":"b195a6ab06c15316be4cf4045550aefcbe4016d4","subject":"Fix \"create database\" failed for postgress","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b195a6ab06c15316be4cf4045550aefcbe4016d4"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-29 03:20:34.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-06-29 07:48:52.000000000","tz":0},"subject":"[CI]: Don\u0027t run periodic jobs on stable branchs","message":"[CI]: Don\u0027t run periodic jobs on stable branchs\n\nwe don\u0027t need to pulish trove image with stable branches everyday.\n\nChange-Id: I72e68d9e7df6a82b4635abac995299be770546ce\n(cherry picked from commit 7a6666f4f301cd4a9a4d2b7380546893f0c1c6f8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f5ed5154db6ffb3c7a20152d8b21c17699c94ceb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f5ed5154db6ffb3c7a20152d8b21c17699c94ceb"}]},"branch":"refs/heads/stable/zed"},"9f8ed20e812942337936410d13fb0ef1fd90970f":{"kind":"REWORK","_number":2,"created":"2023-06-29 07:49:23.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/76/887176/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/76/887176/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/76/887176/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/76/887176/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/76/887176/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/76/887176/2"}}},"commit":{"parents":[{"commit":"b195a6ab06c15316be4cf4045550aefcbe4016d4","subject":"Fix \"create database\" failed for postgress","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b195a6ab06c15316be4cf4045550aefcbe4016d4"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-29 03:20:34.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-06-29 07:49:23.000000000","tz":0},"subject":"[CI]: Don\u0027t run periodic jobs on stable branchs","message":"[CI]: Don\u0027t run periodic jobs on stable branchs\n\nwe don\u0027t need to pulish trove image with stable branches everyday.\n\nChange-Id: I72e68d9e7df6a82b4635abac995299be770546ce\n(cherry picked from commit 7a6666f4f301cd4a9a4d2b7380546893f0c1c6f8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9f8ed20e812942337936410d13fb0ef1fd90970f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9f8ed20e812942337936410d13fb0ef1fd90970f"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
