)]}'
{"id":"openstack%2Ftrove~887244","triplet_id":"openstack%2Ftrove~stable%2Fwallaby~I48f3ddcdc523f8e5c5d65a89b9b1327fab7c8ef3","project":"openstack/trove","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2023-06-29 12:08:15.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"I48f3ddcdc523f8e5c5d65a89b9b1327fab7c8ef3","subject":"[CI]: set voting to false for trove-tempest jobs","status":"MERGED","created":"2023-06-29 03:47:52.000000000","updated":"2023-06-29 13:04:07.000000000","submitted":"2023-06-29 13:02:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887244","meta_rev_id":"0f52ead03ecb752f7304cb5b3369a46a6eca3fa4","_number":887244,"virtual_id_number":887244,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-29 13:02:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-29 12:08:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-06-29 12:08:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-29 04:30:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2cef7d3b83e185242ceca54b2eabebe0d4f74a76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-29 03:47:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e6edc1757e2597c8108c32823a68427f39f5821","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 04:30:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6af31610ef0b486fbfa0fadba021d1ce\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4e9f391f1ee4a458bb7b5a769bad651 : FAILURE in 6m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe168f028da446779324e1fac78ad853 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/020067da822f4cd3a93a231d40d48a18 : SUCCESS in 11m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84cd1118a15e4598833db4a64c031a0e : SUCCESS in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab845a1a52e440ae83c10564ab42eb17 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/023d2908264d48b6922df1db8a1019e2 : SUCCESS in 12m 04s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d16677f6471d4f79b5673c36ee5eb15d : SUCCESS in 11m 26s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/f4051cbb239d425c87552e8a1ef0bc83 : FAILURE in 40m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"86219ad76ec335f64caf5985fd66a6b7c9b7ef26","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-29 06:47:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c12c3571700ac00e4f6a1955502809edc12c5b4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 06:48:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ef83a7859262ea1e927db9c0fb54af47f1533267","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 07:32:21.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3232d74555f14718873cd0bfc3c80566\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0ac1a0cfc2b44b0a9e257f4c5088f73 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6feeb8d792364d388f4d6693044a7ed9 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90a597879db34ee0960590f7243d0b1c : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbda313ede954660952a17475efc0eea : SUCCESS in 5m 46s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/8287662133a24292ac3b58d74fb71a30 : FAILURE in 41m 44s","accounts_in_message":[],"_revision_number":1},{"id":"b7123ea343770e1f7c5d9492730914fad001efa7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-29 07:58:17.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ddcad4a07e9693447ec6d0c72684bc657115e55f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 08:28:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/452c7d6d947643ffbdae2357b593dfb5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c969b9789f6488b87f25855f944daa8 : FAILURE in 6m 15s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6316b5165b746c989f5728739e2183e : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2fad746300034d7099245cccad7ce017 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02706fe9469b460ab93037625b9fdf78 : SUCCESS in 6m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03d9e5a00e834d79b2326932d9ca3a7f : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcfc9a0b12b141c0b430fbf461d59270 : SUCCESS in 6m 38s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0681fbbf8f104882b9c47888467e1811 : SUCCESS in 6m 52s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/498ee9d0b02a4899b4e069ac9dad0963 : FAILURE in 28m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9a0088f28656f1c0c9c3d4fc5af7966bcfebb07c","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-29 12:08:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"dcf66276e4e5225471d4c0baf53e96f4b91f1648","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 12:08:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e4c728f42f65ebb3131df89ee3d70f2936e4b96f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 13:02:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc913b6583bf432f9833da3a613f4137\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c7946e69835456da8d5d6093e446264 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a916f506f8d0428191b84a005f5cffba : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb9d35c7eb884a9fba0bb4044b4919aa : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdcee7dd0eb64ec9bbc658fbce37ef24 : SUCCESS in 5m 34s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/55d62d67c4584b4eb4596f9d88349816 : FAILURE in 34m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b7931cbf3f7bdc0f470af84d48f4bec7bb6d1245","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 13:02:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0f52ead03ecb752f7304cb5b3369a46a6eca3fa4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 13:04:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd19f523f1c54375890f43163e39d0b7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa2fb85df1104a1ba353479add423470 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"40d7df7d62446b5984c9cf0654e500276fa98c7a","revisions":{"83ed66e225b3e1bc8dff286f615484837d44f24f":{"kind":"REWORK","_number":1,"created":"2023-06-29 03:47:52.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/44/887244/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/44/887244/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/44/887244/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/44/887244/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/44/887244/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/44/887244/1"}}},"commit":{"parents":[{"commit":"d249e5e35d0d2850d273643a193de8220d74737f","subject":"[CI] Move queue setting to project level","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d249e5e35d0d2850d273643a193de8220d74737f"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-29 03:45:19.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-29 03:45:19.000000000","tz":480},"subject":"[CI]: set voting to false for trove-tempest jobs","message":"[CI]: set voting to false for trove-tempest jobs\n\nAs nobody maintains this jobs, we need to set voting to false, so\nthat we can pass the zuul checks.\n\nChange-Id: I48f3ddcdc523f8e5c5d65a89b9b1327fab7c8ef3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/83ed66e225b3e1bc8dff286f615484837d44f24f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/83ed66e225b3e1bc8dff286f615484837d44f24f"}]},"branch":"refs/heads/stable/wallaby"},"40d7df7d62446b5984c9cf0654e500276fa98c7a":{"kind":"REWORK","_number":2,"created":"2023-06-29 07:58:17.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/44/887244/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/44/887244/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/44/887244/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/44/887244/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/44/887244/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/44/887244/2"}}},"commit":{"parents":[{"commit":"d249e5e35d0d2850d273643a193de8220d74737f","subject":"[CI] Move queue setting to project level","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d249e5e35d0d2850d273643a193de8220d74737f"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-29 03:45:19.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-06-29 07:58:16.000000000","tz":0},"subject":"[CI]: set voting to false for trove-tempest jobs","message":"[CI]: set voting to false for trove-tempest jobs\n\nAs nobody maintains this jobs, we need to set voting to false, so\nthat we can pass the zuul checks.\n\nChange-Id: I48f3ddcdc523f8e5c5d65a89b9b1327fab7c8ef3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/40d7df7d62446b5984c9cf0654e500276fa98c7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/40d7df7d62446b5984c9cf0654e500276fa98c7a"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
