)]}'
{"id":"openstack%2Ftrove~887348","triplet_id":"openstack%2Ftrove~stable%2F2023.1~I90c4c3694c0010fa57fff1e994ede2d3bf9ba17f","project":"openstack/trove","branch":"stable/2023.1","hashtags":[],"change_id":"I90c4c3694c0010fa57fff1e994ede2d3bf9ba17f","subject":"Change settings for stable/2023.1","status":"MERGED","created":"2023-06-30 02:16:15.000000000","updated":"2023-07-02 12:54:55.000000000","submitted":"2023-07-02 12:53:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887348","meta_rev_id":"e23f5f1ec9508cbbed9def5019d81bd4f433cf9d","_number":887348,"virtual_id_number":887348,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-02 12:53:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-07-02 10:57:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-07-02 10:57:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-30 02:28:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-30 05:14:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"92be6d0e5597ae40c6db5d86340efd55fc6559ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-30 02:16:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c6a187e2e206bf3aa0c9d6fd74c4ad6118a59441","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 02:28:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9e9e90c410b4e01b7c69e72076f0fd2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c7fecfcb229e4cac86e74eb14d40aa55 : SUCCESS in 10m 33s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cc9fe4da338f42c19aeeed9b3c4f1b40 : SUCCESS in 9m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eb09294d20fe901252951ef0982be871d56e99cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 05:14:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ec56693a1f24ba98fe2a8f1315bb5e1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0767b72c8db4e958874e784361b4ced : FAILURE in 7m 14s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90860124c2da45bab025b5fa764e03a7 : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/853fcabf4dce462c9fe633ede4157892 : SUCCESS in 5m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9f9d52de0cca46049f2012bab30f1372 : SUCCESS in 20m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7a6a86b44a44ea9a6b48abca85a37d5 : SUCCESS in 6m 02s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b173fca7ec3e49fe9bd10af6bc5ac439 : SUCCESS in 4m 48s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/ee8402601fb34b7db50a061bbe75f7bb : SUCCESS in 1h 52m 37s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/e4b1049334b4410e87060259b30a15f8 : SUCCESS in 8m 52s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/7b90f9f892f34cd9a461c926fdf709b3 : SUCCESS in 14m 49s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/5717b0b662bd4b8bb332101f09f78c63 : SUCCESS in 1h 51m 09s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2fd06935119d4979878b59c5932dd90c : TIMED_OUT in 2h 02m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e04650af03c9bcc5fa7fcab15edcd9fc8175f6bd","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-07-02 10:57:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cfd1fdcf96a881a321b45cc28621961cc5bf0245","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-02 10:58:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8dfbb80a77e5b51da3c9240f0d510f68f17d9cc8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-02 12:53:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/523323b1d3ec4fcb828c9970b63c45e7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42223cb253bc4a9cb7a6e7ec2e562764 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c38cc66cfb334092a549f2f2596052b1 : SUCCESS in 9m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c23649bd9a3042ca85551799521312fa : SUCCESS in 8m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c4bb45b60294f34a132e563cd7b6c3d : SUCCESS in 5m 28s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/c6a6467c52674e82afa44c7e794e1309 : SUCCESS in 1h 53m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d40b2335fa5382d782cff3e77648385191dc8511","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-02 12:53:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e23f5f1ec9508cbbed9def5019d81bd4f433cf9d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-02 12:54:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33bdaf1437f94988847b9107fdc26212\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff9b715e7302494aae1192e27cf472da : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"53eef3bc7b60a2ed40ee49ff758bee811c1af40e","revisions":{"53eef3bc7b60a2ed40ee49ff758bee811c1af40e":{"kind":"REWORK","_number":1,"created":"2023-06-30 02:16:15.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/48/887348/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/48/887348/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/887348/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/887348/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/887348/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/48/887348/1"}}},"commit":{"parents":[{"commit":"1e7bfafdc551e28f11d1096942ad67ac3a44ade0","subject":"[CI]: fix trove-tempest jobs on stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1e7bfafdc551e28f11d1096942ad67ac3a44ade0"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-30 02:11:57.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-06-30 02:11:57.000000000","tz":480},"subject":"Change settings for stable/2023.1","message":"Change settings for stable/2023.1\n\nChange-Id: I90c4c3694c0010fa57fff1e994ede2d3bf9ba17f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/53eef3bc7b60a2ed40ee49ff758bee811c1af40e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/53eef3bc7b60a2ed40ee49ff758bee811c1af40e"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
