)]}'
{"id":"openstack%2Ftrove~887653","triplet_id":"openstack%2Ftrove~master~Ide59528d9526fdf1c741f0749ca2263721374c65","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2023-07-06 01:18:52.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ide59528d9526fdf1c741f0749ca2263721374c65","subject":"update requirements","status":"ABANDONED","created":"2023-07-05 07:07:56.000000000","updated":"2023-07-06 01:18:52.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"dbd37d04b4bfdcb1fbef4555d160576863283179","_number":887653,"virtual_id_number":887653,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-07-05 09:45:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-05 07:16:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-05 09:45:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"79c1bdb37cb6f3495b8a3b217c718ada1711d8ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-07-05 07:07:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f2b73516bef9338fd3d4fece2ee0cb39001e385","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 07:16:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82263315e37b4db99d4066d27f58ddc2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b10aafd054744c2d8630b67905f24984 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4578b781bb7b49d6a3799417284c43ba : POST_FAILURE in 6m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e9dc565b17784d9bb62485627260875e : SUCCESS in 6m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3e657bcc8e8a8896c25db96b80c1d1655ce3c7ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 09:45:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f991a81219047909e7f8cd781147ff6\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ecd8986addcb4d73beb47d3850f5bf89 : POST_FAILURE in 4m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/674cba5bd9ab49ccb7443ea00a42adc2 : FAILURE in 6m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ed8ae4f1629493e963b30ffd0475ad1 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57641239693348c29be2387fd0ebda33 : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/689af9af859f406faad358fb55316fc3 : SUCCESS in 5m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2409b48284ea4165a721bf5068cda37b : POST_FAILURE in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/478c316557124353b0c088851d42d6c0 : SUCCESS in 5m 25s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/4283f5b742ab4dafadf9ad65c74b0798 : FAILURE in 2h 36m 07s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/0c2edbea64c347708ea9ec820fd6e013 : FAILURE in 2h 36m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dbd37d04b4bfdcb1fbef4555d160576863283179","tag":"autogenerated:gerrit:abandon","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-07-06 01:18:52.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"44de13cc61eab48e474a336e7862c3e975f2d2f9","revisions":{"44de13cc61eab48e474a336e7862c3e975f2d2f9":{"kind":"REWORK","_number":1,"created":"2023-07-05 07:07:56.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/53/887653/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/53/887653/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/53/887653/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/53/887653/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/53/887653/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/53/887653/1"}}},"commit":{"parents":[{"commit":"17af8eab8c9a3673011a0e06690618c5589631ed","subject":"Merge \"mysql: explicitly use utf8mb3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/17af8eab8c9a3673011a0e06690618c5589631ed"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-07-05 07:07:20.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-07-05 07:07:20.000000000","tz":480},"subject":"update requirements","message":"update requirements\n\nAdd flask and pyroute2 module in requirements\n\nChange-Id: Ide59528d9526fdf1c741f0749ca2263721374c65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/44de13cc61eab48e474a336e7862c3e975f2d2f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/44de13cc61eab48e474a336e7862c3e975f2d2f9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
