)]}'
{"id":"openstack%2Ftrove~889375","triplet_id":"openstack%2Ftrove~master~Ib12b92ed4ddd9133ad7a3d079534d02f403dbdc0","project":"openstack/trove","branch":"master","topic":"move_to_ubuntu_jammy","attention_set":{},"removed_from_attention_set":{"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2023-09-06 05:57:37.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"Ib12b92ed4ddd9133ad7a3d079534d02f403dbdc0","subject":"[CI]: Add support for Ubuntu Jammy","status":"MERGED","created":"2023-07-25 07:00:32.000000000","updated":"2023-09-06 08:13:10.000000000","submitted":"2023-09-06 08:11:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889375-move_to_ubuntu_jammy","meta_rev_id":"1219eb5a9c5efb8b64fdb91795e184049f5ef4cf","_number":889375,"virtual_id_number":889375,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-06 08:11:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-06 05:57:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-06 05:57:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-25 07:00:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"140318bf592d7b6ae71d3714bfb3f22785579462","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-07-25 07:00:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6cac1b5540e0056490f3d6d2936df900ee0dc6ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 07:00:52.000000000","message":"Patch Set 1: Verified-1\n\n(5 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/trove on branch stable/2023.1.  The error was:\n\n  Nodeset trove-ubuntu-focal-single does not match existing definition in branch master\n\nThe error appears in the following nodeset stanza:\n\n  nodeset:\n      name: trove-ubuntu-focal-single\n      nodes:\n        - name: primary\n          label: ubuntu-focal\n\n  in \"openstack/trove/zuul.d/nodesets.yaml@stable/2023.1\", line 11, column 3","accounts_in_message":[],"_revision_number":1},{"id":"ce84f2b427f18d2720366ffd48947e16516e1a37","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 07:01:00.000000000","message":"Patch Set 1:\n\n(5 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/trove on branch stable/2023.1.  The error was:\n\n  Nodeset trove-ubuntu-focal-single does not match existing definition in branch master\n\nThe error appears in the following nodeset stanza:\n\n  nodeset:\n      name: trove-ubuntu-focal-single\n      nodes:\n        - name: primary\n          label: ubuntu-focal\n\n  in \"openstack/trove/zuul.d/nodesets.yaml@stable/2023.1\", line 11, column 3","accounts_in_message":[],"_revision_number":1},{"id":"b21bd8b783c9dee18ec296a4e953ac402b15c216","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-07-25 07:05:16.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"39f75b087d63786f10003bfbb0f129e25a2abe13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 07:05:37.000000000","message":"Patch Set 2: Verified-1\n\n(3 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/trove on branch master.  The error was:\n\n  The nodeset \"trove-ubuntu-jammy\" was not found.\n\nThe error appears in the following job stanza:\n\n  job:\n      name: trove-functional-mysql\n      parent: trove-devstack-base\n      nodeset: trove-ubuntu-jammy\n      timeout: 10800\n      vars:\n        devstack_localrc:\n          TROVE_RESIZE_TIME_OUT: 1800\n        trove_resize_time_out: 1800\n        trove_test_datastore: \u0027mysql\u0027\n        trove_test_group: \u0027mysql\u0027\n        trove_test_datastore_version: \u00275.7.29\u0027\n\n  in \"openstack/trove/zuul.d/jobs.yaml@master\", line 79, column 3","accounts_in_message":[],"_revision_number":2},{"id":"dcbf0f67cfffd11e81be37e1d2b481a59e584920","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 07:05:43.000000000","message":"Patch Set 2:\n\n(3 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/trove on branch master.  The error was:\n\n  The nodeset \"trove-ubuntu-jammy\" was not found.\n\nThe error appears in the following job stanza:\n\n  job:\n      name: trove-functional-mysql\n      parent: trove-devstack-base\n      nodeset: trove-ubuntu-jammy\n      timeout: 10800\n      vars:\n        devstack_localrc:\n          TROVE_RESIZE_TIME_OUT: 1800\n        trove_resize_time_out: 1800\n        trove_test_datastore: \u0027mysql\u0027\n        trove_test_group: \u0027mysql\u0027\n        trove_test_datastore_version: \u00275.7.29\u0027\n\n  in \"openstack/trove/zuul.d/jobs.yaml@master\", line 79, column 3","accounts_in_message":[],"_revision_number":2},{"id":"c05470881a8d6aedf257b944f7c11055eb25c255","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-07-25 07:07:42.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"47d1a286e45ce74493b7c2d4cdf0ce018ec41764","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 07:16:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4755fe3701340f38056fa256036411b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/888c669647454608a01d59a2d8a24da6 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c2a2e6251f8f4de0b5f7ebc091dbb562 : SUCCESS in 8m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dad00563664440279ca275b620b321a4 : SUCCESS in 7m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fcd82d48d39eb182abff43876e9c27b52aef3e14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 09:50:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3884bc183d848d68a04e79e0f42d684\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a3711605006496d82ec21452269bedc : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5489cda9ae2249158f024197d578c028 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56dceb21111544708d5a05c72ddd435c : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9cbf69be695540a0b6b96add40071373 : SUCCESS in 6m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/24c8be6ac5444461a8104ef7cdd9b97b : SUCCESS in 8m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7f80be11b234c47b6f73587073f8ff3 : SUCCESS in 5m 10s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b26238da9d7040c6bfe2dd4a596e000f : SUCCESS in 5m 11s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/c1846eabebd64f16bbb81bde22e5fdcf : SUCCESS in 2h 15m 12s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/d5cf8d7de26c402eb041f8698643d07d : FAILURE in 3m 17s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/cbf077e25cb548b68d3d29fd43d5d970 : FAILURE in 3m 46s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/27efe228243e4f788c36aad284ce54e5 : FAILURE in 2h 36m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d718237f657ec27fe994515712d08570a832dc3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-08-07 07:04:04.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"4cbc7336ffdc3d15772f3b9a7c8f9a99608cdf6d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 07:35:06.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a367c6958422402184f53d39ec2159c7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2a4459b30f2a472c85940bde046e0b73 : SUCCESS in 8m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/db5b1999af3045de84a2558890925041 : SUCCESS in 14m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/14234f8f551043e98f9ea3d9a92d5ed0 : SUCCESS in 29m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2f5a434667f608a70e8bd4683992a2476af90f0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 09:20:42.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d9e5f4ac585c41549f6cc841be1552b2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33fea05cc94d4881bed831a81acf77a4 : SUCCESS in 7m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07f4c8701ad2433989a951fce6d577dc : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/569a91a5504544c49e08bedcbb289ba3 : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c478581b78bb44b4856edd8b9d8498d9 : SUCCESS in 5m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/623ae0880ce146d6985169fbc813f539 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc003968b6a14f54bc9934471ec57d96 : SUCCESS in 5m 34s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/42a326f3ef0347899da4be124444d7f3 : SUCCESS in 4m 20s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/cbd427170c4c489a9f0696b0a13730de : SUCCESS in 2h 14m 59s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/80bcd0c9b90848a69a9dc3cb25eb656b : SUCCESS in 2h 05m 56s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/bb9a99b13c9441d5bbf2202cb2e31717 : SUCCESS in 7m 35s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/0d750063c6e04903a900a452faa01fd4 : FAILURE in 4m 03s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/c9e67306cca34e2981479c3cd273e078 : SUCCESS in 2h 11m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a8c2f3d9fa4be7e811d5debee72b0728b562ad86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-09-01 02:14:20.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"79d79d8dd2653bce8c3697f84653873d17dddbcb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 02:29:53.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bb7810a7a10476a949bf3960c2f2ffa\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a67dc9a7922a4f6ea1095987cc5ea3f9 : SUCCESS in 9m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6bee172798004878a9a8d42040026736 : SUCCESS in 12m 22s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/baa9cbd971d2400ca793d0f31ad61fa3 : SUCCESS in 10m 27s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/26e6027c4c824f099896ec5a9fdf9fb9 : RETRY_LIMIT in 3m 33s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6a037c517f92315ac824b5a39f76f5d1efdc1c80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 04:54:50.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f782be3984cf4b5b9ab3a7dd45678c04\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f8e71f574fa4aeeaaeaa5998971fcc2 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f92abd05b93b42208388141ec7b28379 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2b27564fbeb4c72af3705386d3cacbe : SUCCESS in 8m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a0f3e32fa6441ffad68e7e3dda5ddec : SUCCESS in 6m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3da99499acbb477ca64e936ed0e0f826 : SUCCESS in 6m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8ba0385d2a2f481ba68b4bf9f90f32eb : RETRY_LIMIT in 1m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/736ff5fdeb924504a9dde0e0290e2c54 : SUCCESS in 5m 09s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4eaa69ac42ce4146a0b234147895d480 : SUCCESS in 5m 26s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/75141493d99b4c5192d409648ed6cbcd : SUCCESS in 2h 19m 20s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/1db6193f30814fa681e4e4248a5a043a : SUCCESS in 1h 56m 29s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/9c8158c10002421a9d73b00490346c60 : SUCCESS in 9m 43s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/ecfbcd27fbe04540923ac74408bc357d : FAILURE in 4m 12s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/d773f5adfcdc4cb392c18325f66dc280 : SUCCESS in 2h 36m 46s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1ac3a958fe989b554f85b1407cbc24e9a8185913","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-09-01 07:29:22.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"06418b811f36e4ca481e6e06ee8801382b1d588c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 07:46:36.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90926d7c24a74fe381106cbd42eff9d3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fcb0416cdf684b5e92004de25478db7c : SUCCESS in 15m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/adf0783ef3284ab7bb086b18f9059a6c : SUCCESS in 13m 33s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e74e3022de9042cb991404b676c53780 : SUCCESS in 10m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/067995a531524c47bc680e78361e4931 : RETRY_LIMIT in 1m 49s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2705859e590c1c7c6b620517f46d37a8491adfd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 10:35:37.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07c8b9b94bb4447b9d0b531bc01c15fa\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a70580da97a4bac86411907e077cfcb : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f77014e589ed4ca9bb148988ac546dec : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7893295dcf7d4c4394b0b26855c40a60 : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4eee15df50e44adda3eefe36d37ea537 : SUCCESS in 6m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/84eceb1367a14e6c975eb862915a9bdd : SUCCESS in 5m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bf9d06950ad843fe88aa83197ffc61a0 : RETRY_LIMIT in 1m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f800d53adbb4cbcb2e84d5d6466d162 : SUCCESS in 5m 57s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0c9d8044eaf2487aac066920d43b7e4c : SUCCESS in 5m 42s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/1fc1fdfd534d4a2884a06edb71c9610b : TIMED_OUT in 3h 03m 54s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/4c05a5b490ba4d7bb0ea0bbe24f2ab1f : SUCCESS in 1h 32m 21s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/f19e215898f145169c7516e311b852a3 : SUCCESS in 9m 34s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/760e84c2d815477994ecea2275b90022 : SUCCESS in 11m 57s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/6e132272ef904303957d656cd31e85a1 : SUCCESS in 2h 09m 30s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ef59fc1154b0b12dedfe4af5ff8722f347853bef","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-09-06 05:57:37.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"087b1bb089c4df1c7d4239cc0298db988b8b6754","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 05:58:11.000000000","message":"Patch Set 6: -Verified\n\n(3 comments)\n\nStarting gate jobs.\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/trove on branch master.  The problem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following project stanza:\n\n  project:\n      queue: trove\n      templates:\n        - check-requirements\n        - openstack-cover-jobs\n        - openstack-python3-jobs\n        - openstack-python3-jobs-arm64\n        - periodic-stable-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - openstack-tox-cover:\n              voting: true\n          - trove-tox-bandit-baseline:\n              voting: true\n          - trove-tempest:\n              voting: false\n          - trove-tempest-mysql-network-isolation:\n              voting: true\n          - trove-ubuntu-guest-image-build:\n              voting: true\n          - trove-centos8s-guest-image-build:\n              voting: true\n          - trove-tempest-postgres:\n              voting: false\n      gate:\n        jobs:\n          - trove-tempest:\n              voting: false\n      experimental:\n        jobs:\n          - trove-tempest-ipv6-only\n          - trove-functional-mysql\n          - trove-scenario-mysql-single\n          - trove-scenario-mysql-multi\n          - trove-scenario-mariadb-single\n          - trove-scenario-mariadb-multi\n      periodic:\n        jobs:\n          - publish-trove-guest-image-ubuntu-jammy:\n              branches: ^(?!stable/.*).*$\n          - publish-trove-guest-image-ubuntu-jammy-dev:\n              branches: ^(?!stable/.*).*$\n          - publish-trove-datastore-images:\n              branches: ^(?!stable/.*).*$\n\n  in \"openstack/trove/zuul.d/projects.yaml@master\", line 2, column 3","accounts_in_message":[],"_revision_number":6},{"id":"e7012e48fe9c43476842e5e43a15c009ce55fedc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 08:11:02.000000000","message":"Patch Set 6: Verified+2\n\n(3 comments)\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ec73ebd37514d909a33098f893c14bf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3f2d6465e03463a8298e3a6b7448880 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b63ef633cdb04cc8aa25294ad54dc962 : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ab6f7e1d6f14baf9b50026b4b6b324d : SUCCESS in 6m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3b5bc19353554f4cb6a927197bae7294 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dd9d47c8f6c4c269fe586ecbf599763 : SUCCESS in 5m 56s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/3508d316f54441768f1b58c21273309e : SUCCESS in 2h 11m 21s (non-voting)\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/trove on branch master.  The problem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following project stanza:\n\n  project:\n      queue: trove\n      templates:\n        - check-requirements\n        - openstack-cover-jobs\n        - openstack-python3-jobs\n        - openstack-python3-jobs-arm64\n        - periodic-stable-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - openstack-tox-cover:\n              voting: true\n          - trove-tox-bandit-baseline:\n              voting: true\n          - trove-tempest:\n              voting: false\n          - trove-tempest-mysql-network-isolation:\n              voting: true\n          - trove-ubuntu-guest-image-build:\n              voting: true\n          - trove-centos8s-guest-image-build:\n              voting: true\n          - trove-tempest-postgres:\n              voting: false\n      gate:\n        jobs:\n          - trove-tempest:\n              voting: false\n      experimental:\n        jobs:\n          - trove-tempest-ipv6-only\n          - trove-functional-mysql\n          - trove-scenario-mysql-single\n          - trove-scenario-mysql-multi\n          - trove-scenario-mariadb-single\n          - trove-scenario-mariadb-multi\n      periodic:\n        jobs:\n          - publish-trove-guest-image-ubuntu-jammy:\n              branches: ^(?!stable/.*).*$\n          - publish-trove-guest-image-ubuntu-jammy-dev:\n              branches: ^(?!stable/.*).*$\n          - publish-trove-datastore-images:\n              branches: ^(?!stable/.*).*$\n\n  in \"openstack/trove/zuul.d/projects.yaml@master\", line 2, column 3","accounts_in_message":[],"_revision_number":6},{"id":"f79926061f27c545297b3968f060aedc30e3a67b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 08:11:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"1219eb5a9c5efb8b64fdb91795e184049f5ef4cf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 08:13:10.000000000","message":"Patch Set 6:\n\n(3 comments)\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca4901546c37435a9dd9401f090a4f0e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26bb9dee5b2e42f082f9f8cf16ad5c4a : SUCCESS in 49s\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/trove on branch master.  The problem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following project stanza:\n\n  project:\n      queue: trove\n      templates:\n        - check-requirements\n        - openstack-cover-jobs\n        - openstack-python3-jobs\n        - openstack-python3-jobs-arm64\n        - periodic-stable-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - openstack-tox-cover:\n              voting: true\n          - trove-tox-bandit-baseline:\n              voting: true\n          - trove-tempest:\n              voting: false\n          - trove-tempest-mysql-network-isolation:\n              voting: true\n          - trove-ubuntu-guest-image-build:\n              voting: true\n          - trove-centos8s-guest-image-build:\n              voting: true\n          - trove-tempest-postgres:\n              voting: false\n      gate:\n        jobs:\n          - trove-tempest:\n              voting: false\n      experimental:\n        jobs:\n          - trove-tempest-ipv6-only\n          - trove-functional-mysql\n          - trove-scenario-mysql-single\n          - trove-scenario-mysql-multi\n          - trove-scenario-mariadb-single\n          - trove-scenario-mariadb-multi\n      periodic:\n        jobs:\n          - publish-trove-guest-image-ubuntu-jammy:\n              branches: ^(?!stable/.*).*$\n          - publish-trove-guest-image-ubuntu-jammy-dev:\n              branches: ^(?!stable/.*).*$\n          - publish-trove-datastore-images:\n              branches: ^(?!stable/.*).*$\n\n  in \"openstack/trove/zuul.d/projects.yaml@master\", line 2, column 3","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"e1b72b5115d2a47e763210d0b02f217aa4e62956","revisions":{"29fd198bb954b9aaa2a2f2643d42792c29ccc00a":{"kind":"REWORK","_number":1,"created":"2023-07-25 07:00:32.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/75/889375/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/889375/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/889375/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/889375/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/889375/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/889375/1"}}},"commit":{"parents":[{"commit":"c4efd57e3659a2ac075106835ffd807b1f77df35","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c4efd57e3659a2ac075106835ffd807b1f77df35"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-07-25 06:57:34.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-07-25 06:57:34.000000000","tz":480},"subject":"[CI]: Add support for Ubuntu Jammy","message":"[CI]: Add support for Ubuntu Jammy\n\nRun tests on Ubuntu Jammy in zuul.\n\nChange-Id: Ib12b92ed4ddd9133ad7a3d079534d02f403dbdc0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/29fd198bb954b9aaa2a2f2643d42792c29ccc00a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/29fd198bb954b9aaa2a2f2643d42792c29ccc00a"}]},"branch":"refs/heads/master"},"7d3f95eceac3a4fc47cabd8c0a77e636d5e19a3d":{"kind":"REWORK","_number":2,"created":"2023-07-25 07:05:16.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/75/889375/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/889375/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/889375/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/889375/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/889375/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/889375/2"}}},"commit":{"parents":[{"commit":"c4efd57e3659a2ac075106835ffd807b1f77df35","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c4efd57e3659a2ac075106835ffd807b1f77df35"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-07-25 06:57:34.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-07-25 07:04:54.000000000","tz":480},"subject":"[CI]: Add support for Ubuntu Jammy","message":"[CI]: Add support for Ubuntu Jammy\n\nRun tests on Ubuntu Jammy in zuul.\nrefer to:https://governance.openstack.org/tc/reference/runtimes/2023.2.html\n\nChange-Id: Ib12b92ed4ddd9133ad7a3d079534d02f403dbdc0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7d3f95eceac3a4fc47cabd8c0a77e636d5e19a3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7d3f95eceac3a4fc47cabd8c0a77e636d5e19a3d"}]},"branch":"refs/heads/master"},"cf8606c30df80cf76f38258cf2c3b2d15032cc36":{"kind":"REWORK","_number":3,"created":"2023-07-25 07:07:42.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/75/889375/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/889375/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/889375/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/889375/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/889375/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/889375/3"}}},"commit":{"parents":[{"commit":"c4efd57e3659a2ac075106835ffd807b1f77df35","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c4efd57e3659a2ac075106835ffd807b1f77df35"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-07-25 06:57:34.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-07-25 07:07:16.000000000","tz":480},"subject":"[CI]: Add support for Ubuntu Jammy","message":"[CI]: Add support for Ubuntu Jammy\n\nRun tests on Ubuntu Jammy in zuul.\nrefer to:https://governance.openstack.org/tc/reference/runtimes/2023.2.html\n\nChange-Id: Ib12b92ed4ddd9133ad7a3d079534d02f403dbdc0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cf8606c30df80cf76f38258cf2c3b2d15032cc36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cf8606c30df80cf76f38258cf2c3b2d15032cc36"}]},"branch":"refs/heads/master"},"c005bfdc7a70b320fbd65c27d890cac33fc2f2c8":{"kind":"REWORK","_number":4,"created":"2023-08-07 07:04:04.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/75/889375/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/889375/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/889375/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/889375/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/889375/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/889375/4"}}},"commit":{"parents":[{"commit":"b841781fd5a813249959cd0e86a8b170a8792fd7","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b841781fd5a813249959cd0e86a8b170a8792fd7"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-07-25 06:57:34.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-08-07 07:03:42.000000000","tz":480},"subject":"[CI]: Add support for Ubuntu Jammy","message":"[CI]: Add support for Ubuntu Jammy\n\nRun tests on Ubuntu Jammy in zuul.\nrefer to:https://governance.openstack.org/tc/reference/runtimes/2023.2.html\n\nChange-Id: Ib12b92ed4ddd9133ad7a3d079534d02f403dbdc0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c005bfdc7a70b320fbd65c27d890cac33fc2f2c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c005bfdc7a70b320fbd65c27d890cac33fc2f2c8"}]},"branch":"refs/heads/master"},"4c3b35fc6fe4b0a65d569bec53e7ef48d03590cb":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-09-01 02:14:20.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/75/889375/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/889375/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/889375/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/889375/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/889375/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/889375/5"}}},"commit":{"parents":[{"commit":"05c30cbcf6436713dce71e16894a2eb53eac5f12","subject":"Merge \"Fix: failed to create the replicas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/05c30cbcf6436713dce71e16894a2eb53eac5f12"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-07-25 06:57:34.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-09-01 02:14:20.000000000","tz":0},"subject":"[CI]: Add support for Ubuntu Jammy","message":"[CI]: Add support for Ubuntu Jammy\n\nRun tests on Ubuntu Jammy in zuul.\nrefer to:https://governance.openstack.org/tc/reference/runtimes/2023.2.html\n\nChange-Id: Ib12b92ed4ddd9133ad7a3d079534d02f403dbdc0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4c3b35fc6fe4b0a65d569bec53e7ef48d03590cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4c3b35fc6fe4b0a65d569bec53e7ef48d03590cb"}]},"branch":"refs/heads/master","description":"Rebase"},"e1b72b5115d2a47e763210d0b02f217aa4e62956":{"kind":"REWORK","_number":6,"created":"2023-09-01 07:29:22.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/75/889375/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/889375/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/889375/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/889375/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/889375/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/889375/6"}}},"commit":{"parents":[{"commit":"05c30cbcf6436713dce71e16894a2eb53eac5f12","subject":"Merge \"Fix: failed to create the replicas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/05c30cbcf6436713dce71e16894a2eb53eac5f12"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-07-25 06:57:34.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-09-01 07:29:04.000000000","tz":480},"subject":"[CI]: Add support for Ubuntu Jammy","message":"[CI]: Add support for Ubuntu Jammy\n\nRun tests on Ubuntu Jammy in zuul.\nrefer to:https://governance.openstack.org/tc/reference/runtimes/2023.2.html\n\nChange-Id: Ib12b92ed4ddd9133ad7a3d079534d02f403dbdc0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e1b72b5115d2a47e763210d0b02f217aa4e62956"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e1b72b5115d2a47e763210d0b02f217aa4e62956"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
