)]}'
{"id":"openstack%2Ftrove~889789","triplet_id":"openstack%2Ftrove~master~I1998f37b6d81fd451c31311072aaf539434dd6dd","project":"openstack/trove","branch":"master","topic":"fix_err_create_backup_psql","attention_set":{},"removed_from_attention_set":{"28691":{"account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"last_update":"2023-09-06 03:09:47.000000000","reason":"Change was submitted"},"32029":{"account":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"last_update":"2023-09-06 03:09:47.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2023-08-30 09:44:00.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"I1998f37b6d81fd451c31311072aaf539434dd6dd","subject":"Fix error create backup in PostgreSQL","status":"MERGED","created":"2023-07-27 04:06:11.000000000","updated":"2023-09-06 03:10:52.000000000","submitted":"2023-09-06 03:09:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":43,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"889789-fix_err_create_backup_psql","meta_rev_id":"2a5074489c782bc3efd7a469b0bc87605310a3d9","_number":889789,"virtual_id_number":889789,"owner":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-06 03:09:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-06 02:59:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-06 02:59:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-27 04:20:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-27 06:47:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-27 10:20:45.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2023-08-07 02:42:54.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"5aa025955662e9438c9a0c7c63404fae4fc84d95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-07-27 04:06:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f92955a1e1f02b3ada2c874f254feeeb2d0fc6c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 04:20:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7777683ca3e4ba6b2c483c1f1014915\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c13f3215c96947969f469322c0b76dce : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5a01ed2eac4a4bef87fa70d2ab099b61 : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9bd33fdf4070475aa0882b4b05ea6f11 : SUCCESS in 12m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f2df890614446d701e883e5e46514708f95c0c7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 06:47:53.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c1e7b8c3e9f45bcab28f90bbc15e554\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/598fde1eee8a4823bf588f612145e5d6 : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87a0161f2f1c44e28111bddc444abb5f : FAILURE in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b21d2ce3e9394175ae0b289a5e8737df : SUCCESS in 6m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/574de1a986b04b94960557b092042070 : SUCCESS in 6m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a28697018f0545ed873edcce550131af : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23fd8f341a9a465aa424c808f8c228a4 : SUCCESS in 5m 59s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/127099a198684acda0fe7ce74bbccec3 : SUCCESS in 5m 17s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/9389cdae6cc64767935a308f3e4f1e51 : SUCCESS in 2h 40m 19s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/d9c93761ac9b4bfe8e2f434e17b3e132 : FAILURE in 1h 50m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"06dbd3a149f5631ef1db9a4def27559b5c5fe9ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-07-27 07:23:03.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"55dfe1c09806725f95669dfed0fb638e4a42db91","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 07:54:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab5b1931b48042b5aa61beaed2577dab\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b8a6e27797c64d57b1d8df555a023bc5 : SUCCESS in 12m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b52d0ef37cf340f7a74f80580c4f2a9f : SUCCESS in 16m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/30faffd088d947bfa799e246fbd0bc55 : SUCCESS in 15m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"556412213b1631aa2e083a928c7cb984c365abb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 09:47:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b34f6af624444bda8142ffae11dfeca\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eaf03db8c29b47eb864991dfbe197f3c : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cde6fe4f9b2e4da6ab682fc05c63780c : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11d6389c54494034924e897fa6032690 : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5301be40338341c7a0b87466a49b95ca : SUCCESS in 6m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94b9d17e887c43da9c65ddeb173f2fa5 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e31f9043c67d4e4b9c40834def6b7ffd : SUCCESS in 5m 23s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/51705c40673a4f66a4c0fe05da918902 : SUCCESS in 5m 57s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/9b6d742042df4490b49376b7a4412bdc : SUCCESS in 2h 06m 19s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/af66208e95e54c14a0387ff451918022 : FAILURE in 1h 37m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2cb4b1ad33523d6828fedf932e1dab40e16d1b1c","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-07-27 10:20:45.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0e165cd7f087d40995c4ac3587d18e8e65ebf4cf","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-07-28 03:00:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"09f72b6ad113fdba7e056127f3f4735bcc0b92cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-28 03:17:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24585daa4d144a7ca165b3ca0cc2e1d3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cbab6b890d464a82b963e38a4a7356ec : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2b092ddb665f475ba07e3dd6b4d7bc76 : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f3596600166e4777a7500be1efada537 : SUCCESS in 14m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"85b7a51c54bc45bc472f957dee5b4da9dfb06ac1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-28 05:17:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b65fb22044b44b7c8621020a8507e72f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12540ea6ac5c4aed8ddf7af97575fbc3 : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aea17d96053e4055a2027160df3d4aff : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3296b05166af49ac915d76c76c387a12 : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e83658df26c04cb0b7d5bd1c44ba7eb8 : SUCCESS in 6m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cc41a09ad7af4b0483401ae0a2127cfa : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be2658ce0d3749349bfa195743ba7acb : SUCCESS in 5m 29s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f704a21fdac9464098367804877d7c69 : SUCCESS in 6m 27s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/359e4744d41e4fdc95170cb45c9bfe10 : SUCCESS in 2h 14m 53s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/fb772fb74a9048f7882c8b4b7a95403d : FAILURE in 1h 38m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bc0d81dc2563249ad1ac8070aa5b13a6a08f943e","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-07-28 05:49:13.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"30a35ebc4c005eee7b73f137f8ea0aebd4adc6be","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-07-28 07:19:53.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"294fe68d300f4cb3061cfe989e52b0646ee5ee5c","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-07-28 07:20:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2e9f81ab0bdc4b765aa22db1a4e86871cd85e3e0","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-07-31 03:42:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f7a871d2d06e9edf0958cab6afc3dc555a297048","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-07-31 03:47:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"892c3bfc2baa35b0622df4772ef718e3505cf013","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 04:05:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05145cd5f47f4252ae08efe99ed58635\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/baa880e026564c0fb6d6e43997250e49 : SUCCESS in 7m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bfa9bb1a538541dfa85aec6a46d66431 : SUCCESS in 16m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/18286f7ee4444442a250a3e73bc8910e : SUCCESS in 7m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0547ff01db3344f57d9d9a25d379fed9faebd784","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 05:56:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/465113c3471444c28da3d9f2c352d549\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/016a0a3f1b534a8799aeb25437c83f3f : SUCCESS in 8m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/624b48c6998a4e21982460cc84da454e : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c4062eb3c624f328074bd72889091c2 : SUCCESS in 6m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85fbaff051f84ed7b7df8becfaa94f24 : SUCCESS in 6m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e54683fd659e4f199485055b880177ab : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0fb2f74d7b14580a2e029564bd6d9cd : SUCCESS in 5m 31s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b809dedaf9724627a24ccbc96b9e8719 : SUCCESS in 5m 19s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/40f96d7dd8674a2090530c0304e92721 : SUCCESS in 2h 08m 22s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/0e6eb7a5b5f4429b82135b33010897ed : FAILURE in 1h 37m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9c19b1320cfe175d1cb3bbb23edc67faa1cad699","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-07-31 06:11:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ba76c10168dbc705189dad2bede02e40e2266c5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-07-31 06:55:26.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"a90ff47e193ff12f4f998eea202fc97b0caa2174","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 07:02:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efc94e01a4834ea5a1489c816a67cca4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9bae7f7f4c784eec80f21ad1900c534e : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5f2dc777ba4044f1ae03fc6e28e98c1c : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4b6e883bc4104e9eb893a49bb8d40324 : SUCCESS in 5m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"26d685e569a8a35c11b4eb96d66ebf0953b4b489","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 09:59:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67e3b45c791448f0b5e13aefd0a75dfe\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9566fcc74f54edb986f6e5edb65b348 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90493466ace7408b969bc03327831c0d : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c6025392d7c400eba0482f3514e864d : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/841884decc874320970f1bc82f33891b : SUCCESS in 5m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a625340882c48b6b78a36b1333bf201 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b08dfd5eeab4220be65effb561dd054 : SUCCESS in 6m 05s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5fa3eb4ec6844343b30d38984ee151e1 : SUCCESS in 6m 33s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/c1c28c9ced8c43f0a6a1582a88d2aa12 : TIMED_OUT in 3h 02m 59s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/bb27e622286546e88ed1d7a9d654ab19 : FAILURE in 1h 55m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"58b525ab543dfb3b54ec9698270825749c0e95bf","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-08-01 01:39:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"84068e5fdecb2441877b86921b0febda1a87f783","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-08-02 06:07:24.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"38d5f13e6e20452ba46d307da26dc1c7a1b75503","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-08-02 06:09:29.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0182e312f9ce03c1af1dcb704c06a515ac26314c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-08-04 06:19:10.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"fe96d5153b566087132d9add26da10efb28f3bdf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 06:25:40.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01847ef9b51e4ba081e485b7b16019d1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5c12874a5a8b45d59206192846b0d22e : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2de9ae490d754e37bb8779b01d089ddc : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ebce83385a504831815eec96465fbedc : SUCCESS in 5m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"828cfea8057b7c4a70b5ae55f6a71d7e0f5fd2b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 08:47:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d193e5bcb90c4125b9e526dfafa55680\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00f3b555ef2948bab3770120a4675a2b : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb63d879a75c4c6da9be79a332e5bac4 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa5fe0c0cf2041a8afd46c2e2830233f : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/80aed4d5d62d44fa8f0a38c2d3cb802d : SUCCESS in 6m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/218b54c31e6847efa004483d0c211230 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6cae40d1e3c48979e25788530de61f1 : SUCCESS in 4m 30s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/693d677861114c4f90d875640dd301d0 : SUCCESS in 5m 28s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/f47a7611114749c1b01f344b7f44f274 : SUCCESS in 2h 27m 22s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/fdae37eedce74443b1471cd102608672 : SUCCESS in 1h 25m 39s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/0fe211daa5324f90865aebe55e8a45da : FAILURE in 1h 40m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8c0d8aa8b71464f76f578960b47f76be7ee5a44c","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-08-04 09:06:53.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"278063df37ecb763b681b20c733734276a2a78d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-08-04 09:44:01.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"3f5215eb6e41e97f4e90c63d288e844d244e2833","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 09:51:41.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a16081ae62b460ab82516bda75003c1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4530ac2b630f4dd787b90ddfa64e2fbf : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/951e25504bd642e18aae6f58114ba5a3 : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f9e0cef4801e47e0b26089586590aadb : SUCCESS in 4m 52s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9290b28b037c2facccfbcb13fb7fbac7c32e877f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 12:17:06.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/474de6c2bd8b4de8b6df6c1e801e9bf7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08feca12b27946ad80d298ed8e0d15dd : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceec00af756041a5b2aaba6add46e444 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5925dfe6e82449ab3f86a5fd723f1b2 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c0b93d560d214c37a731f11bdaf8c6bf : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29c3984e6ecd4310b145055b0c1f4397 : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0d412e12bf54e999f01aa64c405685e : SUCCESS in 4m 49s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/85aab15bb4f44f1b85db6f9777b7f086 : SUCCESS in 5m 21s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/e34347ab10004545b9b1ef60cdddb3e3 : SUCCESS in 2h 31m 10s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/f0b79cf0a71a42e99d05b5b8b2a1c575 : FAILURE in 54m 46s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/8d1d82e89f9a4510aaab3ddc97ea4dae : SUCCESS in 2h 32m 00s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"19bb084ff01afb668f98806d6f3eeccb5a98e0cc","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-08-05 01:18:18.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cde357487cbf9858946787b64d9366238b27b69d","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-08-05 01:23:50.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5ac8c53ee36bbada473fba30e024e077fa32e1f4","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-08-06 03:20:09.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e01b7b1e785d71810360255c54c187503448449e","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-08-06 03:22:59.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"655c8deccd7400a808ddeb8f004384770bba8533","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-06 03:50:00.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecf96246709b4941b62aae9da300562e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/815a523f8def4afbb5f64b6d005da190 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/534271839f7246c987e63aedefed48b6 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5bc21d273143433a9fb5c4e1c9164a41 : SUCCESS in 18m 09s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"18249e54162c6dcf136a98edd2407f79ae724d96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-06 06:02:06.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09b5b4e4b0e24b7ab68dae0b59d1c6ca\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33d8bf2f5d5447e9a6a76c0c3680c5dc : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af93e7b8061d48d693515d22b8a3bf06 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bcac0c1181da42079132709c1a6bde74 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41847364e23c4804ba8fb15a3beb200d : SUCCESS in 5m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/decd2121d02344599d0703035318f54a : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/308cacc4a8de405db2f7c4e9b225c1e3 : SUCCESS in 4m 56s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/22a48aac97714009bd389a02535d8102 : SUCCESS in 4m 37s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/d88ace8d80db48789dfb4f4016437429 : SUCCESS in 2h 37m 33s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/f50c6b8b01734effbc4cfcdbab9510a9 : SUCCESS in 1h 48m 59s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/224cb85539154c8586b4e273c7c59968 : SUCCESS in 2h 12m 29s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0ed718f0f45d60e15904aad3a7ca8ffac7469fd5","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-08-07 02:11:11.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7b9aa20c5694bcaee46efca138cb3001b70a3b39","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-08-07 02:20:12.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"868f2ad3f3a4faa11e118afebefe8abf8272e210","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-08-07 02:31:21.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a1ed81ac8a72a86280e911a046681e294d2754df","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-08-07 02:42:54.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"2a7b6906c7f1e83821d7895bc5cd19b2be63a81b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-08-07 02:47:04.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"e3759c5220c727bc9713b213f8a81cc41304eec0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 02:54:37.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab07ad0cda264ce883df5f69b736339e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/616297cb27034d708b14be296a6a238e : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/55beb1b010b74e3cb5d07aaeb971e826 : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dc1e72ee0af04c4b9727c97e72ebe788 : SUCCESS in 5m 47s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9247b88363cdaf436cb01e2a4b7a5a03758922e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 05:04:07.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e75d81df2ab4a2e9d31edce8fc4f991\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8183566c7f3c421f8d1b20c16dc23f97 : SUCCESS in 7m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62f268f61b2040ee9a23b3f26235d80a : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9741e2e6036048c28b66a0e64588fe5f : SUCCESS in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/242a1f26b0a54fb58e600ab8366e9051 : SUCCESS in 6m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d246fcd210445f1973053891c93175e : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbe957bc40084bf1847ac39cee117076 : SUCCESS in 5m 37s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1f2b000ed8514ed8862df35c539de40a : SUCCESS in 6m 26s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/2cb7b5d1dbd442fa859e4aad6eb5bf7b : SUCCESS in 2h 15m 45s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/189b3a59b9ee49a89f584b54e960b1ec : SUCCESS in 1h 31m 50s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/76cfe733c4f2497284d5216fbebced14 : FAILURE in 1h 41m 43s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3954b55cb0cc1110bdff052e285cb079b93b191b","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-08-09 02:41:03.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"5eba215c8cc95bfeab261cb939138a09b582d64e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 02:41:50.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"1b8d40c0fda7aa44b86d7aedbe7d1aa5fc06f8b1","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-08-09 02:48:04.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7483a57bdbb8fc04b2ba6500d041f39094389ae4","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-08-09 03:59:53.000000000","message":"Patch Set 6: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"710b23ea7862cd8c24952de4baa18825b2984cf7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 04:00:05.000000000","message":"Patch Set 6: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":6},{"id":"8c629936384e36fc4d7213503165d3bcbb8de94a","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-08-09 04:01:23.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"39af6ae14f1ce015b266df03408247ce747a72bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-08-09 04:02:11.000000000","message":"Uploaded patch set 7.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":7},{"id":"5b6d8051a90bdaa5f92470741eaaf669aaf6b9c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 04:16:21.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/529502d2c686403da494faeb5a578a1c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fe6901369aa74280b622b166aefca6e6 : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/86d41f4b9ce1433981c1f035dae0d3fc : SUCCESS in 12m 19s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cddb2f7df9b945f080234be70cad34f4 : SUCCESS in 4m 57s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"276454b469a1d0a1bf920867bac79da8f1f0c9b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 06:15:24.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fda91eec7e904fb88ca6912ef116a33e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f20dbca043994bbfb9676e3a9162f253 : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82016e4e322e41988bc0a0799de57ff8 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ce5e577cf4f4907b2150fbfae3f360f : SUCCESS in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/02553edb32404a29b7052984241d8d38 : SUCCESS in 6m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2e72334eaedd4ee792ac1d92d273561a : SUCCESS in 7m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca687b3d34724864ba38a1daf6e8e5ea : SUCCESS in 6m 39s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0a999eb380314ae1b40a3c7e54f42436 : SUCCESS in 5m 27s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/4ff8b22a3d0e484d84d620b056513116 : SUCCESS in 2h 12m 07s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/af7b78ed9aba4c84a1a791c695474547 : SUCCESS in 1h 29m 48s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/43be85f7a68746f3884728c2f3dc2990 : SUCCESS in 2h 08m 51s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"c17dcb351a880dbec9d8b469bbac07cf13f7059b","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-08-09 06:46:29.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"fb0e58616e0408854d275cf287a4512080de4ba7","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-08-09 06:54:32.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"94c5ed6d6c487dfdbd7b75c83c96cec3ef823280","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-08-09 08:39:37.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"33703174b59cdb571434b6fdff55dcb6eee1c5bb","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-08-11 07:14:48.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"5601053eacd95a36cfd33cfa214042268525c02e","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-08-11 09:00:32.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c062e825ecc8dd2d5bfbc1f422256b210e23de04","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-08-11 12:56:17.000000000","message":"Patch Set 7: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d9f9a08c81b101cd2585691bd7ab882b503fcada","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-08-14 02:56:18.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"3b44b34a3349dc592a85543a0cc9b26291835a7c","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-08-14 03:01:46.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e08fc2869afdc0f04231bb49b5318575795203e7","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-08-14 03:09:08.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"da8974bdb32f34b3a6b996c8d03c76bc359868c6","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-08-14 03:11:42.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d67704ee9f205351f330de78241276864ada0d42","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-08-14 03:52:37.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"7a7f45acc71237f917201383336cd6c9ff873e9c","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-08-14 06:32:13.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"fb9e35d403ccccd480231386a75ad1c30d7f3fad","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-08-14 06:36:04.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"4b0cee5cebf814b63808180b594e27e0faae7605","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-08-14 06:44:43.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ef190c21a88be2d4b60ad1285297fb166c10901b","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-08-14 06:55:09.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"2923026616dbd0bc15ebc5ed581561163076857d","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-08-15 05:55:41.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"34a416ee3f1e26d7ab0364989e907253ace89b54","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-08-29 09:05:29.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"4d3abc357276a0065a847aae0c18e73340bb74a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-08-30 01:49:43.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"028a09ca345e0b6d9b92d9817754a3d64ac5748d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 02:00:16.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d193ad2cd864c49b447022a959feb25\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7ffcc0268eb94d93b5fa8b6faea713c7 : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/47a7caee7e7e488a90d5d714c2293e16 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/954f6e6b0d514ca9aa4998687a5ff835 : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/976cf7c3e6f842c18528d84de81380b2 : RETRY_LIMIT in 1m 21s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"de6423237346026b9f5262e9b908a2ed24a3d0c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 04:58:11.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff6f1336e9cd4e6abf55e3e00f95a19c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c39589e24ba410b98f141b1dfe8ca49 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f15c3808d964440b03c61941a470639 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc1203bfe6104537afd81575e5bc69e5 : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/18d0e43fd4334d5eaff6fc9216de8a75 : SUCCESS in 5m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2e8a455e109649bc99126065e97e3f1b : SUCCESS in 5m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/96e62abe29aa4af6870618250f8c9b71 : RETRY_LIMIT in 2m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d6aebda6300470dbce7df53dfe2efd8 : SUCCESS in 5m 39s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c2ac4fe2d2e14bf2a613ac454c457f0f : SUCCESS in 4m 37s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/8e509e0c9d6948b582f23253ddf5c6eb : TIMED_OUT in 3h 02m 58s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/19d2ed6c312e4abb8fcfa84ec57bec5d : SUCCESS in 2h 07m 24s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/85fcfd784ddb4f6f9b2f3713dc159f3a : FAILURE in 25m 34s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"c742351c0b12d0086b139a86986a604be60954e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-08-30 06:41:37.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"0e959e4486d0021d6ba86d91e9559c611f7a773d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 06:57:42.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88c10a03811a434d8daac5de68c70ae9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/284b1ba6a32245bb9ebca5217b2b4fe2 : SUCCESS in 13m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b8a2c5f4f8a94671b4e8d9e1d398eb60 : SUCCESS in 13m 28s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/312ec7efd7b440f4b247bb8d4659b8ac : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/abb481a11d0744398434419313588fd9 : RETRY_LIMIT in 1m 33s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"3f17ac1a80ff409a64e0b01267a9b4bc4479ca63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 08:55:45.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c0d971427f1420cae10b57eb10e9eb6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/520149435847475cb2c610c3ab059627 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17ae221d42e94c3cac3b5283171849bf : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b48ffad31f684691a0002def818c6b97 : SUCCESS in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb9ff67f64f44034862bef0ff917f776 : SUCCESS in 5m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4f92f05d05904b84baa4fba972f64382 : SUCCESS in 5m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/61ada6d483b34215b530fcfeb1cbb433 : RETRY_LIMIT in 1m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfa77baeb0cb469095728966d44f1a6f : SUCCESS in 5m 54s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/83a05b57729040889aeb12d6f765786e : SUCCESS in 4m 39s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/211c959c4ffa4f9cad56d6b1b80b5d1f : SUCCESS in 2h 12m 14s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/92d1c44835394895a7b5394665ec340a : SUCCESS in 1h 31m 18s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/790442e0e34f4c0b8dcec5ebbb808fa3 : FAILURE in 2h 08m 27s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"a5619aa220e3ed0b325a3a26d191e32e3bce841b","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-08-30 09:44:00.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"48cf077838ce83177769ebf623f790dbf2c7aadf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-08-31 01:55:25.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"16c5c1b1c9b08159bec47b802e75b71df43c9caa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 02:12:55.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e8a9295b93440cd8da968f36a48b8fd\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6594c50982f94966a4d3141f5d81bca8 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2fface72210944c18e14082aa213157e : SUCCESS in 15m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/816de67558d34be180bf1ba7fc94752f : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/268786409c9f4a1f9382e61796fee644 : RETRY_LIMIT in 3m 15s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"9847aaa23e67cded60c415ea6627b647d077a20f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 04:28:54.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9ec0845494244c59140d72e35626d64\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/017aafe93874428cbd5da30120bb1b4d : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4056e274a9f145458b1dc923515f5727 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/baeade8e0d624e42ac3b7e3ae81e9d64 : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb970909348b400c96fab71b4581e772 : SUCCESS in 5m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d786b4a2b8784cf4821c79bbfd2bcbeb : SUCCESS in 5m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3c7206e127274b9da42a8eb6d3363b70 : RETRY_LIMIT in 3m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8391c2342ec944b087d523b7093a9221 : SUCCESS in 5m 04s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ff28f3a2d77f43818f5f525c5a1bafbd : SUCCESS in 4m 24s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/08391149229c4dd9ac016d425202d4be : SUCCESS in 2h 32m 26s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/60e7c631a10d4f2c97164f3ec534ba23 : SUCCESS in 1h 48m 38s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/80a4f8db85c147c48ba4b3b201c58de4 : SUCCESS in 2h 27m 08s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"440be4ce780e3d565203640d9e386fa5acd53c50","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-09-06 02:59:59.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"40ce1cdce7c2287ee02f3cf6dcd1a1fe04070ac2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 03:00:29.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"9b8a5130c27a6328d804986d46ece99e2705939c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 03:09:47.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f701e13c8da430492d0ea300b8a11ab\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39bb6ec94d2545648a62d9e87e636fd8 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe406ee9c42343bda644ff850f9cfdfd : SUCCESS in 8m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7ba425acf944e7ca674be840d87784e : SUCCESS in 6m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/45be5078fb0142ae817b3a8ff54d4429 : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3fc31cedefc410d83f1b90fbe9845c5 : SUCCESS in 5m 50s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/1cdd3a0dd8c34af5ad1cf3b52a664eaf : FAILURE in 5m 10s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"804dc455ed223118f7fcd11e54f0d784010bf501","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 03:09:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"2a5074489c782bc3efd7a469b0bc87605310a3d9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 03:10:52.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82fca1765c1b4a06a6950bd59eac25bc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4ba5c028d73438ea532593e0c4ac5fe : SUCCESS in 46s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"5424152b047914512cf81a82d4b94526ef786263","revisions":{"b77c56a622470f96a4d58c3d6ebc91d77024213a":{"kind":"REWORK","_number":1,"created":"2023-07-27 04:06:11.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/89/889789/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/89/889789/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/889789/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/889789/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/889789/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/89/889789/1"}}},"commit":{"parents":[{"commit":"c4efd57e3659a2ac075106835ffd807b1f77df35","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c4efd57e3659a2ac075106835ffd807b1f77df35"}]}],"author":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-27 04:05:51.000000000","tz":420},"committer":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-27 04:05:51.000000000","tz":420},"subject":"Fix error create backup in PostgreSQL","message":"Fix error create backup in PostgreSQL\n\nYou can\u0027t do backup in PostgreSQL secondary node, because backup can\u0027t\nget label from *.backup files.\nCurrent, you just can do backup in PostgreSQL primary node. So I want\nwe can do backup in any nodes\n\nStory: #2010837\nTask: #48426\nChange-Id: I1998f37b6d81fd451c31311072aaf539434dd6dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b77c56a622470f96a4d58c3d6ebc91d77024213a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b77c56a622470f96a4d58c3d6ebc91d77024213a"}]},"branch":"refs/heads/master"},"f432dd659abe54ae7fb6785d22bba610443598c8":{"kind":"REWORK","_number":2,"created":"2023-07-27 07:23:03.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/89/889789/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/89/889789/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/889789/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/889789/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/889789/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/89/889789/2"}}},"commit":{"parents":[{"commit":"c4efd57e3659a2ac075106835ffd807b1f77df35","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c4efd57e3659a2ac075106835ffd807b1f77df35"}]}],"author":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-27 04:05:51.000000000","tz":420},"committer":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-27 07:22:50.000000000","tz":420},"subject":"Fix error create backup in PostgreSQL","message":"Fix error create backup in PostgreSQL\n\nYou can\u0027t do backup in PostgreSQL secondary node, because backup can\u0027t\nget label from *.backup files.\nCurrent, you just can do backup in PostgreSQL primary node. So I want\nwe can do backup in any nodes\n\nStory: #2010837\nTask: #48426\nChange-Id: I1998f37b6d81fd451c31311072aaf539434dd6dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f432dd659abe54ae7fb6785d22bba610443598c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f432dd659abe54ae7fb6785d22bba610443598c8"}]},"branch":"refs/heads/master"},"dd9d7b8156079457b0d75092cbc36272e6418d25":{"kind":"REWORK","_number":3,"created":"2023-07-31 06:55:26.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/89/889789/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/89/889789/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/889789/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/889789/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/889789/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/89/889789/3"}}},"commit":{"parents":[{"commit":"c4efd57e3659a2ac075106835ffd807b1f77df35","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c4efd57e3659a2ac075106835ffd807b1f77df35"}]}],"author":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-27 04:05:51.000000000","tz":420},"committer":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-31 06:55:11.000000000","tz":420},"subject":"Fix error create backup in PostgreSQL","message":"Fix error create backup in PostgreSQL\n\nYou can\u0027t do backup in PostgreSQL secondary node, because backup can\u0027t\nget label from *.backup files.\nCurrent, you just can do backup in PostgreSQL primary node. So I want\nwe can do backup in any nodes\n\nStory: #2010837\nTask: #48426\nChange-Id: I1998f37b6d81fd451c31311072aaf539434dd6dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dd9d7b8156079457b0d75092cbc36272e6418d25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dd9d7b8156079457b0d75092cbc36272e6418d25"}]},"branch":"refs/heads/master"},"8eb186f7e1a46e25f780ef8687fcd9606be478e5":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-08-04 06:19:10.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/89/889789/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/89/889789/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/889789/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/889789/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/889789/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/89/889789/4"}}},"commit":{"parents":[{"commit":"651cee350fa74dac901a833c4e6bf4b11b6f1b55","subject":"Merge \"Fix postgresql database creation failures from prepare func\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/651cee350fa74dac901a833c4e6bf4b11b6f1b55"}]}],"author":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-27 04:05:51.000000000","tz":420},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-08-04 06:19:10.000000000","tz":0},"subject":"Fix error create backup in PostgreSQL","message":"Fix error create backup in PostgreSQL\n\nYou can\u0027t do backup in PostgreSQL secondary node, because backup can\u0027t\nget label from *.backup files.\nCurrent, you just can do backup in PostgreSQL primary node. So I want\nwe can do backup in any nodes\n\nStory: #2010837\nTask: #48426\nChange-Id: I1998f37b6d81fd451c31311072aaf539434dd6dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8eb186f7e1a46e25f780ef8687fcd9606be478e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8eb186f7e1a46e25f780ef8687fcd9606be478e5"}]},"branch":"refs/heads/master","description":"Rebase"},"116e9de8746fcd1fc07427f6f83baaf780f11606":{"kind":"REWORK","_number":5,"created":"2023-08-04 09:44:01.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/89/889789/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/89/889789/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/889789/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/889789/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/889789/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/89/889789/5"}}},"commit":{"parents":[{"commit":"c4efd57e3659a2ac075106835ffd807b1f77df35","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c4efd57e3659a2ac075106835ffd807b1f77df35"}]}],"author":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-27 04:05:51.000000000","tz":420},"committer":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-08-04 09:43:43.000000000","tz":420},"subject":"Fix error create backup in PostgreSQL","message":"Fix error create backup in PostgreSQL\n\nYou can\u0027t do backup in PostgreSQL secondary node, because backup can\u0027t\nget label from *.backup files.\nCurrent, you just can do backup in PostgreSQL primary node. So I want\nwe can do backup in any nodes\n\nStory: #2010837\nTask: #48426\nChange-Id: I1998f37b6d81fd451c31311072aaf539434dd6dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/116e9de8746fcd1fc07427f6f83baaf780f11606"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/116e9de8746fcd1fc07427f6f83baaf780f11606"}]},"branch":"refs/heads/master"},"2a7da20c83e583066f978c424844b9683c062f47":{"kind":"REWORK","_number":6,"created":"2023-08-07 02:47:04.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/89/889789/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/89/889789/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/889789/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/889789/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/889789/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/89/889789/6"}}},"commit":{"parents":[{"commit":"c4efd57e3659a2ac075106835ffd807b1f77df35","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c4efd57e3659a2ac075106835ffd807b1f77df35"}]}],"author":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-27 04:05:51.000000000","tz":420},"committer":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-08-07 02:46:47.000000000","tz":420},"subject":"Fix error create backup in PostgreSQL","message":"Fix error create backup in PostgreSQL\n\nYou can\u0027t do backup in PostgreSQL secondary node, because backup can\u0027t\nget label from *.backup files.\nCurrent, you just can do backup in PostgreSQL primary node. So I want\nwe can do backup in any nodes\n\nStory: #2010837\nTask: #48426\nChange-Id: I1998f37b6d81fd451c31311072aaf539434dd6dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2a7da20c83e583066f978c424844b9683c062f47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2a7da20c83e583066f978c424844b9683c062f47"}]},"branch":"refs/heads/master"},"9e9acf58122df1e33d42d6d1a7fe914df225cadb":{"kind":"REWORK","_number":7,"created":"2023-08-09 04:02:11.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/89/889789/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/89/889789/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/889789/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/889789/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/889789/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/89/889789/7"}}},"commit":{"parents":[{"commit":"c4efd57e3659a2ac075106835ffd807b1f77df35","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c4efd57e3659a2ac075106835ffd807b1f77df35"}]}],"author":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-27 04:05:51.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2023-08-09 04:01:53.000000000","tz":420},"subject":"Fix error create backup in PostgreSQL","message":"Fix error create backup in PostgreSQL\n\nYou can\u0027t do backup in PostgreSQL secondary node, because backup can\u0027t\nget label from *.backup files.\nCurrent, you just can do backup in PostgreSQL primary node. So I want\nwe can do backup in any nodes\n\nStory: #2010837\nTask: #48426\nChange-Id: I1998f37b6d81fd451c31311072aaf539434dd6dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9e9acf58122df1e33d42d6d1a7fe914df225cadb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9e9acf58122df1e33d42d6d1a7fe914df225cadb"}]},"branch":"refs/heads/master"},"6b6c23d10b810da1a8477435347885c5c3f86f76":{"kind":"REWORK","_number":8,"created":"2023-08-30 01:49:43.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/89/889789/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/89/889789/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/889789/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/889789/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/889789/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/89/889789/8"}}},"commit":{"parents":[{"commit":"c4efd57e3659a2ac075106835ffd807b1f77df35","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c4efd57e3659a2ac075106835ffd807b1f77df35"}]}],"author":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-27 04:05:51.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2023-08-30 01:49:26.000000000","tz":420},"subject":"Fix error create backup in PostgreSQL","message":"Fix error create backup in PostgreSQL\n\nYou can\u0027t do backup in PostgreSQL secondary node, because backup can\u0027t\nget label from *.backup files.\nCurrent, you just can do backup in PostgreSQL primary node. So I want\nwe can do backup in any nodes\n\nStory: #2010837\nTask: #48426\nChange-Id: I1998f37b6d81fd451c31311072aaf539434dd6dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6b6c23d10b810da1a8477435347885c5c3f86f76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6b6c23d10b810da1a8477435347885c5c3f86f76"}]},"branch":"refs/heads/master"},"476c6a94500da69f9a47b985b006f86e4a8d2a57":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2023-08-30 06:41:37.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/89/889789/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/89/889789/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/889789/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/889789/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/889789/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/89/889789/9"}}},"commit":{"parents":[{"commit":"fe13333a6107bd4aa45a0764a81b51ad48246121","subject":"Merge \"Change StrictRedis usage to Redis\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fe13333a6107bd4aa45a0764a81b51ad48246121"}]}],"author":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-27 04:05:51.000000000","tz":420},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-08-30 06:41:37.000000000","tz":0},"subject":"Fix error create backup in PostgreSQL","message":"Fix error create backup in PostgreSQL\n\nYou can\u0027t do backup in PostgreSQL secondary node, because backup can\u0027t\nget label from *.backup files.\nCurrent, you just can do backup in PostgreSQL primary node. So I want\nwe can do backup in any nodes\n\nStory: #2010837\nTask: #48426\nChange-Id: I1998f37b6d81fd451c31311072aaf539434dd6dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/476c6a94500da69f9a47b985b006f86e4a8d2a57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/476c6a94500da69f9a47b985b006f86e4a8d2a57"}]},"branch":"refs/heads/master","description":"Rebase"},"5424152b047914512cf81a82d4b94526ef786263":{"kind":"REWORK","_number":10,"created":"2023-08-31 01:55:25.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/89/889789/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/89/889789/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/889789/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/889789/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/889789/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/89/889789/10"}}},"commit":{"parents":[{"commit":"c4efd57e3659a2ac075106835ffd807b1f77df35","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c4efd57e3659a2ac075106835ffd807b1f77df35"}]}],"author":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-27 04:05:51.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2023-08-31 01:55:06.000000000","tz":420},"subject":"Fix error create backup in PostgreSQL","message":"Fix error create backup in PostgreSQL\n\nYou can\u0027t do backup in PostgreSQL secondary node, because backup can\u0027t\nget label from *.backup files.\nCurrent, you just can do backup in PostgreSQL primary node. So I want\nwe can do backup in any nodes\n\nStory: #2010837\nTask: #48426\nChange-Id: I1998f37b6d81fd451c31311072aaf539434dd6dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5424152b047914512cf81a82d4b94526ef786263"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5424152b047914512cf81a82d4b94526ef786263"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
