)]}'
{"id":"openstack%2Ftrove~889797","triplet_id":"openstack%2Ftrove~master~Ie0268951db7a9e711089714cecf7fbf311eecec8","project":"openstack/trove","branch":"master","topic":"change_uid_gid","attention_set":{},"removed_from_attention_set":{"28691":{"account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"last_update":"2024-08-17 06:48:24.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2024-08-12 03:31:54.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"Ie0268951db7a9e711089714cecf7fbf311eecec8","subject":"Add UID/GID configurations for each datastore","status":"MERGED","created":"2023-07-27 07:09:42.000000000","updated":"2024-08-17 06:49:28.000000000","submitted":"2024-08-17 06:48:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":46,"unresolved_comment_count":21,"has_review_started":true,"submission_id":"889797-change_uid_gid","meta_rev_id":"ceace4116ff1a8dcfe267941886966971e7db37f","_number":889797,"virtual_id_number":889797,"owner":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-17 06:48:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-08-17 06:06:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-08-17 06:06:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-27 07:28:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-27 07:37:59.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"CC"},{"updated":"2023-07-27 08:07:50.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2023-07-27 09:52:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f232715e9e9c2eed04ece01d3b9d20989f616aed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-07-27 07:09:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a32bae902afd9c8dbd7affc72c90d1bb2676a979","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 07:28:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40a19a3b1d94434b8c260b3ea841429f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a2ad1bcf9b4645aa957118063d5884dd : SUCCESS in 16m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7fc60889657a443ea3986c2f620759a6 : SUCCESS in 16m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7f56cfa985c444328ffab4ab9d011133 : SUCCESS in 16m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5fed6466139850f0bab27047d216dd8ff75b7512","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-07-27 07:35:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"be03853c50c9ac726fd97fe427a720f1db91d94e","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-07-27 07:37:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bafd8143d78ff70509132cb3b6af047759303864","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-07-27 07:49:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d487dff983ede5a6d74c67f2b93f1bd9c5756d6f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 07:55:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5a626fe0d7c40e59117c10948ba84e7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/136c48c47f304a80a60f685ac386e62f : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e5a7efc765f245b5a8cbbfd4750c7af5 : SUCCESS in 14m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/83ca1a4aa7e942398a64f5129e8397f5 : SUCCESS in 7m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d0fc4170bc2eb7c674626dfa72b4c7c0972af0dc","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-07-27 08:07:50.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4341befde33a82e990719ba024e0366279126fc6","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-07-27 08:13:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"be785ae04c9402908cb80cfa849657df291f2035","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-07-27 08:25:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"55decb4bdfa4b1c3108ac18902f5225102629afd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 09:52:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/486500cfdfdd4055a54b58947c9c108b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64afb32dc10542949a1bfc9d7fa5704f : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46c01df19a014745b0566031b1a8d33e : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fae46b0db7f64e2f8e5a1dae9b858bb5 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/17c80ba566cc4945a80d228703e632ca : SUCCESS in 5m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/00a85e2f787942deae135aee0a29e464 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de9edc2275d74e51823472b5f1dccd59 : SUCCESS in 7m 18s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e85182dea3e04e77947d60df43f4fa78 : SUCCESS in 5m 11s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/9c078ff8426b4719bfa297042d2dceff : SUCCESS in 2h 15m 19s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/7bc0357de3ff44518d2d8fdb1319006f : SUCCESS in 2h 14m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a0ac4f67e07efb83df05533a4c45ecc257b6daa6","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-04-02 03:19:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3f642e65305b8c6d1918f5beb7a4371500023410","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-04-02 03:19:52.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"8c8e5e17e92b68e8b50cea3457376634ac7468eb","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-04-02 03:22:46.000000000","message":"Patch Set 3: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f636eeac475b5c7fc63fac3bc00fe14dad008d45","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-04-02 03:27:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e90749536447db59fda81c482cc1c6486b4f3299","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 03:35:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ff673db274a45648dfc8a511544bdb7\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b43ca2731e9042a8b6795b0df973de4a : SUCCESS in 10m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a5b36cbe9f3045e69cb2a00e6cb9c098 : SUCCESS in 4m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ddbe634cbe911d3dc5c993f9a456a531881c1998","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 04:16:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74c0a662e81545fabc1b87b19b202055\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3900c33829a441b98242d2fd5c9160f0 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d70e5d7b779e4ab0bc4cb5a6eaa7d285 : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f9fdef6444744a69a1f9e2a1bac77151 : SUCCESS in 5m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1ed20faebbbc41e28aa69c7c487b361b : SUCCESS in 4m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/95ee53ff378344f2936b9e36064784d0 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/857003d3ba9148cebef90e87385d4c93 : SUCCESS in 5m 13s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/f43cc10185c544cb9117d4df826673fb : SUCCESS in 42m 36s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/0764730ff087487d94c3012a22d4e1f2 : FAILURE in 32m 14s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/520ebd5f60d041f79c8f0aa0bd7cf954 : FAILURE in 42m 08s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/3acc9ca07c414d2fa5dbef89acf4d2b8 : SUCCESS in 39m 03s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/084e31e69d18495e95bc359c52b45fc4 : FAILURE in 41m 26s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/15f8f293c7a8451c9a1a4c4b71ba6d0b : FAILURE in 29m 56s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/49c2da00b9b4449fa830bee2ee4c668e : SUCCESS in 44m 50s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/a40969f4e35c4be9bd87832b55c00423 : FAILURE in 40m 40s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/b779f3a765934a4d8e994c9e7c80fa39 : FAILURE in 29m 57s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/072e0795104c4fa389c88ea2a30a83df : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":3},{"id":"45eae29e245fdbb730687310270b0124b296bc85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-04-02 10:00:00.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"812f879ff739262162d0bb154686cb7eafc0779c","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-04-02 10:00:14.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"87492e48d153465648a8737be798f02428266a4c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 10:06:25.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2dc7184c3d8144b786b493949d99e4a6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/39e779cb8f544d18a047cee9da756923 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/52af866794624dcda08059ae27e6d99b : SUCCESS in 4m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"363bf9ec68d3d8bef154d830d6e84ee84f528252","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 13:07:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c62597ac48b94716914d6a17643ecd94\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8031034cae949febd196746ce5805e2 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53fb8aac03b2446983e7b4e5c4aae369 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c218ef19abf64fb8836eb43511f06d89 : SUCCESS in 5m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/09999a1666e144e1b257de9c21e228b3 : SUCCESS in 6m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b92efd7a09904a0fa4dce45e457904ea : SUCCESS in 9m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53a062b66058421eaae690da17fe2497 : SUCCESS in 4m 53s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/bd698bd20f5c47219c94319ede44ab6c : SUCCESS in 1h 06m 32s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/2336813f1d9745928e5488c05856c39f : SUCCESS in 2h 29m 02s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/96da7b6382b64ae9b0aa66b7437369ea : SUCCESS in 1h 57m 41s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/0f36e56c55ba490893cf0c8a7ea663a7 : SUCCESS in 1h 06m 20s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/fd2fd8c7b3c042d891d4dc93409c8923 : SUCCESS in 2h 07m 40s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/27998e8f4e924b6a9d8f104257bb5d16 : SUCCESS in 2h 09m 01s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/111e4eb1706a4582a51346ca04e88f87 : SUCCESS in 1h 13m 59s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/60895f9e1c5e49199dea6484505209c1 : SUCCESS in 2h 16m 24s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/06061c8c436647829132fe3d7a43bd78 : FAILURE in 1h 19m 23s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cfab86956b5644c98073874b47eb3e29 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":4},{"id":"a671837a9205f0b62ebf8c51bb809a52bb491b67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-05-13 01:58:55.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"998246896ec041a04c34bef1691327d5d7edcbbc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 02:05:57.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88696e6cbf0149e9971436d15d388133\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e5103f62bb204b3688e91b84023ae1bc : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e5be91cb17f449ecbb7e7fed91c9e825 : SUCCESS in 4m 42s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5503ab41d7b51b40fde64771bff2bfde04a5441a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 03:57:38.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb7a4204ae0a4929835807f1ba23266e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb561bede9784ee7b1bdbe177095c3c3 : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/852b73e09c554eb99813c04a0a97f39c : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93e7fb5e101e47178d51fee29f04a058 : SUCCESS in 6m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/67c1e4e5aeb949d98b9011de8647828c : SUCCESS in 4m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0e85550beea843348e7e3c05328770ab : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a051da74bd6451c9de0430c6c4c31fd : SUCCESS in 6m 00s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/f9adb471a2e94d7db158ae801bf4eb57 : SUCCESS in 48m 20s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/8bf5be92073240a0be6938869e650041 : SUCCESS in 1h 57m 21s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/3504394b0bfe4694b699d837cefc09bf : SUCCESS in 1h 43m 52s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b9e8e37e52854e0d87a7f1ff9ad1281f : SUCCESS in 49m 55s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/62ff50edd1dc4f2db303a4b07b03cb94 : SUCCESS in 1h 51m 03s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/53d4a8f23c224797aad69036b8d7f33b : SUCCESS in 1h 22m 45s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/dee78f3fad3c42db95937d1f9ad8f7a1 : SUCCESS in 58m 35s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/8611faec488e4b329c0585d172741d7d : FAILURE in 1h 16m 28s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/b1998c19dfec4023af77c077982b7a74 : FAILURE in 1h 10m 22s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/deadb59c6d794ddf8a68fad83d7559da : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":5},{"id":"6d2e13d38cedc7294f432595dd687792361f9aa4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-05-21 07:34:55.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"5001e5ff1125621f07d0693856e908978e484fa9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 07:45:02.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7817a48a9ec64815bbb72c75ab26bd62\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0cdaada7a57044359c5a96bcc2c1f6de : FAILURE in 4m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/859e3d45c04140c4a22263fa6de17795 : FAILURE in 8m 25s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"aa1b021416fc8972d7a83dc86e9e2c39d40899b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 08:06:26.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/46aa656b9dd746a9b4f89d516f85d558\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc1356b7340046aabb1d6a84087e7ac3 : FAILURE in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c278811f79de4550826ba962a1f35ed3 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab4c5b9ee48b47c79927ec44a274b3f9 : FAILURE in 5m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f0f83c0005d34383a8166bd19cce7f28 : FAILURE in 4m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3a8358ac3fe54b05a22ea14e25bd5ce7 : FAILURE in 8m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8de2d488f2af47c7aa66dc3daf9dcc6e : SUCCESS in 4m 36s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/2e7e8440835a4c0abcf633f4dd5bf593 : FAILURE in 18m 07s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/2579bd93a54b4bc19780907d18c2804f : FAILURE in 19m 49s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/478538c38b684d1da1354f02efce5360 : FAILURE in 19m 42s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/7d583f2187c142939ced5bbf5e090ff8 : FAILURE in 19m 10s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/7a86323f281540429851a66b850ced86 : FAILURE in 21m 18s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/cf7ccea8b9d34b5a85f23edb0b392c45 : FAILURE in 23m 54s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/8f4bfc6b8a054e46b9b78112d7b4eeaa : FAILURE in 20m 34s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/2251ea0b06e2476193cb16170f552e98 : FAILURE in 21m 21s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/c09e0a03a7db4e4d81775cb1df1ecfe4 : FAILURE in 20m 25s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cd4911377d0345fbb18f0afa0350e7ab : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":6},{"id":"b3153aea475c76bf6747fc851b6ffb28e96cfc35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-07-11 02:29:40.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"71fbc6a53ff9952b259e6c45e0824440ae0b6244","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 02:46:05.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aec5a40ee56245768a6cbb6945ca2a46\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/099483ad8d0d41b7a48ef1b6054d503d : SUCCESS in 14m 32s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b2607f2ee2534ad0a629efdf74b91c3b : SUCCESS in 10m 31s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"822f8dbb4866b46136f26b4e78cf6fc947a29405","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 04:35:31.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d43d82d5b11489f971d25b2fd965a50\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7ff8de8ba76454198c724fb4219d6a4 : SUCCESS in 7m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c781cf596f74e9caff16b19a638b877 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee2df0e8cab246878dfb7c0622d4319c : SUCCESS in 6m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c8c6556ca1f947b1ae0439e0ca1e022d : SUCCESS in 4m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b71eed275f6748c09e77bdafaa45c85d : SUCCESS in 9m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d84665a30b4749628ea3feb5fcc260d2 : SUCCESS in 6m 16s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/24d3d66b67ff48e1acbac0eddffbd964 : SUCCESS in 39m 27s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/911eabdb6d4147eaaa22eddf42dc99dd : SUCCESS in 1h 42m 59s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/471cc6b60cae4e7aa01c76dc5959eead : SUCCESS in 1h 20m 12s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/c423a9f24e634926a50095a5e87a9e65 : SUCCESS in 43m 20s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/520cafb0d66a49d3a33f45fb54e02ba4 : SUCCESS in 1h 30m 57s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/9c09d3e4634a411a8fb5e7f0854e04f1 : SUCCESS in 1h 16m 15s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/a67ea4dc2a9f43e38267d34c8c9aee94 : SUCCESS in 46m 27s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/37343ae92e06451892d4ab8defbf92ab : SUCCESS in 1h 30m 13s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/a5395531552b48c3a29125a544d96f11 : FAILURE in 1h 10m 42s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c285a4fca6b843fca609ee0295c12eef : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":7},{"id":"e1d594de439cf76fe88f8ed7c2628e804652d9dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-07-17 04:23:11.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"75ebdd932b5581d285e1f0616b57a4542c16072f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 04:29:14.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/357bfe41711442b5a6cff1400daaa38a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ccee9af00e4642ecbcb9383a10f3033c : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7bfc1bdf11484a7ea9fc06ba472fec4a : SUCCESS in 4m 56s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"d575feb6c6e3727b57b762b4fbeb4355ec308950","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 05:03:47.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/de70eb11fdca490a95d834ccdc8ed415\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55e9c3c6bd784e008e55f0613600c073 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c6ddd0917d7485fb0ca8eaf8db7e825 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/779eea64bcc1497f97a0ae01a3dd0122 : SUCCESS in 5m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fb4fd3f883a645148903802f83338d63 : SUCCESS in 5m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9c981c400bfb44d5a932ff7969e0e1d5 : SUCCESS in 8m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e333e2e71554e25bfe4b206179f60c3 : SUCCESS in 5m 04s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/444b72211fc044239363153edf2648cd : FAILURE in 26m 20s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/f403d75a171e4d19a0519fa505094cd8 : FAILURE in 38m 00s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/4223392b80d941538fbc064b2d32f0e8 : FAILURE in 36m 20s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/26da2abd57594d3c9b0d403278c765e8 : FAILURE in 28m 07s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/aa3120dc63f04e2c86a269b34aa638cf : FAILURE in 37m 24s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/9337b45a0ff04039889115604e3335aa : FAILURE in 37m 27s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/d42c26060cde4134b45a2c070018a0eb : FAILURE in 25m 04s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/8184124bd64a4d29bb8e7232bd222c2e : FAILURE in 37m 38s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/4f9e993eedd04f02a483a905a18de067 : FAILURE in 38m 58s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f35f64ad669e4f7b9d60ecc3c2416e31 : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":8},{"id":"e88d65cfc4e9116209c05e1c35c7aff4c3947e5e","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-07-22 10:15:04.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"1979675837edca3a7af4eef53f744557fb774693","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 10:30:05.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ffca6555631436994637eda017ed887\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0cc33997cd844de9a4f7882d5ceacb4d : SUCCESS in 13m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9af99b1bfbea46179bf6372960308cae : SUCCESS in 4m 46s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"55d03cb8be182022eadcbf3407ae9315a23c9f15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 11:05:14.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f9afee43175744f48e9c58669a1868ab\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b6130ab049647f583660c99ca178e09 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/719cd491a518471dbc446ccae24b636e : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1addc3b2e3b0445592628a1182335765 : SUCCESS in 5m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/97718e8b08cb42d795a86dba1fca3045 : SUCCESS in 6m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5fb9897a592146b8ad65e20020728793 : SUCCESS in 11m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d255cc9092840f1a1884da3c2b0254e : SUCCESS in 6m 49s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/3fd36ef4cdea47f3bf94ee69f6e677b9 : FAILURE in 34m 26s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/22a3132a99794db684e771be800fdc63 : FAILURE in 46m 38s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/a29ec257054945918a97ecf4a0179924 : FAILURE in 38m 02s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/be838f29df8f4aef8f35b9e025e4ad24 : FAILURE in 26m 52s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/02f88c5dcb56409c94e3b50a4e0cb2d8 : FAILURE in 40m 37s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/44eb93a2e52e48ca8689b6d4844173ce : FAILURE in 38m 22s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/15a4caae03b848519f8cea0fc5a8533e : FAILURE in 25m 24s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/f286c82569fb4778b5e416ce9e47a0b4 : FAILURE in 39m 00s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/ca0c2acde99e425c8821422fcb83220a : FAILURE in 37m 29s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8df2918ad88d48ee9bd54d1ddecc6031 : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":8},{"id":"29ed1fac20618bd415d9426b2f3a400b6a080eec","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-07-24 01:24:59.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"785a66b700a414fa20c19296ebe62ff9c1d50a21","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-07-24 14:16:28.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"d28c359f1719f380b212c39772506944e27caac3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-07-25 09:57:07.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"bcb53a686faf8004b3b5bcab032b2685cbd735d0","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-07-25 10:01:38.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"4e6f89dc8ee20d800d0f6b7a49ba7ac731db008a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 10:05:13.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab391b57242140d2b2a1d603d5b096de\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f5d3a44b02a445119314dfae0f499384 : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ff56c0361974471995a9fd842f99a998 : SUCCESS in 4m 25s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"95aae934c48213d1f53c66e03224b6c9dbfb972a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 11:42:13.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99f66dccb2694a968c60d96a4f95ccc3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/465c05c9cb9247eb983ebbb84e51dd65 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/172c2bc9f23d4d6883fadba93c48a0cc : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4cc9977560a4cbdb18fa00fd3166705 : SUCCESS in 5m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ae585091afc34fca9ae58796a12892dd : SUCCESS in 4m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c0be5fa997934cae8c2c0036d42b81cf : SUCCESS in 9m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2017118a00d44298e362cb3c2e10d16 : SUCCESS in 4m 28s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/034ee18209da4e92bfdcf9b92ddf750f : SUCCESS in 46m 28s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/753a90f1c5dc41c4b0b0c4d30a0550b2 : SUCCESS in 1h 44m 04s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/608abb9a30314f15a1c86963d4ac2373 : SUCCESS in 1h 25m 21s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/0d71a5136b3148dba39da3caeba86c40 : SUCCESS in 41m 16s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/746cfaec4a0b4336a27f7b657bbeb44b : SUCCESS in 1h 29m 38s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/71b50193efb14d6fad0a864196e27bc2 : SUCCESS in 1h 14m 35s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/968d1bf84a09425c9ecec8240f1e08cc : SUCCESS in 44m 00s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/ad3fe7af7c51478a913e93956d4af0e3 : SUCCESS in 1h 27m 53s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/5390c8f3fcc8486c9d764924c00fc10b : SUCCESS in 1h 14m 08s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6de7b4f8e6aa4d1987d66f294c7853a0 : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":9},{"id":"64a00a6590a797d7063947f2c128b8b9be962f39","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-08-02 06:59:10.000000000","message":"Patch Set 9: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":9},{"id":"5e5f840c73897dda07f9806fceaa6cd783934824","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-08-09 04:58:26.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"25ee3c88989933beba660e4a9664e2d23d9389f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-08-09 04:59:33.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"e27fe327a1856113052aef3cc168e3f4828ad403","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 05:17:19.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5406c066ca804f65923d8cdc3c847cff\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5859b2fa785b4391bbcef97535b11d94 : SUCCESS in 15m 55s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fc433d6a60ec42efa2971990b62baa95 : SUCCESS in 9m 13s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"0350c91700387bb0c5ee85b1612a14010a90a8b9","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-08-09 05:55:03.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"67cc080d6c121ea76c469b8efc0ecd7292c85f37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 06:59:54.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87911967982d4399becb61ca1d62fd66\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22845acd2ec548d795f330103684c5e2 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c23d4ff4e5ca40588070c29c6074dc89 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/91f14dea3fc7477db43080643024a51e : SUCCESS in 5m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/720cd61ed86f4dfbb9ce24c2069e6e47 : SUCCESS in 4m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a3bbc19fc71646ba93a356e9b53e49e9 : SUCCESS in 8m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d528d8f57ae4069bd6ac962666eb939 : SUCCESS in 5m 04s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/dd2ec259fca9479496284ffa38061e07 : SUCCESS in 1h 01m 02s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/3dd50f71324f4b4483f2b6f6cd14df5c : SUCCESS in 1h 57m 34s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/8ba2e7713e30472790e3b3c5523b33b1 : SUCCESS in 1h 44m 06s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/75b4f81465c74f70a40b18562d33d800 : SUCCESS in 58m 22s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/9f3f4ef2ec344e4b9c7b0f99079669b2 : SUCCESS in 1h 47m 33s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/9c7e2fc4a8164c62867c62bd02837fd0 : SUCCESS in 1h 33m 43s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/2d425a22eb334b9da98c1a6fb7228304 : SUCCESS in 53m 33s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/2f9a04d1f2a54a81898706ba033ed06b : SUCCESS in 1h 48m 39s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/5a635fa25a564b3a95a9e0cb3aa82c11 : SUCCESS in 1h 32m 03s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b5676b13fd034b40809e56b10f08faf3 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":10},{"id":"a5883fd28330ff853c28105270ead8956f1d8012","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-08-09 07:17:34.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"80b290c56e3b2cc4749da316eb53eefebeb0565b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 07:48:22.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/716deb7503e7435ab29ebfdd25c74499\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/92bb7d7079cf4d0784b5d8eda4302724 : SUCCESS in 14m 10s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/393e1e45c82245b192300e6e9619fe91 : SUCCESS in 14m 57s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"31d3c916f511c373e16d1be3599f8b1953ffe894","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-08-09 09:06:23.000000000","message":"Patch Set 11:\n\n(5 comments)","accounts_in_message":[],"_revision_number":11},{"id":"bf4b3419f83bbb40b2ab997f9f30bf31ce093dc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 09:22:46.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c940fbea09be4686bbba3c27c1cfff3b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57ddc507683d4a8095f17ddb2f2c5a86 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33b89cabeaa142c583ac8a8f27ff9a6e : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7fa1ea8cbb0d4bbdac34503a53974ea7 : SUCCESS in 5m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e950999a81ea427e901f1c91094a1c0c : SUCCESS in 4m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/07a98bb0fa1c4ddda121a3242207319e : SUCCESS in 7m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bfd11d367c1415ba95638973fd5cbe4 : SUCCESS in 4m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6211ec2808dd405bab717748676a092e : SUCCESS in 6m 24s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/65455cb67cd740dc824793a65de2200e : SUCCESS in 45m 50s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/04b575ff405642f1b74515ae8ab53f5b : SUCCESS in 1h 42m 53s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/4caa3f38848c41eb99d6fd1c82a721d6 : SUCCESS in 1h 49m 21s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/7170970c7175436ea009fa17477eddac : SUCCESS in 43m 07s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/58c5cde49fff4649be4cc3a27ccb6f58 : SUCCESS in 2h 02m 02s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/32ab1b196a15494d8eb95e4f27f2742d : SUCCESS in 1h 49m 58s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/6b9ab0e0526941c091a3602fb1a4225f : SUCCESS in 56m 12s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/bd16860765784d76a2a5c0b94d29af3a : FAILURE in 1h 11m 27s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/4636d92a92c64857a37debb3d2768be8 : SUCCESS in 1h 10m 03s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ac260cc891804ba5b88ea225a83e201f : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":11},{"id":"caf8bf403ee60fb43089e9532dd6f5a3f1918896","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-08-09 10:24:16.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"55f36583357201dca02f9b3aa956f7369cc07a1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-08-09 10:25:24.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"98c187e82417155d4f65dc59f986252ab3102e5e","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-08-09 10:28:33.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"bcb84e56249131ca542d095ebdd9946033fd78c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-08-09 10:34:13.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"a35c8a33079df3df6fa40ce43ef79619104cb3b2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 10:47:39.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4be8b0024f744021ba6f0217c6d9cade\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c13bf2a3b66045b2af94ab342f088254 : SUCCESS in 10m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3c2322ea2d78424ea15caca515f2b445 : SUCCESS in 10m 29s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"94d75c5a55026b358c7ce9e1411ae282491b0f07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 12:52:26.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/62cd800b4c7b424caf92b065c7f51fbf\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0fa7cbbd819a40238b2bc6c13e165a72 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72bc8854643e45ee8e12537938b0c983 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e85c99e69fa425398235a7c67335372 : SUCCESS in 5m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5bbaa89fe3c44b179a65f5670bb4bf6e : SUCCESS in 5m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/27df4b9972214427ab4b13616413185f : SUCCESS in 8m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb3a56ee62be41ba98a3b160c0080336 : SUCCESS in 7m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9623aa9b8738418c9a6db36730f92285 : FAILURE in 3m 30s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/9ebb42a7d8204f9f91bcc1fc8b5997dd : SUCCESS in 1h 02m 16s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/4173b49b7ebc421a952d34eed161ff54 : SUCCESS in 2h 17m 32s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/e16b0659ae364ee5aa131b8cc6ac6d60 : SUCCESS in 2h 05m 43s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/209960b801f1445d9fea56249c1c79f3 : SUCCESS in 54m 11s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/49c8938edb684c869f5740d18f703849 : SUCCESS in 2h 00m 35s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/be302f94a8a142c9be0f8bbec249d93e : SUCCESS in 1h 34m 05s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/0e1c68cfce7844a08b63797f1b9edce6 : SUCCESS in 53m 59s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/8fd5abebfa7244eb8591aea6e6ab6e18 : SUCCESS in 1h 52m 45s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/66082e4ef4ca4830811ef890db99199c : SUCCESS in 1h 36m 14s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f378a135d24548a092e50d9b2c05c120 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":14},{"id":"736cc231c0b57f6539c8f6d0ad4cf4ff86e12c9c","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-08-11 02:32:03.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"7c688a604c2d822b86e3138b77cb240d6834365a","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-08-12 02:17:41.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"ed017f1a7e77b42faa063b20fb8793ae5f0c9c31","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-08-12 02:20:24.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"197f633faa9fac6b9da73b9cdf4d9641bf47260b","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-08-12 02:45:54.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"4e65b1baae0bdb468df712046547ca769aa6c63a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-12 03:05:37.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bf22120984d4836b59cb9ca6bd25120\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/573490a2d68549d4996f17865cf32738 : SUCCESS in 17m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c4be2c37306f403c917c87cb41171ce2 : SUCCESS in 9m 55s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"fd0fdbf85829ea4f87db754d4db64b839ebd413a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-08-12 03:11:05.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"1b15e9035bda63630174351ae353a796d58b44f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-12 03:27:50.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c9d68ca7c034bf594d02c68d6a543f0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bcbe8b1a963d4d6585cf10244738989a : SUCCESS in 14m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9c0ac8f0d3244655bedf838a88af4c9f : SUCCESS in 14m 36s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"27f08dfb3d5164be4d3a90e25123cd869f554a8c","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-08-12 03:31:54.000000000","message":"Patch Set 14: Code-Review-1\n\n(14 comments)","accounts_in_message":[],"_revision_number":14},{"id":"d4ae4a2109697f4b9c710e1287dc7d1543e64190","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-08-12 04:48:49.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"a4b1f41cd4c6f8be615995baba2f6435fe4239be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-12 05:11:46.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fed2a8a164f14f2f83faedff8121a525\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a84deb3b282443a970323428400231b : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f63348daf8b41a5936f2b87ed7dd63a : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d7d76620006485d8e12813a815c2b7f : SUCCESS in 6m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1172c8f43efd485ab1cc5486516674d8 : SUCCESS in 5m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/051b97678594421e87a82d17fea9e7b5 : SUCCESS in 11m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a9ba06acb704144a1990e1a4dca95a5 : SUCCESS in 4m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac451c1fbc254ae398085901d469898f : SUCCESS in 5m 54s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/1160df6ae2494603a9499c8143212870 : SUCCESS in 54m 56s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/c4b8231d093544e5ab91ac718484945c : SUCCESS in 1h 57m 31s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/145caa5dbf1a45dd9aa581df23269821 : SUCCESS in 1h 45m 34s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/c437422bb4d442f7a2e1292dbbf3bd76 : SUCCESS in 51m 22s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/bb7f96bdaa5445de8206c87bfaaf607c : SUCCESS in 1h 58m 54s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/a53c8387eab04d0f9edce80bb3d13ef5 : SUCCESS in 1h 35m 58s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/8059744e31ea4a40af34f6053a39abdc : SUCCESS in 41m 09s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/cf7d4a424748443f8be385e217baca77 : SUCCESS in 1h 24m 29s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/d60b3032fd9e4dedb25b9199b057114d : SUCCESS in 1h 11m 59s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/db62e1c059074fe98cd5908bf7e99e74 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":15},{"id":"ac5695a34f79b800246568df0e745c6f0dcdf679","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-08-12 07:54:06.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"ef3269fc201a4d84054722900c7b16360db374df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-12 08:10:15.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18830bd4160844e3a2ae5235c6eed1f0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ec0caeecf276433fb5fee1c8c1c245bf : SUCCESS in 14m 02s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/de747a04c3374ca58395a5753f735863 : SUCCESS in 14m 06s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"ed345add2c28419999c2affde65a9783e2940c06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-12 09:34:02.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c4b1e98cd1b4094b293b1899fd09b95\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/729eb537e7314ea59f66eaed62fa2bbb : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93199240edb944a4a3b5e5e2bc6f4c85 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea3d1fa9736e4a23a0589df04187237d : SUCCESS in 6m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9ddeacaada3a47e08ff1e8e9cf62a6b5 : SUCCESS in 4m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4a8e1ca88af1460982f2974f13ddf3dd : SUCCESS in 8m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/273c8ab13bca4585b9668e544578657c : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/973b88f57da44320827889e389b15da1 : SUCCESS in 6m 01s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/ca1bb2d8d27042bc8ceaea2f89462f42 : SUCCESS in 45m 27s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/8469431566c4472ba066bd265cc3f523 : SUCCESS in 1h 38m 49s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/1abb7e62a5cf4b3f805417d1e78438ba : SUCCESS in 1h 24m 55s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/a5cf8bf83d7a46e1a1af5f15ee80b759 : SUCCESS in 42m 06s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/837ead2ea3f5485cb7ea3ee2086dec24 : SUCCESS in 1h 28m 35s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/4602fb3d98354695aaf70fc88265f563 : SUCCESS in 1h 18m 40s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/abea00b6452c478f833db8de3f1cf53a : SUCCESS in 47m 08s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/eae951e4016e47c9a223454d3938e022 : SUCCESS in 1h 30m 34s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/221a21d3f58348ed8432fe571c854928 : SUCCESS in 1h 13m 51s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c9ce5d8f29be41b5923f46a73f93d257 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":16},{"id":"2af52752c874e046f0279f0751fbbe987ea2f099","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-08-16 06:10:49.000000000","message":"Patch Set 17: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"feacc93117bacde0898f3f86e6ffc10171d912db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-08-16 06:16:41.000000000","message":"Patch Set 18: Commit message was updated.","accounts_in_message":[],"_revision_number":18},{"id":"f2f425e40cfb729201c92391b2f2341c3b834b90","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-08-16 06:19:47.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"5542e8a365cf82e1c003aabe52bb2807ec7667f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 06:30:35.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/230beef47ef5488eacc11c6f1d3b66cd\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/81846f77f333403aaaf5836e9c1e6a81 : SUCCESS in 8m 55s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/aa39ebe42f954197bf2db5ddb6bd1327 : SUCCESS in 12m 00s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"bb16450fc0c0d05fd1b79b8a67d0e3bb595b5dce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 07:55:27.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7101e09ebe564cc194e15f3561182f6c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d98f9bfdf3340abb2b75cfe55a0e24a : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2023e025da8d47c5aa2d5bb8a9c2c91a : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/317b46b2285849f19c71e7d3b895af18 : SUCCESS in 4m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0f70283162ab42a482a431edb949f40f : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/85223efe13a24ece842d3e0f9caa9dce : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b38229cb5b2409ea08c3a491b7d71ce : SUCCESS in 4m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/abd4f9cd80f74ac5b2af3aef09d1b177 : SUCCESS in 4m 25s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/7fb32680614a46deb0cd848442621602 : SUCCESS in 45m 29s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/b338444e10f041669d0aa4b082e8644e : SUCCESS in 1h 37m 32s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/e3fb522585cc43d8bd4987c5a3a4b748 : SUCCESS in 1h 19m 30s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/5a31bc1fb5fc408f9b025dda6dcd731b : SUCCESS in 42m 18s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/341d0e011f904325a2ea7306ecc25ae0 : SUCCESS in 1h 28m 59s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/f48afe3f9a274f6e8d96ebf80333bc0f : SUCCESS in 1h 13m 48s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/cf0216aacb4f4e25a890df163125b7f6 : SUCCESS in 46m 08s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/69f6b334d4594ba5bec904ba0791fa52 : SUCCESS in 1h 30m 00s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/3401877c6124474884b137088fb8a1fb : SUCCESS in 1h 08m 49s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8a2969665d50482e924f3abca24f17ea : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":18},{"id":"2c0b1e802fb204df18b9b81ec22152ab5b7fbc35","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-08-17 06:06:32.000000000","message":"Patch Set 18: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":18},{"id":"099848c6088a361f1d34571e0565f3a7fbeb28fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-17 06:07:02.000000000","message":"Patch Set 18: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":18},{"id":"443f3973facf0ed3132269e23c8fbe9260241583","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-17 06:48:24.000000000","message":"Patch Set 18: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49494236514049ffa61f02be6c12d26c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47eb547e67ca4a1dba88bca102358492 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/744b4ff3218240e2947841e2381b20c1 : SUCCESS in 4m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2dd54647c89b48cf8db1fc58f4f5addb : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70ed2d9968d545d5a51f46f8c172d86b : SUCCESS in 4m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc48026fbbc14a1dad2baa8adea53979 : SUCCESS in 4m 59s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/80a33a134e69477cab79f6c6b4fc00a3 : SUCCESS in 40m 49s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/81483e8c377d4a3686b04a56ea849cae : SUCCESS in 37m 49s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/1708650233b64d4e9c74ad671dd6b50f : SUCCESS in 40m 19s","accounts_in_message":[],"_revision_number":18},{"id":"f3c4d8712b98be1d6d911608720df172d41058df","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-17 06:48:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":18},{"id":"ceace4116ff1a8dcfe267941886966971e7db37f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-17 06:49:28.000000000","message":"Patch Set 18:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f359398b132c4bfab47acc4cbe31cee4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4651929f09464c00b714c54f59c5be2b : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8063e4337ad4ba5a14826268c46321b : SUCCESS in 43s","accounts_in_message":[],"_revision_number":18}],"current_revision_number":18,"current_revision":"27778a90d0f9094286f018b5e7b5a6fc712c47b4","revisions":{"fa231d028df5290390f7b279b9d3b4eeca8faf5f":{"kind":"REWORK","_number":1,"created":"2023-07-27 07:09:42.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/97/889797/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/97/889797/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/97/889797/1"}}},"commit":{"parents":[{"commit":"c4efd57e3659a2ac075106835ffd807b1f77df35","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c4efd57e3659a2ac075106835ffd807b1f77df35"}]}],"author":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-27 07:09:23.000000000","tz":420},"committer":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-27 07:09:23.000000000","tz":420},"subject":"Change uid/gid of folder managed by trove","message":"Change uid/gid of folder managed by trove\n\nWhen trove prepare container was grant permission for folders\nwith default user/group is \"database\" (id 1001).\nBut, some docker image use provision database required other user.\nSo, we need make flexible to make database container can run\n\nStory: #2010827\nTask: #48386\nChange-Id: Ie0268951db7a9e711089714cecf7fbf311eecec8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fa231d028df5290390f7b279b9d3b4eeca8faf5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fa231d028df5290390f7b279b9d3b4eeca8faf5f"}]},"branch":"refs/heads/master"},"2fb04b5a0aae85928bbf1347fc040f30b507dc0c":{"kind":"REWORK","_number":2,"created":"2023-07-27 07:35:41.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/97/889797/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/97/889797/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/97/889797/2"}}},"commit":{"parents":[{"commit":"c4efd57e3659a2ac075106835ffd807b1f77df35","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c4efd57e3659a2ac075106835ffd807b1f77df35"}]}],"author":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-27 07:09:23.000000000","tz":420},"committer":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-27 07:35:27.000000000","tz":420},"subject":"Change uid/gid of folder managed by trove","message":"Change uid/gid of folder managed by trove\n\nWhen trove prepare container was grant permission for folders\nwith default user/group is \"database\" (id 1001).\nBut, some docker image use provision database required other user.\nSo, we need make flexible to make database container can run\n\nStory: #2010827\nTask: #48386\nChange-Id: Ie0268951db7a9e711089714cecf7fbf311eecec8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2fb04b5a0aae85928bbf1347fc040f30b507dc0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2fb04b5a0aae85928bbf1347fc040f30b507dc0c"}]},"branch":"refs/heads/master"},"04fdf447fcf6a4d5c4d9d67220cfdcd0f943ee7c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-04-02 03:19:52.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/97/889797/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/97/889797/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/97/889797/3"}}},"commit":{"parents":[{"commit":"a2775ec2aa459a0275ac86180ce0f091ca8ad384","subject":"Merge \"Get rid of nose and Proboscis\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a2775ec2aa459a0275ac86180ce0f091ca8ad384"}]}],"author":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-27 07:09:23.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2024-04-02 03:19:52.000000000","tz":0},"subject":"Change uid/gid of folder managed by trove","message":"Change uid/gid of folder managed by trove\n\nWhen trove prepare container was grant permission for folders\nwith default user/group is \"database\" (id 1001).\nBut, some docker image use provision database required other user.\nSo, we need make flexible to make database container can run\n\nStory: #2010827\nTask: #48386\nChange-Id: Ie0268951db7a9e711089714cecf7fbf311eecec8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/04fdf447fcf6a4d5c4d9d67220cfdcd0f943ee7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/04fdf447fcf6a4d5c4d9d67220cfdcd0f943ee7c"}]},"branch":"refs/heads/master","description":"Rebase"},"26e1467ff336c1c2a6e3b832eecee781202c1d70":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-04-02 10:00:00.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/97/889797/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/97/889797/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/97/889797/4"}}},"commit":{"parents":[{"commit":"c8d1848c0f8afdf7c8e68f059d40c2c065d21f0f","subject":"Merge \"Bump hacking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c8d1848c0f8afdf7c8e68f059d40c2c065d21f0f"}]}],"author":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-27 07:09:23.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2024-04-02 10:00:00.000000000","tz":0},"subject":"Change uid/gid of folder managed by trove","message":"Change uid/gid of folder managed by trove\n\nWhen trove prepare container was grant permission for folders\nwith default user/group is \"database\" (id 1001).\nBut, some docker image use provision database required other user.\nSo, we need make flexible to make database container can run\n\nStory: #2010827\nTask: #48386\nChange-Id: Ie0268951db7a9e711089714cecf7fbf311eecec8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/26e1467ff336c1c2a6e3b832eecee781202c1d70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/26e1467ff336c1c2a6e3b832eecee781202c1d70"}]},"branch":"refs/heads/master","description":"Rebase"},"bf61db397b4e5a1aea0a39a99f010a8a0eb5ef01":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-05-13 01:58:55.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/97/889797/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/97/889797/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/97/889797/5"}}},"commit":{"parents":[{"commit":"3a69bd353b66f82dfbedd21e416616368cf4cb52","subject":"Remove extras and enum34","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3a69bd353b66f82dfbedd21e416616368cf4cb52"}]}],"author":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-27 07:09:23.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2024-05-13 01:58:55.000000000","tz":0},"subject":"Change uid/gid of folder managed by trove","message":"Change uid/gid of folder managed by trove\n\nWhen trove prepare container was grant permission for folders\nwith default user/group is \"database\" (id 1001).\nBut, some docker image use provision database required other user.\nSo, we need make flexible to make database container can run\n\nStory: #2010827\nTask: #48386\nChange-Id: Ie0268951db7a9e711089714cecf7fbf311eecec8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bf61db397b4e5a1aea0a39a99f010a8a0eb5ef01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bf61db397b4e5a1aea0a39a99f010a8a0eb5ef01"}]},"branch":"refs/heads/master","description":"Rebase"},"6ee9feb3ae0345b49232dc4b7b21aaaf18bf852e":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-05-21 07:34:55.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/97/889797/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/97/889797/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/97/889797/6"}}},"commit":{"parents":[{"commit":"7501804a9f321525e7f5e68f4c51ff5cbd45be70","subject":"Merge \"reno: Update master for unmaintained/zed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7501804a9f321525e7f5e68f4c51ff5cbd45be70"}]}],"author":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-27 07:09:23.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2024-05-21 07:34:55.000000000","tz":0},"subject":"Change uid/gid of folder managed by trove","message":"Change uid/gid of folder managed by trove\n\nWhen trove prepare container was grant permission for folders\nwith default user/group is \"database\" (id 1001).\nBut, some docker image use provision database required other user.\nSo, we need make flexible to make database container can run\n\nStory: #2010827\nTask: #48386\nChange-Id: Ie0268951db7a9e711089714cecf7fbf311eecec8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6ee9feb3ae0345b49232dc4b7b21aaaf18bf852e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6ee9feb3ae0345b49232dc4b7b21aaaf18bf852e"}]},"branch":"refs/heads/master","description":"Rebase"},"ec36d6cc972b0c977d2ec65e2ec782867894059a":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-07-11 02:29:40.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/97/889797/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/97/889797/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/97/889797/7"}}},"commit":{"parents":[{"commit":"be9bd3f3a034d9e31b32fa194db8e42ed0ee4155","subject":"Use policy to determine admin context","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/be9bd3f3a034d9e31b32fa194db8e42ed0ee4155"}]}],"author":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-27 07:09:23.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2024-07-11 02:29:40.000000000","tz":0},"subject":"Change uid/gid of folder managed by trove","message":"Change uid/gid of folder managed by trove\n\nWhen trove prepare container was grant permission for folders\nwith default user/group is \"database\" (id 1001).\nBut, some docker image use provision database required other user.\nSo, we need make flexible to make database container can run\n\nStory: #2010827\nTask: #48386\nChange-Id: Ie0268951db7a9e711089714cecf7fbf311eecec8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ec36d6cc972b0c977d2ec65e2ec782867894059a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ec36d6cc972b0c977d2ec65e2ec782867894059a"}]},"branch":"refs/heads/master","description":"Rebase"},"282d49c9e77acedc2400a8cf11f58ebd04b03b67":{"kind":"REWORK","_number":8,"created":"2024-07-17 04:23:11.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/97/889797/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/97/889797/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/97/889797/8"}}},"commit":{"parents":[{"commit":"be9bd3f3a034d9e31b32fa194db8e42ed0ee4155","subject":"Use policy to determine admin context","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/be9bd3f3a034d9e31b32fa194db8e42ed0ee4155"}]}],"author":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-27 07:09:23.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2024-07-17 04:23:00.000000000","tz":420},"subject":"Change uid/gid of folder managed by trove","message":"Change uid/gid of folder managed by trove\n\nWhen trove prepare container was grant permission for folders\nwith default user/group is \"database\" (id 1001).\nBut, some docker image use provision database required other user.\nSo, we need make flexible to make database container can run\n\nStory: #2010827\nTask: #48386\nChange-Id: Ie0268951db7a9e711089714cecf7fbf311eecec8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/282d49c9e77acedc2400a8cf11f58ebd04b03b67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/282d49c9e77acedc2400a8cf11f58ebd04b03b67"}]},"branch":"refs/heads/master"},"ec5732e991fec8aee0925d3337d598a293669a65":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2024-07-25 09:57:07.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/97/889797/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/97/889797/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/97/889797/9"}}},"commit":{"parents":[{"commit":"3f521cb7f577635ee74b2a48c70f874dab2f9e30","subject":"[CI] fix trove gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3f521cb7f577635ee74b2a48c70f874dab2f9e30"}]}],"author":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-27 07:09:23.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2024-07-25 09:57:07.000000000","tz":0},"subject":"Change uid/gid of folder managed by trove","message":"Change uid/gid of folder managed by trove\n\nWhen trove prepare container was grant permission for folders\nwith default user/group is \"database\" (id 1001).\nBut, some docker image use provision database required other user.\nSo, we need make flexible to make database container can run\n\nStory: #2010827\nTask: #48386\nChange-Id: Ie0268951db7a9e711089714cecf7fbf311eecec8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ec5732e991fec8aee0925d3337d598a293669a65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ec5732e991fec8aee0925d3337d598a293669a65"}]},"branch":"refs/heads/master","description":"Rebase"},"768be826d1618f1197e78da033b982baeb292c90":{"kind":"REWORK","_number":10,"created":"2024-08-09 04:59:33.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/97/889797/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/97/889797/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/97/889797/10"}}},"commit":{"parents":[{"commit":"be9bd3f3a034d9e31b32fa194db8e42ed0ee4155","subject":"Use policy to determine admin context","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/be9bd3f3a034d9e31b32fa194db8e42ed0ee4155"}]}],"author":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-27 07:09:23.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2024-08-09 04:59:09.000000000","tz":420},"subject":"Change uid/gid of folder managed by trove","message":"Change uid/gid of folder managed by trove\n\nWhen trove prepare container was grant permission for folders\nwith default user/group is \"database\" (id 1001).\nBut, some docker image use provision database required other user.\nSo, we need make flexible to make database container can run\n\nStory: #2010827\nTask: #48386\nChange-Id: Ie0268951db7a9e711089714cecf7fbf311eecec8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/768be826d1618f1197e78da033b982baeb292c90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/768be826d1618f1197e78da033b982baeb292c90"}]},"branch":"refs/heads/master"},"aee863fda2f906fb4c5c13c0579fb81558d728a9":{"kind":"REWORK","_number":11,"created":"2024-08-09 07:17:34.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/97/889797/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/97/889797/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/97/889797/11"}}},"commit":{"parents":[{"commit":"be9bd3f3a034d9e31b32fa194db8e42ed0ee4155","subject":"Use policy to determine admin context","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/be9bd3f3a034d9e31b32fa194db8e42ed0ee4155"}]}],"author":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-27 07:09:23.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2024-08-09 07:17:22.000000000","tz":420},"subject":"Change uid/gid of folder managed by trove","message":"Change uid/gid of folder managed by trove\n\nWhen trove prepare container was grant permission for folders\nwith default user/group is \"database\" (id 1001).\nBut, some docker image use provision database required other user.\nSo, we need make flexible to make database container can run\n\nStory: #2010827\nTask: #48386\nChange-Id: Ie0268951db7a9e711089714cecf7fbf311eecec8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/aee863fda2f906fb4c5c13c0579fb81558d728a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/aee863fda2f906fb4c5c13c0579fb81558d728a9"}]},"branch":"refs/heads/master"},"aee9f25e1e694562a6c22e50cc63a4db160f7ab4":{"kind":"REWORK","_number":12,"created":"2024-08-09 10:24:16.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/97/889797/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/97/889797/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/97/889797/12"}}},"commit":{"parents":[{"commit":"be9bd3f3a034d9e31b32fa194db8e42ed0ee4155","subject":"Use policy to determine admin context","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/be9bd3f3a034d9e31b32fa194db8e42ed0ee4155"}]}],"author":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-27 07:09:23.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2024-08-09 10:24:01.000000000","tz":420},"subject":"Change uid/gid of folder managed by trove","message":"Change uid/gid of folder managed by trove\n\nWhen trove prepare container was grant permission for folders\nwith default user/group is \"database\" (id 1001).\nBut, some docker image use provision database required other user.\nSo, we need make flexible to make database container can run\n\nStory: #2010827\nTask: #48386\nChange-Id: Ie0268951db7a9e711089714cecf7fbf311eecec8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/aee9f25e1e694562a6c22e50cc63a4db160f7ab4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/aee9f25e1e694562a6c22e50cc63a4db160f7ab4"}]},"branch":"refs/heads/master"},"4a76f7600afb76c6009e7c08db03da6b1d2ad713":{"kind":"REWORK","_number":13,"created":"2024-08-09 10:25:24.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/97/889797/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/97/889797/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/97/889797/13"}}},"commit":{"parents":[{"commit":"be9bd3f3a034d9e31b32fa194db8e42ed0ee4155","subject":"Use policy to determine admin context","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/be9bd3f3a034d9e31b32fa194db8e42ed0ee4155"}]}],"author":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-27 07:09:23.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2024-08-09 10:25:11.000000000","tz":420},"subject":"Change uid/gid of folder managed by trove","message":"Change uid/gid of folder managed by trove\n\nWhen trove prepare container was grant permission for folders\nwith default user/group is \"database\" (id 1001).\nBut, some docker image use provision database required other user.\nSo, we need make flexible to make database container can run\n\nStory: #2010827\nTask: #48386\nChange-Id: Ie0268951db7a9e711089714cecf7fbf311eecec8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4a76f7600afb76c6009e7c08db03da6b1d2ad713"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4a76f7600afb76c6009e7c08db03da6b1d2ad713"}]},"branch":"refs/heads/master"},"8a8a8738b13cb607d484a311b042c30d7f329086":{"kind":"REWORK","_number":14,"created":"2024-08-09 10:34:13.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/97/889797/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/97/889797/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/97/889797/14"}}},"commit":{"parents":[{"commit":"be9bd3f3a034d9e31b32fa194db8e42ed0ee4155","subject":"Use policy to determine admin context","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/be9bd3f3a034d9e31b32fa194db8e42ed0ee4155"}]}],"author":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-27 07:09:23.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2024-08-09 10:34:00.000000000","tz":420},"subject":"Change uid/gid of folder managed by trove","message":"Change uid/gid of folder managed by trove\n\nWhen trove prepare container was grant permission for folders\nwith default user/group is \"database\" (id 1001).\nBut, some docker image use provision database required other user.\nSo, we need make flexible to make database container can run\n\nStory: #2010827\nTask: #48386\nChange-Id: Ie0268951db7a9e711089714cecf7fbf311eecec8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8a8a8738b13cb607d484a311b042c30d7f329086"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8a8a8738b13cb607d484a311b042c30d7f329086"}]},"branch":"refs/heads/master"},"b058d16cbd9d93a95b10a6a1f2a8b884b0816903":{"kind":"REWORK","_number":15,"created":"2024-08-12 03:11:05.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/97/889797/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/97/889797/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/97/889797/15"}}},"commit":{"parents":[{"commit":"be9bd3f3a034d9e31b32fa194db8e42ed0ee4155","subject":"Use policy to determine admin context","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/be9bd3f3a034d9e31b32fa194db8e42ed0ee4155"}]}],"author":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-27 07:09:23.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2024-08-12 03:10:52.000000000","tz":420},"subject":"Change uid/gid of folder managed by trove","message":"Change uid/gid of folder managed by trove\n\nWhen trove prepare container was grant permission for folders\nwith default user/group is \"database\" (id 1001).\nBut, some docker image use provision database required other user.\nSo, we need make flexible to make database container can run\n\nStory: #2010827\nTask: #48386\nChange-Id: Ie0268951db7a9e711089714cecf7fbf311eecec8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b058d16cbd9d93a95b10a6a1f2a8b884b0816903"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b058d16cbd9d93a95b10a6a1f2a8b884b0816903"}]},"branch":"refs/heads/master"},"5c06417d786bf4e8dfb8fe74b460d7e2c1ef49ac":{"kind":"REWORK","_number":16,"created":"2024-08-12 07:54:06.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/97/889797/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/97/889797/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/97/889797/16"}}},"commit":{"parents":[{"commit":"be9bd3f3a034d9e31b32fa194db8e42ed0ee4155","subject":"Use policy to determine admin context","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/be9bd3f3a034d9e31b32fa194db8e42ed0ee4155"}]}],"author":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-27 07:09:23.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2024-08-12 07:53:27.000000000","tz":420},"subject":"Change uid/gid of folder managed by trove","message":"Change uid/gid of folder managed by trove\n\nWhen trove prepare container was grant permission for folders\nwith default user/group is \"database\" (id 1001).\nBut, some docker image use provision database required other user.\nSo, we need make flexible to make database container can run\n\nStory: #2010827\nTask: #48386\nChange-Id: Ie0268951db7a9e711089714cecf7fbf311eecec8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5c06417d786bf4e8dfb8fe74b460d7e2c1ef49ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5c06417d786bf4e8dfb8fe74b460d7e2c1ef49ac"}]},"branch":"refs/heads/master"},"f039719f2a316f7fdd36382df8cc4d6a879944c4":{"kind":"NO_CODE_CHANGE","_number":17,"created":"2024-08-16 06:10:49.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/97/889797/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/97/889797/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/97/889797/17"}}},"commit":{"parents":[{"commit":"be9bd3f3a034d9e31b32fa194db8e42ed0ee4155","subject":"Use policy to determine admin context","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/be9bd3f3a034d9e31b32fa194db8e42ed0ee4155"}]}],"author":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-27 07:09:23.000000000","tz":420},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2024-08-16 06:10:49.000000000","tz":0},"subject":"ADD UID/GID configurations for each datastore","message":"ADD UID/GID configurations for each datastore\n\nWhen trove prepares a container, it grants permissions for folders with the\ndefault user/group \"database\" (ID 1001). However, some Docker images require\na different user for provisioning the database. Therefore, we need to make it\nflexible so that the database container can run with the required user and group,\nregardless of the default settings.\n\nStory: #2010827\nTask: #48386\nChange-Id: Ie0268951db7a9e711089714cecf7fbf311eecec8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f039719f2a316f7fdd36382df8cc4d6a879944c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f039719f2a316f7fdd36382df8cc4d6a879944c4"}]},"branch":"refs/heads/master","description":"Edit commit message"},"27778a90d0f9094286f018b5e7b5a6fc712c47b4":{"kind":"NO_CODE_CHANGE","_number":18,"created":"2024-08-16 06:16:41.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/97/889797/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/97/889797/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/97/889797/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/97/889797/18"}}},"commit":{"parents":[{"commit":"be9bd3f3a034d9e31b32fa194db8e42ed0ee4155","subject":"Use policy to determine admin context","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/be9bd3f3a034d9e31b32fa194db8e42ed0ee4155"}]}],"author":{"name":"Bo Tran Van","email":"ministry.96.nd@gmail.com","date":"2023-07-27 07:09:23.000000000","tz":420},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2024-08-16 06:16:41.000000000","tz":0},"subject":"Add UID/GID configurations for each datastore","message":"Add UID/GID configurations for each datastore\n\nWhen trove prepares a container, it grants permissions for folders with\nthe default user/group \"database\" (ID 1001). However, some Docker images\nrequire a different user for provisioning the database. Therefore, we\nneed to make it flexible so that the database container can run with\nthe required user and group, regardless of the default settings.\n\nStory: #2010827\nTask: #48386\nChange-Id: Ie0268951db7a9e711089714cecf7fbf311eecec8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/27778a90d0f9094286f018b5e7b5a6fc712c47b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/27778a90d0f9094286f018b5e7b5a6fc712c47b4"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
