)]}'
{"id":"openstack%2Ftrove~891657","triplet_id":"openstack%2Ftrove~master~I7af019fc3c4ce295e4e4b4ccafae5a170a7b7833","project":"openstack/trove","branch":"master","topic":"fix-ci","attention_set":{},"removed_from_attention_set":{"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2023-08-18 07:17:15.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"I7af019fc3c4ce295e4e4b4ccafae5a170a7b7833","subject":"[CI]: fix trove api started failed","status":"MERGED","created":"2023-08-17 07:13:30.000000000","updated":"2023-08-18 10:27:14.000000000","submitted":"2023-08-18 10:26:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"891657-fix-ci","meta_rev_id":"89901012442cf1f4bade863cf1f69ab7788f3ced","_number":891657,"virtual_id_number":891657,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-18 10:26:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-18 07:17:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-18 07:17:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-17 07:20:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-17 07:54:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7a8ad5e8fd3d5dbb0b46577994e022b709a4b7b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-08-17 07:13:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b38591e99d9b37ecc339f18797025ef83cf9b0e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 07:20:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5053a6efedb0446b9b9f8e35a46ca649\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bc1c36cdf9a04b1b98d16b8c68fa837d : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f6f41182a1c5426c866bdabdefaaed66 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c721fbe822d74d68a6c20814e85bcc21 : SUCCESS in 5m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a219a323cda0fe5f54f881fcc5a073a8982176e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 07:54:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd5fcb2d251e4a4492633c3764903dd1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94e621447f204bd6b0ac6bcbe12c0916 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55a00604f9eb49089f80bf7c4d2f8716 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/887c87c08dc44456aeb5b1fd7c25b408 : SUCCESS in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/767bfdb062764151829dd623d3ba6748 : SUCCESS in 5m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6bba0a2c615d44b3b5c77d572b999d6a : SUCCESS in 6m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a002700d78d14b51b59e3c2e2662a339 : SUCCESS in 6m 33s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b4a9d08c06024682816392cb070735e9 : SUCCESS in 5m 59s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/3b955da33c5645afaf733b85b094b368 : FAILURE in 38m 49s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/a3817a6995ed47e1a2db043304aff4b7 : FAILURE in 39m 12s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/7f525fbd82894bc2b104a32f83cf5aaf : FAILURE in 38m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9b19a7cb555dd9af556ff6deb6209f65c27c73d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-08-17 10:14:17.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"0908bfaf3c70dfe03865c9db2917fdeec0581392","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-08-17 10:20:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"418db966d647e279912b6939f32f8cb5546911f2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 10:32:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a9c39294f3a4f0c9ef8c7c5904c9b40\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1505c71d82514e3bb0098ef448d28832 : SUCCESS in 10m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f385f0be6c404c45abf451954a93cd11 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/899f6759310e458bb8aeb997868a4442 : SUCCESS in 7m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1db293e35864abfe86c03b0da857360f595ae55f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 10:39:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5f3d17458d0d4a4a96c5dfe77a9a729d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bbe6dc6d385c4e1dae1448c12f914414 : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89a75d2f6edd436db21eb27ff31a64d7 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f67322ef37004f79bfa5dcef6985c126 : SUCCESS in 6m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa7129a53eab45ce984ba52481d488a2 : SUCCESS in 5m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8a6389c9887b45c6bbbfd89780d6e275 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5b7b22f13574587ab20acdcf9dcf436 : SUCCESS in 5m 43s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/acf8a25aa27a40d3b4ac00aed22df062 : SUCCESS in 6m 52s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/22213a4bf0424b659717a14f129d2bbe : FAILURE in 16m 59s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/3d42767e53b74a008b01bdcb920d6e8c : FAILURE in 16m 35s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/160a3dbf07a94037a7a4e6b402fc1dcc : FAILURE in 16m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"210c8ebec6c503fa2e855bc88ae7c0e9ccec9eee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-08-17 12:07:09.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"f70fc04658b8c172a082211cdf44bd44282f8145","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 12:15:32.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c570e23b7954de794a9ee3f51a7d6d7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/27a612b9fb7544b194c1012d63021fb6 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/251f5650c39642ccb4b75069e34c83f0 : SUCCESS in 7m 28s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ec4502546fd846efb10557724051267d : SUCCESS in 5m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"349cbc856b78bd685330a857764d97e66d2b0c88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 14:43:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de249ba479db4281947fe87025915e28\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4cb20b4b12440b0ae58072dbf35f53f : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a763a1d137643349a621ad2934dd85c : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0aa6c91f59d7455794802da739aa3384 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/692eb0c4a33544eab71bbac67d035290 : SUCCESS in 6m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fcfa305851354a118ac9d33591777eb3 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81e55b9b74f94672ade0288697b22c8b : SUCCESS in 6m 04s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8dfde11115e0473a9a16b898cf4a4573 : SUCCESS in 5m 37s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/a3458c38357d4dddab4f9549e0213fb1 : SUCCESS in 2h 34m 33s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/d7111c2f8ec74024acf51afe24cc1bee : SUCCESS in 1h 43m 46s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/36793fecebe44dcaa55bc6d9b982d20a : FAILURE in 2h 26m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4e0bcacef76f6c753a7103b1b2ec482eb50c8808","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-08-18 02:52:35.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"876470c72a94585a62fc61de94ec7397fda6fe13","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 03:04:10.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15853330cd724fe5accdb1f5db2f2a72\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fae7e6cadeca4b0cb7ffbba4c8e153ce : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/185619772f924565affa95d7867f5fee : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/be260e62dfeb43b8ba6139f1863bf0fa : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e5b5897c6779472cb625c589bfe5c701 : RETRY_LIMIT in 2m 34s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"acf85b58483d43bc67358278440a6544b8413f9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 05:10:59.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/908e7c1ed9cc43eca5174f987b84ecd8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3c64e133f5a4afbb025e4360646102b : SUCCESS in 10m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a87b7924c82d4272b37adca0bc144c93 : SUCCESS in 6m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8533ca881d46407e83883c6438a5099c : SUCCESS in 8m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0a1715f578a24bf99e5a7944b7230f1c : SUCCESS in 7m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/599f08beefcb453a8f525bf433514a15 : SUCCESS in 5m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/575756ca09d246c3b5ad386406a7ea88 : RETRY_LIMIT in 3m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38efea18b98241e18bab1aa3d5c758cc : SUCCESS in 5m 50s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/de138607684d4fb6aaec11c1e85028c3 : SUCCESS in 5m 19s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/637f965e829f45c3aeba33166e8aed19 : SUCCESS in 2h 16m 36s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/5080fd15571441cea634dd21dbc7b575 : SUCCESS in 1h 32m 47s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/e90dea2936bf4d9ca85957552289d462 : SUCCESS in 2h 17m 05s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7a0ea6e1fb517efc282632ba2b296fae3eff2561","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-08-18 07:17:15.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"8a660a6d52a4b2b27f909327517b77ed3a71d04f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 07:17:43.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"e8a37ae39e8e6a4a6b3ae9521291659b1dccdb81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 10:26:11.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/107b4d4498ea44908bd97dae9d046559\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b6aa568ac344b0f842ffd9b81810d9c : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67a70325073b4da4b65f037aefe1181d : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f7134f7332e4d76b0038e61aecd0976 : SUCCESS in 5m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/80a8d11938014d6d9be45642d3f725ef : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c459a983ad14d5db68d648e7d377cac : SUCCESS in 4m 49s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/173502eb1f10485d9b2bb53a8de8869d : TIMED_OUT in 3h 03m 34s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"cc628df71c44fd7a07f822c033e61f2ff193269e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 10:26:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"89901012442cf1f4bade863cf1f69ab7788f3ced","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 10:27:14.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/342305e2648b4b50a1ce09eef331cce7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06c7e3edc33c4013afe0b84459f69f79 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d29efdf1a6a0cca53cd34114b9eb3e8f7bc39a6e","revisions":{"48be8c6031f84ecafb24039c21078a955fdd9833":{"kind":"REWORK","_number":1,"created":"2023-08-17 07:13:30.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/57/891657/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/57/891657/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/891657/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/891657/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/891657/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/57/891657/1"}}},"commit":{"parents":[{"commit":"b841781fd5a813249959cd0e86a8b170a8792fd7","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b841781fd5a813249959cd0e86a8b170a8792fd7"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-08-17 07:12:29.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-08-17 07:12:29.000000000","tz":480},"subject":"[CI]: fix trove api started failed","message":"[CI]: fix trove api started failed\n\nthis may caused by: https://review.opendev.org/c/openstack/devstack/+/558930\n\nChange-Id: I7af019fc3c4ce295e4e4b4ccafae5a170a7b7833\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/48be8c6031f84ecafb24039c21078a955fdd9833"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/48be8c6031f84ecafb24039c21078a955fdd9833"}]},"branch":"refs/heads/master"},"e11be991bc19e0a1d28ac463ddb3e42986e3cc26":{"kind":"REWORK","_number":2,"created":"2023-08-17 10:14:17.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/57/891657/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/57/891657/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/891657/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/891657/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/891657/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/57/891657/2"}}},"commit":{"parents":[{"commit":"b841781fd5a813249959cd0e86a8b170a8792fd7","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b841781fd5a813249959cd0e86a8b170a8792fd7"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-08-17 07:12:29.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-08-17 10:14:02.000000000","tz":480},"subject":"[CI]: fix trove api started failed","message":"[CI]: fix trove api started failed\n\nthis may caused by: https://review.opendev.org/c/openstack/devstack/+/558930\n\nChange-Id: I7af019fc3c4ce295e4e4b4ccafae5a170a7b7833\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e11be991bc19e0a1d28ac463ddb3e42986e3cc26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e11be991bc19e0a1d28ac463ddb3e42986e3cc26"}]},"branch":"refs/heads/master"},"e731c37a511cf326c5ca6e477a4d0114ffa78b68":{"kind":"REWORK","_number":3,"created":"2023-08-17 10:20:05.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/57/891657/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/57/891657/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/891657/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/891657/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/891657/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/57/891657/3"}}},"commit":{"parents":[{"commit":"b841781fd5a813249959cd0e86a8b170a8792fd7","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b841781fd5a813249959cd0e86a8b170a8792fd7"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-08-17 07:12:29.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-08-17 10:19:37.000000000","tz":480},"subject":"[CI]: fix trove api started failed","message":"[CI]: fix trove api started failed\n\nthis may caused by: https://review.opendev.org/c/openstack/devstack/+/558930\n\nChange-Id: I7af019fc3c4ce295e4e4b4ccafae5a170a7b7833\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e731c37a511cf326c5ca6e477a4d0114ffa78b68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e731c37a511cf326c5ca6e477a4d0114ffa78b68"}]},"branch":"refs/heads/master"},"9ff7613b846e1c6bf6e4427755393dc860b4d4c4":{"kind":"REWORK","_number":4,"created":"2023-08-17 12:07:09.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/57/891657/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/57/891657/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/891657/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/891657/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/891657/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/57/891657/4"}}},"commit":{"parents":[{"commit":"b841781fd5a813249959cd0e86a8b170a8792fd7","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b841781fd5a813249959cd0e86a8b170a8792fd7"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-08-17 07:12:29.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-08-17 12:06:53.000000000","tz":480},"subject":"[CI]: fix trove api started failed","message":"[CI]: fix trove api started failed\n\nthis may caused by: https://review.opendev.org/c/openstack/devstack/+/558930\n\nChange-Id: I7af019fc3c4ce295e4e4b4ccafae5a170a7b7833\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9ff7613b846e1c6bf6e4427755393dc860b4d4c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9ff7613b846e1c6bf6e4427755393dc860b4d4c4"}]},"branch":"refs/heads/master"},"d29efdf1a6a0cca53cd34114b9eb3e8f7bc39a6e":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2023-08-18 02:52:35.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/57/891657/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/57/891657/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/891657/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/891657/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/891657/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/57/891657/5"}}},"commit":{"parents":[{"commit":"b841781fd5a813249959cd0e86a8b170a8792fd7","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b841781fd5a813249959cd0e86a8b170a8792fd7"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-08-17 07:12:29.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-08-18 02:52:33.000000000","tz":0},"subject":"[CI]: fix trove api started failed","message":"[CI]: fix trove api started failed\n\nthis may be caused by: https://review.opendev.org/c/openstack/devstack/+/558930\n\nChange-Id: I7af019fc3c4ce295e4e4b4ccafae5a170a7b7833\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d29efdf1a6a0cca53cd34114b9eb3e8f7bc39a6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d29efdf1a6a0cca53cd34114b9eb3e8f7bc39a6e"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
