)]}'
{"id":"openstack%2Ftrove~893006","triplet_id":"openstack%2Ftrove~master~I17e42b4e8ac60e75d4e2fb22618fde055ff3e358","project":"openstack/trove","branch":"master","topic":"typo","hashtags":[],"change_id":"I17e42b4e8ac60e75d4e2fb22618fde055ff3e358","subject":"Fix typo","status":"MERGED","created":"2023-08-29 03:21:47.000000000","updated":"2023-08-29 12:12:51.000000000","submitted":"2023-08-29 12:11:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893006-typo","meta_rev_id":"d8d75995291244870e246ca0a967674910659d30","_number":893006,"virtual_id_number":893006,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-29 12:11:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-29 08:56:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-29 08:56:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-29 03:41:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-29 05:39:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"21bc184af99532789acb9ca5a8bf0106d7914633","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-08-29 03:21:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e8d2c3f8e8afff43ad3e443a84274d42da0c44a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 03:41:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66baf545617043a187b6390a5ab8cae4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/795a230b1cd246ff8091568475e0b45e : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d2c31db0ad334c588e6e8f6a869e53f2 : SUCCESS in 17m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0e5b981652bb4267b2f980b2cc2f5bbf : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/eb7734319e39461b8260dfe6f64b15bd : RETRY_LIMIT in 1m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"05ef6206eee41b2a4a7011bf299cb91dd9d9fce5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 05:39:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0318850d4ba54e3b8a0fb39164bd5653\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d987732cf3945658f5bd5bb6ee14463 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a60ac4fca3e5461e8855683568d60c26 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5aa536e13d5c41b69557c819968a9e10 : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63f80ff0879f487aaa0f10397e1c6b64 : SUCCESS in 5m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/88db99f23919462c980f9d29aeaa3280 : SUCCESS in 5m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/938f3417e67b48f7a489ce18d6348b9e : RETRY_LIMIT in 1m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad3c1f471abb411696c67361ba89a085 : SUCCESS in 5m 33s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/19059c85362a42abbb4c8c77d2545c9d : SUCCESS in 5m 23s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/cec750200034478c90acccc2cefad672 : SUCCESS in 2h 16m 09s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/bb79470a9cfa4526a8f22ee893a6a045 : SUCCESS in 1h 30m 57s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/351ec0cfdfa3405f992395cd13888132 : SUCCESS in 2h 10m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"14563c1fdff229551ddfe166cbf3319871005f37","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-08-29 08:56:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4519aae13db3db114e4b4bf59d0f219bc044adb4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 08:56:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9317701a258df1375e0a8fb611935c37044d09bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 12:11:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60a06cec05e64d8eb3a164e3d655ca60\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c61ad2746f14c64aacf6bf62dbc96ab : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08206e52363b431aaf6220352f08d953 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46dd626bba5049a7ad74e2ffc34017c9 : SUCCESS in 5m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf84604a98d449d39f507d2c2f02a6ff : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/524bf9dd9a744f5c91e51d588c922bf5 : SUCCESS in 5m 09s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/3ecc96daedcb4036ae98cc11e5fd1893 : TIMED_OUT in 3h 02m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d17e78838cc469f140de9ba2b5bf214096941b32","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 12:11:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d8d75995291244870e246ca0a967674910659d30","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 12:12:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92a49e1a713043d8aa3f0262f6dada60\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8041956659004ae09cc6b0d51d442eec : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ccaa00d45285f33132ece4fd8881221c36242cc6","revisions":{"ccaa00d45285f33132ece4fd8881221c36242cc6":{"kind":"REWORK","_number":1,"created":"2023-08-29 03:21:47.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/06/893006/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/06/893006/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/06/893006/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/06/893006/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/06/893006/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/06/893006/1"}}},"commit":{"parents":[{"commit":"4288b8f78250cc3a1c93b019e2c3b4bf78df177c","subject":"Merge \"[CI]: fix trove api started failed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4288b8f78250cc3a1c93b019e2c3b4bf78df177c"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-08-29 03:14:43.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-08-29 03:21:30.000000000","tz":480},"subject":"Fix typo","message":"Fix typo\n\ntrivial fix\n\nChange-Id: I17e42b4e8ac60e75d4e2fb22618fde055ff3e358\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ccaa00d45285f33132ece4fd8881221c36242cc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ccaa00d45285f33132ece4fd8881221c36242cc6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
