)]}'
{"id":"openstack%2Ftrove~893866","triplet_id":"openstack%2Ftrove~master~Iab9d4849d50c01ff35ce159a988fde95593f7ebf","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2023-09-07 01:18:24.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"Iab9d4849d50c01ff35ce159a988fde95593f7ebf","subject":"Fix Zuul warnings with regexps","status":"MERGED","created":"2023-09-06 11:56:36.000000000","updated":"2023-09-07 04:12:46.000000000","submitted":"2023-09-07 04:09:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893866","meta_rev_id":"5206940cceaa10ca980dc34e57b5ac194cbeabee","_number":893866,"virtual_id_number":893866,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-07 04:09:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-07 01:18:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-07 01:18:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-06 11:56:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9a4b51b3a98ed0771373c5d1ebbb3e4858f8f0f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-09-06 11:56:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d195feaabb2ae67c532f240e1ee131bc58ae214a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 11:56:59.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/trove on branch master.  The error was:\n\n  extra keys not allowed @ data[\u0027periodic\u0027][\u0027jobs\u0027][0][\u0027publish-trove-guest-image-ubuntu-jammy\u0027][\u0027negate\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      queue: trove\n      templates:\n        - check-requirements\n        - openstack-cover-jobs\n        - openstack-python3-jobs\n        - openstack-python3-jobs-arm64\n        - periodic-stable-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - openstack-tox-cover:\n              voting: true\n          - trove-tox-bandit-baseline:\n              voting: true\n          - trove-tempest:\n              voting: false\n          - trove-tempest-mysql-network-isolation:\n              voting: true\n          - trove-ubuntu-guest-image-build:\n              voting: true\n          - trove-centos8s-guest-image-build:\n              voting: true\n          - trove-tempest-postgres:\n              voting: false\n      gate:\n        jobs:\n          - trove-tempest:\n              voting: false\n      experimental:\n        jobs:\n          - trove-tempest-ipv6-only\n          - trove-functional-mysql\n          - trove-scenario-mysql-single\n          - trove-scenario-mysql-multi\n          - trove-scenario-mariadb-single\n          - trove-scenario-mariadb-multi\n      periodic:\n        jobs:\n          - publish-trove-guest-image-ubuntu-jammy:\n              branches: ^stable/.*$\n              negate: true\n          - publish-trove-datastore-images:\n              branches: ^stable/.*$\n              negate: true\n\n  in \"openstack/trove/zuul.d/projects.yaml@master\", line 2, column 3","accounts_in_message":[],"_revision_number":1},{"id":"71b6dc41547eb44a437aa5cfd0bd3716c3aa63c6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 11:57:04.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/trove on branch master.  The error was:\n\n  extra keys not allowed @ data[\u0027periodic\u0027][\u0027jobs\u0027][0][\u0027publish-trove-guest-image-ubuntu-jammy\u0027][\u0027negate\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      queue: trove\n      templates:\n        - check-requirements\n        - openstack-cover-jobs\n        - openstack-python3-jobs\n        - openstack-python3-jobs-arm64\n        - periodic-stable-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - openstack-tox-cover:\n              voting: true\n          - trove-tox-bandit-baseline:\n              voting: true\n          - trove-tempest:\n              voting: false\n          - trove-tempest-mysql-network-isolation:\n              voting: true\n          - trove-ubuntu-guest-image-build:\n              voting: true\n          - trove-centos8s-guest-image-build:\n              voting: true\n          - trove-tempest-postgres:\n              voting: false\n      gate:\n        jobs:\n          - trove-tempest:\n              voting: false\n      experimental:\n        jobs:\n          - trove-tempest-ipv6-only\n          - trove-functional-mysql\n          - trove-scenario-mysql-single\n          - trove-scenario-mysql-multi\n          - trove-scenario-mariadb-single\n          - trove-scenario-mariadb-multi\n      periodic:\n        jobs:\n          - publish-trove-guest-image-ubuntu-jammy:\n              branches: ^stable/.*$\n              negate: true\n          - publish-trove-datastore-images:\n              branches: ^stable/.*$\n              negate: true\n\n  in \"openstack/trove/zuul.d/projects.yaml@master\", line 2, column 3","accounts_in_message":[],"_revision_number":1},{"id":"8d7c31f34c93ed9e39e56a0ce8b7970b6f2f1137","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-09-06 12:05:05.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ea801dd53b6ab4f1de0b722b1dc58f738da89680","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 12:15:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/767a88190e2848af9befdd23a2258d62\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2d6cd3ae06af4400925194fccf1ebae8 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dbba539118d8410f877530678e5f0093 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5d26aacaf19b478698e04ecaa6fe7b6a : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ac28c4e87f824093b6d470553b3664bf : RETRY_LIMIT in 1m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b9fd5ada4dab2a0e05f64ffbd52749bb40fe903e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 12:18:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a98efd277c2a45bca588b92356801041\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1717e36fe12a4b239625d462fc661a34 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30fa052d44214e93bc86adc1c7a0f766 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5781a42a6a7d42129d14aaa41e17e8bf : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/749920ce5189401790be174a0d4c407a : SUCCESS in 5m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6a245f3704e94d169d629bbf9a2ec41e : SUCCESS in 5m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/223c810fe14d4c25927ec14aab87e5cb : RETRY_LIMIT in 1m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a20600fba2164974a0adffc5f13d3f97 : SUCCESS in 5m 11s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/59c8530a179d415d93a2d6a170e6e114 : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":2},{"id":"408c6f208702a771c01abd1fc877935490f15f56","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-09-07 01:18:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2c4eeb307a167e146f57e4ce6c42f53bb35b0104","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 01:18:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e7f9ded06e33409368aecb12663ea7a999ef7d4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 04:09:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64a68fde09ac4344960871d99a85de8b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c97cb8be53440b5925b8b908797bcdb : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3db831c7e71b44e18388f54355df77a8 : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ebd377ed1337437a8f98a21955556a97 : SUCCESS in 6m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c97cf39865b14d85849ef840a2b5998e : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7669fd88f35f457db4e2d8cb34b719ec : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":2},{"id":"307c3742a79d0127109f9566f9766baa63b4846b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 04:09:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5206940cceaa10ca980dc34e57b5ac194cbeabee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 04:12:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de4ddd8859b94e90a2edc19f179bf064\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1bfc3517506480d8a048755e7c19bb3 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e4079989db09bd2545735559c32a5195e532bf9f","revisions":{"61929c7afecd10fc4da21186470b81fa06e43d94":{"kind":"REWORK","_number":1,"created":"2023-09-06 11:56:36.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/66/893866/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/66/893866/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/893866/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/893866/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/893866/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/66/893866/1"}}},"commit":{"parents":[{"commit":"a34ed4fcbb474e3818e59cbe0fc318b30fd3341f","subject":"Merge \"[CI]: Add support for Ubuntu Jammy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a34ed4fcbb474e3818e59cbe0fc318b30fd3341f"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-09-06 11:54:40.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-09-06 11:54:40.000000000","tz":480},"subject":"Fix Zuul warnings with regexps","message":"Fix Zuul warnings with regexps\n\nzuul now reports warnings when using perl-style regexp:\n| The RE2 syntax error is: invalid perl operator: (?!\nfor\n| branches: ^(?!stable)\n\nThis commit fixes it by using the correct RE2 syntax\n\nmore details: https://zuul-ci.org/docs/zuul/latest/project-config.html#regular-expressions\n\nChange-Id: Iab9d4849d50c01ff35ce159a988fde95593f7ebf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61929c7afecd10fc4da21186470b81fa06e43d94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61929c7afecd10fc4da21186470b81fa06e43d94"}]},"branch":"refs/heads/master"},"e4079989db09bd2545735559c32a5195e532bf9f":{"kind":"REWORK","_number":2,"created":"2023-09-06 12:05:05.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/66/893866/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/66/893866/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/893866/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/893866/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/893866/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/66/893866/2"}}},"commit":{"parents":[{"commit":"a34ed4fcbb474e3818e59cbe0fc318b30fd3341f","subject":"Merge \"[CI]: Add support for Ubuntu Jammy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a34ed4fcbb474e3818e59cbe0fc318b30fd3341f"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-09-06 11:54:40.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-09-06 12:04:48.000000000","tz":480},"subject":"Fix Zuul warnings with regexps","message":"Fix Zuul warnings with regexps\n\nzuul now reports warnings when using perl-style regexp:\n| The RE2 syntax error is: invalid perl operator: (?!\nfor\n| branches: ^(?!stable)\n\nThis commit fixes it by using the correct RE2 syntax\n\nmore details: https://zuul-ci.org/docs/zuul/latest/project-config.html#regular-expressions\n\nChange-Id: Iab9d4849d50c01ff35ce159a988fde95593f7ebf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e4079989db09bd2545735559c32a5195e532bf9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e4079989db09bd2545735559c32a5195e532bf9f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
