)]}'
{"id":"openstack%2Ftrove~893984","triplet_id":"openstack%2Ftrove~master~I6100ef09901259e2189b3e327e5d4875a37f2721","project":"openstack/trove","branch":"master","topic":"py311","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-09-07 04:09:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6100ef09901259e2189b3e327e5d4875a37f2721","subject":"Fix bindep.txt for python 3.11 job(Debian Bookworm)","status":"MERGED","created":"2023-09-06 22:02:49.000000000","updated":"2023-09-07 04:11:45.000000000","submitted":"2023-09-07 04:09:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893984-py311","meta_rev_id":"d7ee6936caa257b313471484f377312ca65bdc89","_number":893984,"virtual_id_number":893984,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-07 04:09:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-07 01:17:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-07 01:17:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-06 22:16:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-07 00:51:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-07 01:17:53.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"dbb72f1cff81dc5ff5149e620a2f16f17f3e0c99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-09-06 22:02:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"26f007eef586d969a04b57f468824fc526fd7c06","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 22:16:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89ed47425e884ae5b0502f7093082054\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/28ec956392b04bc390c9c1e12e783bc1 : SUCCESS in 12m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3b160d9fd2864e35940307bba5cbfeaf : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/702714e5b8634726adbd70f2b5e85bee : SUCCESS in 11m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b5744033456a48639c7529a7431ba60d : RETRY_LIMIT in 1m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8a131da942f405843f96afa6fcafd3c86d410500","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 00:51:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02991a980e6e46d7bdc199f28a69a85b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c75affdff3ed422b885c81caf01c1202 : SUCCESS in 4m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/faa6061dc0da48389cc06a532fb3e4de : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a40b089281046af82948895b935e194 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/902d703e6f3a4aa5ba9adb7e0ecae371 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/02be5d281fd842a79a31e7337fda8ecd : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/72a83a93cb8a430f9c8b8454a732546f : SUCCESS in 4m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5b0e0c546ea345c2a232a436ddc8a84c : RETRY_LIMIT in 2m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be94d36e656e422481c5e3a9c3b4d1d5 : SUCCESS in 4m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ca094bb2f424422aea10358fbce9e42 : SUCCESS in 4m 16s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/ae7ad1a6d551419185c03deb7d229633 : SUCCESS in 2h 39m 43s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/d0d1e94550934dd1a14ad00a1a08ce34 : SUCCESS in 1h 40m 21s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/2126b14de1fa40ac8eae73b1a48b4af8 : SUCCESS in 2h 07m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"082787ebc5cb99a6526445aa6db8a4b18e492da4","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-09-07 01:17:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9d26f71a41f8709e397e6b5c355c8bd5536bbcd8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 01:18:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e90741ff376694a8d7a755bc7250217ba976b9ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 04:09:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c88745de7d0f4d5fb4022fc154de3bf4\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3861e426106f45a29e5edd7622a2481f : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f5688ac7ce04089be45b0ab9465bdf4 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a3712accc1f4aeebf210211d44538f4 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e93b7717e8b44319b8da1bb0906adf30 : SUCCESS in 5m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/224555d23de24ef280daf1ab0532d897 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18e75a2506ea4d4f9935b4450d0db59c : SUCCESS in 5m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/56ed1b05fad54d0c8bdbce57c7217177 : SUCCESS in 3m 58s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/863790f1c91d48688daaa3ad1859ad9a : FAILURE in 2h 49m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cd912e3aca1ba99bbc13c2e58eba7e2b12212197","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 04:09:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d7ee6936caa257b313471484f377312ca65bdc89","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 04:11:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cde29086ae0d41abb9e896419dc51120\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/0389322ff5a04953bca4280f7c9d03b2 : SUCCESS in 42s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/789a54304a7442c6acc150355d6c2c00 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80c2e09b098c4cdbaa23f1f7a0d527f2 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2bd3ead04948d381ae1cba1de2dabe0daed98cab","revisions":{"2bd3ead04948d381ae1cba1de2dabe0daed98cab":{"kind":"REWORK","_number":1,"created":"2023-09-06 22:02:49.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/84/893984/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/84/893984/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/893984/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/893984/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/893984/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/84/893984/1"}}},"commit":{"parents":[{"commit":"a34ed4fcbb474e3818e59cbe0fc318b30fd3341f","subject":"Merge \"[CI]: Add support for Ubuntu Jammy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a34ed4fcbb474e3818e59cbe0fc318b30fd3341f"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-09-06 21:51:51.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-09-06 21:51:51.000000000","tz":-300},"subject":"Fix bindep.txt for python 3.11 job(Debian Bookworm)","message":"Fix bindep.txt for python 3.11 job(Debian Bookworm)\n\nPython 3.11 job run on Debian Bookworm which does\nnot the mysql pkg for example, mariadb-server, libmysqlclient-dev\ninstead they have mariadb. Updating bindep.txt file\nto adjust the pkgs as per Debian bookworm.\n\nChange-Id: I6100ef09901259e2189b3e327e5d4875a37f2721\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2bd3ead04948d381ae1cba1de2dabe0daed98cab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2bd3ead04948d381ae1cba1de2dabe0daed98cab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
