)]}'
{"id":"openstack%2Ftrove~895058","triplet_id":"openstack%2Ftrove~stable%2F2023.2~Iacfb6f9f9d845e19fb549e417775c74949df3418","project":"openstack/trove","branch":"stable/2023.2","topic":"create-2023.2","hashtags":[],"change_id":"Iacfb6f9f9d845e19fb549e417775c74949df3418","subject":"Update .gitreview for stable/2023.2","status":"MERGED","created":"2023-09-14 01:23:33.000000000","updated":"2023-09-14 09:32:10.000000000","submitted":"2023-09-14 09:30:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"895058-create-2023.2","meta_rev_id":"1cdbb9abadcb7a653c046f44b725f05da8e7e2ba","_number":895058,"virtual_id_number":895058,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-14 09:30:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-14 09:14:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-14 09:14:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-14 01:48:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-14 09:14:25.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"c96959b6d406587c48409ed7e02fd303eba0fc2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2023-09-14 01:23:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab9622f273da3570c44e3606e8d01c56b1f227d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 01:48:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4810761b45704065b258f1fca1a98e9e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ef5afca441148eabc23eaa8981c5fbd : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14145d70246e45748cf654119ce3f6a2 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae6510d9278d4557a172d6118884f897 : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/207f36cd53ab439a9695cd240d6f84f9 : SUCCESS in 6m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aa97525d3d8448d180319536d7d60af2 : SUCCESS in 6m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5298c33b4a914313876a9143d74ce188 : RETRY_LIMIT in 2m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/464e11b4409b42018229438fd52814ed : SUCCESS in 5m 26s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3a0b01fde6824cb9b8279deafb5ca740 : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":1},{"id":"f59087f6275ca327bab34a93cb23c4a0c0fd8ec4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 01:57:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a9634dcab9a45bca2cef93d8a32ed11\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6faf27135d2849c797a8c00b13fddce9 : SUCCESS in 15m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d5c08b6768854902a74b7425a1e6fa3a : SUCCESS in 13m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/eef3543bc0fd4cadb832932bddab98e4 : SUCCESS in 14m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0a37baa00ac241deb5f90db88e4c8639 : RETRY_LIMIT in 2m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"03f090d62fb9e70b07817cd411dbfd91f754b26d","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-09-14 09:14:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d791f02a6cf58e2e4064eafdbe844f90173f6acc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 09:14:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"29e774ae09d8563400be586c14c85ca306f76276","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 09:30:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6986d2a1d23e4c598755ff7d33a66131\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e3e377d29ed419dbf93496165bb7c7d : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe9cab43508f49bfbc14f653233910ba : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66b544e590b248fe82a67ca484c89f15 : SUCCESS in 5m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/00bce53685224608896e296120215744 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89cc9d7bb2b14e8f94442da39e919fee : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":1},{"id":"208f85f132cae5f55fd564a67628ac82854af931","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 09:30:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1cdbb9abadcb7a653c046f44b725f05da8e7e2ba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 09:32:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7be63e811a9044f5b313353aaf11fbb4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90fc31e8297b48be93cebad963cdaa9c : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"73d553f6637e5c8c9a873d0cf07914c13a71d93a","revisions":{"73d553f6637e5c8c9a873d0cf07914c13a71d93a":{"kind":"REWORK","_number":1,"created":"2023-09-14 01:23:33.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/58/895058/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/58/895058/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/58/895058/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/58/895058/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/58/895058/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/58/895058/1"}}},"commit":{"parents":[{"commit":"0a6211645a12a7ae64259ebc47be4995f532e433","subject":"Merge \"Fix Zuul warnings with regexps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0a6211645a12a7ae64259ebc47be4995f532e433"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-09-14 01:23:32.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-09-14 01:23:32.000000000","tz":0},"subject":"Update .gitreview for stable/2023.2","message":"Update .gitreview for stable/2023.2\n\nChange-Id: Iacfb6f9f9d845e19fb549e417775c74949df3418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/73d553f6637e5c8c9a873d0cf07914c13a71d93a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/73d553f6637e5c8c9a873d0cf07914c13a71d93a"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
