)]}'
{"id":"openstack%2Ftrove~897637","triplet_id":"openstack%2Ftrove~master~I46902522645fc9b2697072dffe6373b527a2d962","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2023-10-25 06:30:37.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"I46902522645fc9b2697072dffe6373b527a2d962","subject":"[CI]: test mysql 5.7 version instead of 5.7.29","status":"MERGED","created":"2023-10-08 09:46:05.000000000","updated":"2023-10-25 08:59:36.000000000","submitted":"2023-10-25 08:55:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897637","meta_rev_id":"31dfab359557444a49ec712cdbe05be029a112d5","_number":897637,"virtual_id_number":897637,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-25 08:55:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-25 06:30:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-25 06:30:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-08 10:00:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-10-08 11:23:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c6d32f2e9fd6e4282dbfb2d06e49513790772dd4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-10-08 09:46:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20b5e409cc4e9b690f4ee42aefa49f1324f2929f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-08 10:00:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f5c4706e70e4ed99cbb1827d195fd91\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/32c871af00f043eb992fc80cc167debc : SUCCESS in 11m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0ad0becff54c449397fc8657adb27beb : SUCCESS in 11m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cd2f407a00fb40bf8d3faea2b7747aca : SUCCESS in 12m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a7cb982618d74c518381f64aa569074c : RETRY_LIMIT in 2m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0161e057b4603a2261dfb331f7e9c15795d1005c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-08 11:23:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f9895f6e4acf487c96260a7ed6a91785\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/494302cc7f574f6bb10f57153e3602a6 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75eed110a70d4784800630166ac07e73 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93ccb16421d64f97b004f28ad3b4dfeb : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c89f2221ad184023a82da432334769da : SUCCESS in 5m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ce3e569c8eca408da5ec13a214120b6c : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/56cccbf7d8de41d8915ce489e3f57967 : RETRY_LIMIT in 1m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb32d1e4b82241dfb36984f5e0b5e227 : SUCCESS in 5m 31s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8863ed62b20446fcae96ca7ec15fd6a1 : SUCCESS in 4m 07s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/77d594dcedc2447693e26920b655a278 : FAILURE in 1h 34m 37s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/02bd2a3cf4ce47419a8369092c5ee967 : FAILURE in 1h 09m 32s","accounts_in_message":[],"_revision_number":1},{"id":"71c3ab839b349d8e807e20c65f590dbd99021587","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-10-08 12:41:09.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9da1ad7ac976283f31f758654148e61e018cac0c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-08 12:56:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/928ecb5256a143c8838a3c3a43668400\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/865ab9949c2a4f58872b20671336e1f2 : SUCCESS in 12m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/efc07ec832ca4d4ab1c907e5694a8b96 : SUCCESS in 11m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4fd1884b63034b84914eaf291162fd92 : SUCCESS in 12m 22s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/46e0b91be0d94e55a065ada60295a7bb : RETRY_LIMIT in 2m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"42ed8ac3dd2a00f4da5b15629e90b841b657178d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-08 15:11:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7e3715fb46844e69af30b68f5a05279\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02d3f6363b0141a697585cea2be27b9e : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99defcc5048e44dfb46fe4d2d90972be : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d45817361f848bc858c3b6d597b3589 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/95df576244c6433ab1cd734fb9a3dcad : SUCCESS in 5m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7bf416d6017e48eda53dc113013ff259 : SUCCESS in 4m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fc51a7de4b384c34b678108c61b3d146 : RETRY_LIMIT in 1m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0148b745dc14ec4a63699604e8d878e : SUCCESS in 4m 38s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ee2e7bb3954241e589e50964c71361c1 : SUCCESS in 4m 32s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/195fe943685e490aa412666ecbbfbc00 : SUCCESS in 2h 28m 53s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/2217c12b2efe44de972af46132022f7b : SUCCESS in 1h 41m 37s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/04a0b4111791486782b81dfb146f9693 : SUCCESS in 2h 11m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d98162775790b1763cc029a9ec2426525f6c7aaf","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-10-23 02:47:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1cbcf83b318bd1f0a7224c2a3e92a9cb9171ecf7","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-10-23 02:47:53.000000000","message":"Patch Set 2: Code-Review-1 Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"a8778057ee34c4b29e38c566190d839298cb15d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 02:48:07.000000000","message":"Patch Set 2: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":2},{"id":"8533ff1b15004390a1332d77f230653454a96d38","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-10-23 05:33:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"85601326be8d9e1fdd7c1705679eaeaee6dbd8ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 07:59:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5757a6519d324bdbb17fc3b8309659d2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d31c12b88c5a4df487ffec94a5f8c101 : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3728bba4898942fabc850486c9611b53 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80187ebff8174b78aa9fc8be9e642e2e : SUCCESS in 5m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e451238c8efc4043929957ed661046d1 : RETRY_LIMIT in 1m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fca96e9b7bf472987dce6affbbcf321 : SUCCESS in 5m 41s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/53209d4438d7454a89377c730a74b751 : SUCCESS in 4m 23s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/561ec240a0f04919819046e230b6c518 : SUCCESS in 2h 24m 56s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/c6cc096710284a74854a5284a2b881c5 : SUCCESS in 1h 34m 19s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/118fa51688e54d309d7874eaffe6220c : SUCCESS in 2h 08m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"18028ed4e2f4a859ab4b35c89037741fb0662013","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-10-23 07:59:35.000000000","message":"Patch Set 2: -Code-Review -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"fb5c67088a9a936b948d5d0054de8ea3fbf5dd2a","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-10-23 07:59:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"81e54d1093bfda5e4863da1ea9f5c707804ff473","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 10:23:48.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d41ed29193484284994ce91271761978\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3df1a8915c44686a1e6deef2e172c87 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/038cec401ffa4686b8bbe0d40f9713e2 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/325d46c4067244b9a52d0f27810b5212 : SUCCESS in 5m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0f537199c2d24a5c993aea619fbf02d9 : RETRY_LIMIT in 1m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb004fad53c5488e9f48fa45df8f2aa1 : SUCCESS in 5m 04s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/34ad09428ba34e00a254f50b8d04b9c1 : SUCCESS in 4m 47s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/fb2e2aa271394de08a7a8974d614fbf0 : SUCCESS in 2h 22m 58s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/1251d59289b84ba497324b7ebf9d9c32 : SUCCESS in 1h 40m 37s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/6a42859cb2dd4fcf80d83c12cf00323d : SUCCESS in 2h 06m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"479115ac2c6e3aeda7d2494f15cf5a5306f616f0","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-10-23 15:01:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a33586bdcd5819d74598dfd189c65f9b386a70c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 15:13:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/956257c1d45d402c805cf8d221a6de42\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/227a2312549a4335993a033d8e010b65 : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/91156d18de0742a4b6b8ad6f4ccd36cb : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e198b65c7f2a40fdb7efb7a789479660 : RETRY_LIMIT in 1m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2c89f1880a0a5bcf7f14c5ab36e794dbff44317d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 17:29:35.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/615640ad710443378c78299c88814d43\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6f4fc6335fd4e7785f061ac5c576ef7 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7de16173dcf94f388ac3671f4cc141dd : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f35e48c69b554416abe4e816c3a72a37 : SUCCESS in 5m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/356f2df31c0a4c72abdebbb69be264e4 : RETRY_LIMIT in 2m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5556a69824e4ed7924f9080d88adf69 : SUCCESS in 5m 50s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8f74143280d94488ad941e69a53e518c : SUCCESS in 5m 15s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/05eb22e5105d4ee4a0759239416dab92 : SUCCESS in 2h 26m 40s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/2f729cb3da374defa3b719fbc35afa7a : SUCCESS in 1h 36m 00s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/e6984a6fdc2b4693973ae7726772243b : SUCCESS in 2h 08m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d2b42d29053e710ab5b0c1ba59b5c530688594e3","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-10-24 01:03:45.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"d5cc3a9b7b716ef293328d18c9eb6e1d9fa8ea40","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-10-24 01:22:24.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1d9dc7237833272dcab63e2218956532c966219d","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-10-24 02:46:33.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_26285\u003e with the following votes:\n\n* Code-Review+2 by \u003cGERRIT_ACCOUNT_26285\u003e\n* Workflow+1 by \u003cGERRIT_ACCOUNT_26285\u003e\n","accounts_in_message":[{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"_revision_number":2},{"id":"f3d86cc458f40c6be69219e464d4cf5658533b92","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-10-24 02:46:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"329f29a22497fb84a36d4adb9614a1f0678f231d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-24 04:00:13.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a0b822856aad48dc8dc30864e05c2eeb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d3fc9dfc0ef4af3ac3c5987abbc5ff0 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be311d5dec1b48589998aa1bfb313bc7 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2d9e0ab8c7d4aa6a8a7125e721f95ac : SUCCESS in 5m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a9659014076f421293057b396862589d : RETRY_LIMIT in 1m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05bbff6ea0af4d8db5490ed8cd968f43 : SUCCESS in 5m 44s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5d2edb1401ae42e196fe75507ed9941b : SUCCESS in 5m 56s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/5354f4fc08034dcf9722716d5187e22e : SUCCESS in 2h 36m 10s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/0276609d10ba4a3abcd05ef58b58ad1d : SUCCESS in 1h 46m 12s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/294d58e0156541b38c4cdecc1f1af478 : SUCCESS in 2h 20m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"139c3b79fa75b9343df62751462af6e1d6642e20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-10-25 01:12:21.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"adebf31c934268de8ed0e404ea23fd253970be4c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 01:37:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12195d205714408bb0c90f0b341d1a03\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f9c42e03dcc447628bd178e16ad53c91 : SUCCESS in 21m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4cdd3025e65c41a682bfa10dbb16723e : SUCCESS in 22m 04s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7e9dbd1508d24b6bb027aeae788a405a : SUCCESS in 5m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"202f985b08255575476883fb319943f94c276f42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 03:38:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/832b3c8cd1684754bf547f7eb6ce62f2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4576a466a5b54c1cb010702ad35a5489 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f87cfbe47e54755800b4fdaa9c25524 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c35516963a34b7badb7e92acda00f0e : SUCCESS in 5m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2ebd6eaffd0245acb9ee1b01c51573c6 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69351fbda10b49188ae36298a8edc49e : SUCCESS in 6m 19s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f5fe6c3da8bb4c03a2d1219f89cd3941 : SUCCESS in 5m 20s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/cd7bc6b5c94e4060a70b18592de6a9fc : SUCCESS in 2h 24m 23s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/25a7df93268d40e1ba7cfcf818aad051 : SUCCESS in 1h 36m 55s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/4d5d5faddbc8403893cb1e688779d2ba : SUCCESS in 2h 12m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"dbb25fa5c4547e22f38bae28465a956101a12101","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-10-25 06:30:37.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9d8d801736f48d9ac494921f130037c4b0131077","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 06:31:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ef51d265b6e03067edf1c49126be50b814f26e71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 08:55:09.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25d0998de4de4cecac30779794bf6409\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c076da5db26d433c8c671e8b9f6a84ab : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db62ed93fd374f748eb94d3fd7a65d93 : SUCCESS in 5m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f822a7053eb84f0591487c86c5acf912 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/436776db99e444ab83408e0fbca4da53 : SUCCESS in 4m 49s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/2acda70b469b4a6b8dbd5acf2ff36789 : SUCCESS in 2h 23m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"965578e14b76fba65e413562c95a238083b97f56","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 08:55:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"31dfab359557444a49ec712cdbe05be029a112d5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 08:59:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3624c17adba74126b3f9d31dace5e687\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c94f25090ac452ca0ce12b6c7aa8d76 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d438839cf9ab4d5af4cf73544ad7af82183e9f4d","revisions":{"3ea0eaf8c1f0c06b84e4b46e60ecede195e59714":{"kind":"REWORK","_number":1,"created":"2023-10-08 09:46:05.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/37/897637/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/37/897637/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/37/897637/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/37/897637/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/37/897637/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/37/897637/1"}}},"commit":{"parents":[{"commit":"3bbeeb87e074016db45e6df397b0b379210a5549","subject":"Merge \"Separate backup docker image for each database version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3bbeeb87e074016db45e6df397b0b379210a5549"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-10-08 09:45:11.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-10-08 09:45:11.000000000","tz":480},"subject":"[CI]: test mysql 5.7 version instead of 5.7.29","message":"[CI]: test mysql 5.7 version instead of 5.7.29\n\nChange-Id: I46902522645fc9b2697072dffe6373b527a2d962\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ea0eaf8c1f0c06b84e4b46e60ecede195e59714"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ea0eaf8c1f0c06b84e4b46e60ecede195e59714"}]},"branch":"refs/heads/master"},"46ad35ee2bbda648dd958d3dad8dc0e461c5b214":{"kind":"REWORK","_number":2,"created":"2023-10-08 12:41:09.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/37/897637/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/37/897637/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/37/897637/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/37/897637/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/37/897637/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/37/897637/2"}}},"commit":{"parents":[{"commit":"3bbeeb87e074016db45e6df397b0b379210a5549","subject":"Merge \"Separate backup docker image for each database version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3bbeeb87e074016db45e6df397b0b379210a5549"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-10-08 09:45:11.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-10-08 12:40:51.000000000","tz":480},"subject":"[CI]: test mysql 5.7 version instead of 5.7.29","message":"[CI]: test mysql 5.7 version instead of 5.7.29\n\nChange-Id: I46902522645fc9b2697072dffe6373b527a2d962\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/46ad35ee2bbda648dd958d3dad8dc0e461c5b214"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/46ad35ee2bbda648dd958d3dad8dc0e461c5b214"}]},"branch":"refs/heads/master"},"d438839cf9ab4d5af4cf73544ad7af82183e9f4d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-10-25 01:12:21.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/37/897637/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/37/897637/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/37/897637/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/37/897637/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/37/897637/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/37/897637/3"}}},"commit":{"parents":[{"commit":"87ba3b787600c11cb24a5c995aa1cce532573ef4","subject":"[CI] fix openstack-tox-py311 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/87ba3b787600c11cb24a5c995aa1cce532573ef4"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-10-08 09:45:11.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-10-25 01:12:21.000000000","tz":0},"subject":"[CI]: test mysql 5.7 version instead of 5.7.29","message":"[CI]: test mysql 5.7 version instead of 5.7.29\n\nChange-Id: I46902522645fc9b2697072dffe6373b527a2d962\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d438839cf9ab4d5af4cf73544ad7af82183e9f4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d438839cf9ab4d5af4cf73544ad7af82183e9f4d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
