)]}'
{"id":"openstack%2Ftrove~899434","triplet_id":"openstack%2Ftrove~master~I9bd497c85c6eba67485840d07a80cca8eed27994","project":"openstack/trove","branch":"master","topic":"add-region-name","attention_set":{},"removed_from_attention_set":{"9426":{"account":{"_account_id":9426,"name":"Gabriel Adrian Samfira","email":"gsamfira@cloudbasesolutions.com","username":"gsamfira"},"last_update":"2023-10-27 09:18:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9bd497c85c6eba67485840d07a80cca8eed27994","subject":"Add region name to trove-guestagent","status":"MERGED","created":"2023-10-26 22:01:06.000000000","updated":"2024-01-16 03:44:49.000000000","submitted":"2023-10-27 09:18:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"899434-add-region-name","meta_rev_id":"cfa514de8da720fb656ae3871e8b6413f41536a4","_number":899434,"virtual_id_number":899434,"owner":{"_account_id":9426,"name":"Gabriel Adrian Samfira","email":"gsamfira@cloudbasesolutions.com","username":"gsamfira"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-27 09:18:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"all":[{"value":2,"date":"2023-10-27 06:54:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"all":[{"value":1,"date":"2023-10-27 06:54:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-26 22:17:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-10-27 00:27:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-27 06:54:29.000000000","updated_by":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"REVIEWER"},{"updated":"2024-01-11 21:19:30.000000000","updated_by":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"reviewer":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"state":"CC"},{"updated":"2024-01-12 07:02:19.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"CC"}],"messages":[{"id":"91594110c48474574862d44887b5ec257b2f3ded","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9426,"name":"Gabriel Adrian Samfira","email":"gsamfira@cloudbasesolutions.com","username":"gsamfira"},"date":"2023-10-26 22:01:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bbbb8f7c7c9f84795b572038f1182d16bc6c9195","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9426,"name":"Gabriel Adrian Samfira","email":"gsamfira@cloudbasesolutions.com","username":"gsamfira"},"date":"2023-10-26 22:03:07.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"070e3725d2ef837248abf3a6415e1098c23c30a5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 22:17:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba90fcdffe5e45ed8c5489a4ac276330\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9ffc031d7bf2431e99568ef6d76734cb : SUCCESS in 12m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2b8340204f69468f84eacd8e7cf97f51 : SUCCESS in 12m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3405f7951d24449b82a1601106f04922 : SUCCESS in 11m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9684188cbf8d581f55387decf9e98a536f4b9c4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 00:27:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ede1003a364d4686a67cd69698ba3d99\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e31326680d31440899c1fdf4f88da142 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/235411a76b2b4dda912a816bb87c3fc2 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9448ddaea594f23aa383c54f43e774b : SUCCESS in 5m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1cfef7d53c3c44fa94fc2dbe852a58ef : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6364f8643bf74de4a81d03c89b033f7f : SUCCESS in 5m 27s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a5cafee91f9c4d508faa743e7276ce96 : SUCCESS in 5m 23s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/1eb61ea3ea04430bb01f5df763f368ce : SUCCESS in 2h 23m 44s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/b43dccf7340a40ceafb636bb4b559633 : SUCCESS in 1h 36m 48s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/cfb35342b2b04c2bafd69508b79bab2e : SUCCESS in 2h 10m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"98a11633236f78508ab1c07925716a13b42ea8f7","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-10-27 06:54:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d3eb227d9748b4c4a5047e97bc53ffc399f88ab9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 06:54:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cb3b946c673ee7519440479f4d3e768ea6100bee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 09:18:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4ef147fa2dc44c499d1c5ed026ce02d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8b908173fcf4229850fbd0f2ae6e7fb : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac7a230981094cbf89ddfac27e0c089d : SUCCESS in 5m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dd24dacbdef6441aaba89afee6f051d6 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/428ef812ab2247ed9372bca43d5805c1 : SUCCESS in 5m 31s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/9c8e7a8e173f4a238e1a7992584571ef : SUCCESS in 2h 23m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3c92d9804cd43cc2bed7ffff0bd3e704752ee7fa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 09:18:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"451f17264e90da64f59442815b5e4e905733fc28","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 09:20:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b72b254cbc4447ec8d174851e0f7453c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afb083604fae4922a7e2246b582a8148 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2},{"id":"dd1152cf05e94e72fe60a75755a8fff460f806b4","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2024-01-11 21:19:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"46cc9e4ffc87c8ea5b5bbee76482e1757a607a4c","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-01-12 07:02:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"70d50ca3e602ee3bab96e38ff83d57b8c150d7b4","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-01-12 07:31:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d2b26191990cc261e6031e568f84f584b2a167ae","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2024-01-12 17:35:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4315ee5774e26af118d628ae574b26ea728604f5","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2024-01-12 20:49:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"81e8d828f97a9b884092942b4332957c757e5549","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-01-13 03:01:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"254726bf14d32e118b885c8ab1ea5c8416937ff0","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-01-16 01:13:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cfa514de8da720fb656ae3871e8b6413f41536a4","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-01-16 03:44:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e998b6886602575127ebe613e56cee3a5a01c6c6","revisions":{"f82e63de781ca8f3f1daa3ae6f5844901d5de006":{"kind":"REWORK","_number":1,"created":"2023-10-26 22:01:06.000000000","uploader":{"_account_id":9426,"name":"Gabriel Adrian Samfira","email":"gsamfira@cloudbasesolutions.com","username":"gsamfira"},"ref":"refs/changes/34/899434/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/34/899434/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/34/899434/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/34/899434/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/34/899434/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/34/899434/1"}}},"commit":{"parents":[{"commit":"b5e85ae2aee3b7d77b3b32bd2de0bc23430de847","subject":"follow up: 3bbeeb87e074016db45e6df397b0b379210a5549","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b5e85ae2aee3b7d77b3b32bd2de0bc23430de847"}]}],"author":{"name":"Gabriel Adrian Samfira","email":"gsamfira@cloudbasesolutions.com","date":"2023-10-26 17:26:34.000000000","tz":180},"committer":{"name":"Gabriel Adrian Samfira","email":"gsamfira@cloudbasesolutions.com","date":"2023-10-26 21:55:08.000000000","tz":180},"subject":"Add region name to trove-guestagent","message":"Add region name to trove-guestagent\n\nIn a shared keystone multi-region setup, we must set the region_name\nwhen using any of the OpenStack clients. Without the region name,\nclients may attempt to use the API endpoints in the wrong region.\nBackups currently fail if there are multiple regions created in the\nsame keystone catalog, due to the fact that the trove guest agent\nattempts to connect to the S3 endpoint in another region.\n\nStory: 2010950\nTask:  48994\n\nChange-Id: I9bd497c85c6eba67485840d07a80cca8eed27994\nSigned-off-by: Gabriel Adrian Samfira \u003cgsamfira@cloudbasesolutions.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f82e63de781ca8f3f1daa3ae6f5844901d5de006"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f82e63de781ca8f3f1daa3ae6f5844901d5de006"}]},"branch":"refs/heads/master"},"e998b6886602575127ebe613e56cee3a5a01c6c6":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-10-26 22:03:07.000000000","uploader":{"_account_id":9426,"name":"Gabriel Adrian Samfira","email":"gsamfira@cloudbasesolutions.com","username":"gsamfira"},"ref":"refs/changes/34/899434/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/34/899434/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/34/899434/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/34/899434/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/34/899434/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/34/899434/2"}}},"commit":{"parents":[{"commit":"b5e85ae2aee3b7d77b3b32bd2de0bc23430de847","subject":"follow up: 3bbeeb87e074016db45e6df397b0b379210a5549","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b5e85ae2aee3b7d77b3b32bd2de0bc23430de847"}]}],"author":{"name":"Gabriel Adrian Samfira","email":"gsamfira@cloudbasesolutions.com","date":"2023-10-26 17:26:34.000000000","tz":180},"committer":{"name":"Gabriel Adrian Samfira","email":"gsamfira@cloudbasesolutions.com","date":"2023-10-26 22:02:49.000000000","tz":180},"subject":"Add region name to trove-guestagent","message":"Add region name to trove-guestagent\n\nIn a shared keystone multi-region setup, we must set the region_name\nwhen using any of the OpenStack clients. Without the region name,\nclients may attempt to use the API endpoints in the wrong region.\nBackups currently fail if there are multiple regions created in the\nsame keystone catalog, due to the fact that the trove guest agent\nattempts to connect to the S3 endpoint in another region.\n\nStory: 2010950\nTask: 48994\n\nChange-Id: I9bd497c85c6eba67485840d07a80cca8eed27994\nSigned-off-by: Gabriel Adrian Samfira \u003cgsamfira@cloudbasesolutions.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e998b6886602575127ebe613e56cee3a5a01c6c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e998b6886602575127ebe613e56cee3a5a01c6c6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
