)]}'
{"id":"openstack%2Ftrove~900403","triplet_id":"openstack%2Ftrove~master~I5635a7e5fb2fe8bbd6e6dfa56fe7969f3a03cf9c","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2023-11-13 07:31:03.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"I5635a7e5fb2fe8bbd6e6dfa56fe7969f3a03cf9c","subject":"[CI] Add mariadb10.4 jobs in zuul.","status":"MERGED","created":"2023-11-08 09:41:05.000000000","updated":"2023-11-13 10:04:43.000000000","submitted":"2023-11-13 10:00:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900403","meta_rev_id":"a4a6de2da8c51a5989208676eaad90223738aaf8","_number":900403,"virtual_id_number":900403,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-13 10:00:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-13 07:31:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-13 07:31:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-08 09:53:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-08 11:55:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"09e7e02e8a04999e2e9b54b682a31b5a69b82cfe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-11-08 09:41:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ffdeaf3d7db2fdad9c32d7365373fa5e8e3bc9df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 09:53:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a86ec02dca14fbcac21cd8fb537fe16\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4fcc44d83090499aa80c2f628738d931 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dabf53ae2dbc4598b1598eced7791f21 : SUCCESS in 8m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e290cc57dcd74e388642e1377cc68475 : SUCCESS in 11m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"caca04258e20d058b21ac1001527fc70619a6a90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 11:55:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e8c74ad67821496b8498329244b4d5f2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37e9f533487f4299ac80b743fce4a9ec : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56d9bc118c344a448b834c5c3f8f5079 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b27ce823eae94f7fb25ddd2e77da1d15 : SUCCESS in 6m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b086581781e74ac5abba03c911240a7a : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79e5f44f835f4820b223fd9422d14a7a : SUCCESS in 5m 41s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/b623ed930f574470aba1e60d6c99bd50 : SUCCESS in 1h 53m 23s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/cbcd906f6b1044b7a8f8d1d8c53794ce : SUCCESS in 1h 45m 31s\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/1cf27730d15342328b7189707d6c0fdd : FAILURE in 2h 12m 13s\n- trove-tempest-ubuntu-base-mariadb10.11 https://zuul.opendev.org/t/openstack/build/a5bb212bc7474963be73ebbc6c315b35 : FAILURE in 27m 36s\n- trove-tempest-ubuntu-backup-mariadb10.11 https://zuul.opendev.org/t/openstack/build/a51e4eb6e678442d9bf1646e5b4d6e24 : FAILURE in 38m 51s\n- trove-tempest-ubuntu-replication-mariadb10.11 https://zuul.opendev.org/t/openstack/build/f6689bcc64ff4ebc8713199d52891cb8 : FAILURE in 39m 11s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/845a8ab17a0b424fb67c5f85f19ae346 : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":1},{"id":"f83fc8104be24c59f02c9b39c820cd3c67c67102","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-11-08 14:51:53.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"210fe15d953fe50edfdf1e354ff85881a1b0fe97","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 15:12:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b407bc81cbe4cf1a773f1fb1b8fb5dc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2888a02379114418b6b1b53a6a31acf5 : SUCCESS in 15m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bf088a3b1e93489b9ecf0dfa3829580c : SUCCESS in 17m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7968a38533af47ba9a5d22855d32499f : SUCCESS in 6m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f1afca753181cc5ff33941f1c3fdb5e1df127e8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 16:41:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d1f37f2b51e47b1a5553e581c292eed\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ebf09d703db4732a00884dec82ef9a7 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/289f6084bff74786a76d579692bd2e23 : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cca472534a724f9ba0b6f56383880592 : SUCCESS in 7m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b0d9fa70dcb04b78a9aef654d6601503 : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a75ded586e05483db00c95e9eea706b5 : SUCCESS in 5m 44s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/fc61100e165c4220846eb129995f1c64 : SUCCESS in 1h 27m 30s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/3627590af1f9486eaadc704a51560782 : FAILURE in 1h 47m 58s\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/eaa62114387247b388e5a91951db5072 : SUCCESS in 1h 31m 21s\n- trove-tempest-ubuntu-base-mariadb10.11 https://zuul.opendev.org/t/openstack/build/d0afbb243d6346c09fe03b6bd56675a7 : SUCCESS in 1h 19m 31s\n- trove-tempest-ubuntu-backup-mariadb10.11 https://zuul.opendev.org/t/openstack/build/770d42eeb2ae4f2b9d748554f4f5a199 : FAILURE in 38m 19s\n- trove-tempest-ubuntu-replication-mariadb10.11 https://zuul.opendev.org/t/openstack/build/3b0b5c1894004f1b8b02ab1f2d411046 : FAILURE in 40m 11s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3e3742c86cf6473e8165a2130e90c1fd : SUCCESS in 5m 56s","accounts_in_message":[],"_revision_number":2},{"id":"e42ab453a2e2e58b73372f9c1b16586de7fc327c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-11-09 03:21:16.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"2a4a32cd2192946b70e8dbf41ff2ffd345681137","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 03:37:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7e85c9d23ff43caa60c44d03f76ff26\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1136c7c8737f4c12a9b555157b694ffd : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/16c112a40039434eb48314dbc29c7bbe : SUCCESS in 14m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6f94f6eec7c942df84bad4d60361daa8 : SUCCESS in 5m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"04a08063b963d9fd8f2e1cf489c6662d3be51e13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 06:26:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f0a5434fe6b4b76b8a53ed0f3ea15ee\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1609fdbe91ca4f818f91aba2e95b5288 : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d4f7e77da9f4564b5fa9aa9a9467cf3 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86b080f9dcde4ac98074e49c3381fdd8 : SUCCESS in 6m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d14cbc27caf14df0b9583fbdfbd7fee5 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46603dc8cf1f433b9823caf576c43a5c : SUCCESS in 5m 35s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/6f1db03fedad43b585e41a1ec3b82574 : SUCCESS in 1h 51m 56s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/7d5bff2e956243a89d72d849e3f8e5f8 : SUCCESS in 2h 12m 39s\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/f666e72dd6824fa089e60f6e371d111d : SUCCESS in 1h 40m 28s\n- trove-tempest-ubuntu-base-mariadb10.11 https://zuul.opendev.org/t/openstack/build/b9aac52277c940f8959914ff941cd637 : SUCCESS in 1h 38m 15s\n- trove-tempest-ubuntu-backup-mariadb10.11 https://zuul.opendev.org/t/openstack/build/633de959d6534454a62976910b096c73 : TIMED_OUT in 3h 02m 36s\n- trove-tempest-ubuntu-replication-mariadb10.11 https://zuul.opendev.org/t/openstack/build/4413ab32058b4eb3815d09d0198d1d71 : FAILURE in 2h 08m 42s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/26e87874aedd440480841c7071ca470e : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":3},{"id":"e4678d4d731a437409258f3adbf043f618f6a713","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-11-12 14:47:55.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"3cfb869375443ae31db65c8d86c349cebf018427","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-12 15:06:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c24fb7241a894c408185c5be32a9b35f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3c6ea852417e4e65986c12041a5ba9b7 : SUCCESS in 12m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d7219bc4e6cb48d1bcbdaee60de4ac4b : SUCCESS in 16m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0639f3ba46f647fe98ad544bd7abe077 : SUCCESS in 5m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7a146f5f36bd24fb08e2421f439e4f94a0ecf51b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-12 16:47:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76065c1169fb44beb87b796ddb345902\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81aa3b8426b04acd8abe89700576e925 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c3c15afb5c344e393bde7ad92635771 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa7d113d8ca24ae18528fb7b1605e032 : SUCCESS in 5m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/653eb739fc6a43b5bc823fab156c1eb4 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/135fcfb26a5c41ee8fc8725b7c814fa5 : SUCCESS in 4m 24s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/49d7b432eb544ffabecaf486136a4918 : SUCCESS in 1h 33m 49s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/a7e2448431834a8ab0b17ec509913984 : SUCCESS in 1h 50m 38s\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/2afc897741904f99a2e1550266616200 : SUCCESS in 1h 34m 03s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/8dac61dbb3b24f86a0214c9ae85cdd18 : SUCCESS in 1h 27m 25s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/e569511dbc3941a59b68f2d6b1e4772c : SUCCESS in 1h 53m 07s\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/79b559e304e0403681e1b6ebf0cb30bd : SUCCESS in 1h 30m 08s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b0100a0ff9264a32bc9287074dce6d94 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":4},{"id":"f66a644d0659da266a34d540fce779723821847b","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-11-13 07:31:03.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"582cc20a7b1d926659d5d145be5623872f6c9a6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 07:31:39.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c62cc1bb549d23910b4339cf634863d1f30a6c29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 10:00:23.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/783fd1e214074280b502d161e37a932b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2fd5786a8c241429e3de28a0c1ac34d : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26e8429aa38848d0888c52e16cf3eb8a : SUCCESS in 5m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/94e9619c0a914bc5abc68746582f9593 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c75283230d064c4bb35a51ca6c6c224b : SUCCESS in 4m 13s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/726774901f6d4b3cb2cc94c0acb69f78 : FAILURE in 2h 27m 39s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"44a584a5a7c85dd7f87da1db55178f5efef7761e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 10:00:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"a4a6de2da8c51a5989208676eaad90223738aaf8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 10:04:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c89d146d8404421c933e13ab86cfc6be\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14df52c05ab84869b112b31814ee3d20 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"603a2f48f0e9dc42cebb53df95f18344e32df8f5","revisions":{"56aa1ef5e13f9f794d5a3051b70bb3cff140f817":{"kind":"REWORK","_number":1,"created":"2023-11-08 09:41:05.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/03/900403/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/03/900403/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/900403/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/900403/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/900403/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/03/900403/1"}}},"commit":{"parents":[{"commit":"a60b63d7abd444c2ef44bfde6369d826913e1e1b","subject":"Merge \"[CI] Improve CI tests for mysql\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a60b63d7abd444c2ef44bfde6369d826913e1e1b"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-11-08 09:32:47.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-11-08 09:41:01.000000000","tz":480},"subject":"[CI] Add mariadb10.11 jobs in zuul.","message":"[CI] Add mariadb10.11 jobs in zuul.\n\nAs version 10.11 for mariadb is the latest long-term\nstable version, so we need to test it in zuul.\n\nChange-Id: I5635a7e5fb2fe8bbd6e6dfa56fe7969f3a03cf9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/56aa1ef5e13f9f794d5a3051b70bb3cff140f817"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/56aa1ef5e13f9f794d5a3051b70bb3cff140f817"}]},"branch":"refs/heads/master"},"6ab3917f7af38ec2a42d774900b9fd74f82b70d4":{"kind":"REWORK","_number":2,"created":"2023-11-08 14:51:53.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/03/900403/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/03/900403/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/900403/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/900403/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/900403/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/03/900403/2"}}},"commit":{"parents":[{"commit":"a60b63d7abd444c2ef44bfde6369d826913e1e1b","subject":"Merge \"[CI] Improve CI tests for mysql\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a60b63d7abd444c2ef44bfde6369d826913e1e1b"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-11-08 09:32:47.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-11-08 14:51:37.000000000","tz":480},"subject":"[CI] Add mariadb10.11 jobs in zuul.","message":"[CI] Add mariadb10.11 jobs in zuul.\n\nAs version 10.11 for mariadb is the latest long-term\nstable version, so we need to test it in zuul.\n\nChange-Id: I5635a7e5fb2fe8bbd6e6dfa56fe7969f3a03cf9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6ab3917f7af38ec2a42d774900b9fd74f82b70d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6ab3917f7af38ec2a42d774900b9fd74f82b70d4"}]},"branch":"refs/heads/master"},"c5d3f1c46af493ef2ee657d8d638fcee755b2b2b":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-11-09 03:21:16.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/03/900403/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/03/900403/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/900403/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/900403/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/900403/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/03/900403/3"}}},"commit":{"parents":[{"commit":"a60b63d7abd444c2ef44bfde6369d826913e1e1b","subject":"Merge \"[CI] Improve CI tests for mysql\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a60b63d7abd444c2ef44bfde6369d826913e1e1b"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-11-08 09:32:47.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-11-09 03:18:37.000000000","tz":480},"subject":"[CI] Add mariadb10.11 jobs in zuul.","message":"[CI] Add mariadb10.11 jobs in zuul.\n\nAs version 10.11 for mariadb is the latest long-term\nstable version, so we need to test it in zuul.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/900498\n\nChange-Id: I5635a7e5fb2fe8bbd6e6dfa56fe7969f3a03cf9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c5d3f1c46af493ef2ee657d8d638fcee755b2b2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c5d3f1c46af493ef2ee657d8d638fcee755b2b2b"}]},"branch":"refs/heads/master"},"603a2f48f0e9dc42cebb53df95f18344e32df8f5":{"kind":"REWORK","_number":4,"created":"2023-11-12 14:47:55.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/03/900403/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/03/900403/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/900403/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/900403/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/900403/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/03/900403/4"}}},"commit":{"parents":[{"commit":"a60b63d7abd444c2ef44bfde6369d826913e1e1b","subject":"Merge \"[CI] Improve CI tests for mysql\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a60b63d7abd444c2ef44bfde6369d826913e1e1b"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-11-08 09:32:47.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-11-12 14:43:49.000000000","tz":480},"subject":"[CI] Add mariadb10.4 jobs in zuul.","message":"[CI] Add mariadb10.4 jobs in zuul.\n\nAdd mariadb10.4 tests in zuul.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/900498\n\nChange-Id: I5635a7e5fb2fe8bbd6e6dfa56fe7969f3a03cf9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/603a2f48f0e9dc42cebb53df95f18344e32df8f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/603a2f48f0e9dc42cebb53df95f18344e32df8f5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
