)]}'
{"id":"openstack%2Ftrove~901169","triplet_id":"openstack%2Ftrove~master~I6156b289659dc9130effb1666c3cd2e14366bffe","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"28691":{"account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"last_update":"2023-12-08 05:19:26.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2023-12-08 03:36:28.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"I6156b289659dc9130effb1666c3cd2e14366bffe","subject":"Fix guest agent failed to use swiftclient over TLS","status":"MERGED","created":"2023-11-16 14:55:44.000000000","updated":"2023-12-08 05:24:03.000000000","submitted":"2023-12-08 05:19:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901169","meta_rev_id":"4d4ed7c3658817bf28a4af8ca6c1a3eefe68c54a","_number":901169,"virtual_id_number":901169,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-08 05:19:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-12-08 03:36:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-12-08 03:36:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-16 15:12:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-16 16:40:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-17 10:04:16.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"REVIEWER"}],"messages":[{"id":"266734a6d30c4edb04290bd368b82401fe7c4df8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-11-16 14:55:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"68a807e9b4019e0f8ee18007d4d9289416efc4f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-11-16 14:58:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e906066ec485eaa7285886bc41eebccda1ac33b7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 15:12:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50dd98cec1044719afe8aa5175326393\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a230c77f1171424396f3ecc158e64047 : SUCCESS in 8m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0b86309c479943688945e435379a0d0f : SUCCESS in 8m 47s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8287a2090b7f4394b5b5193c737f0b44 : SUCCESS in 11m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2bac43e146145a61e75dad9b93838ace3affef8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 16:40:46.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2dc16b37e2f44702a90a681a12fd366b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35dc94a1cecf4311ad5a55bcebb64079 : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8eefdd24375f4dc5870b302c79851c1d : FAILURE in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e95b60cdb9c64aff9d57eabaac44ea20 : SUCCESS in 6m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7eea98247da0440b8ee538e8647a28e1 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4f453ee089c4a3593fd195ea8a21b2b : SUCCESS in 5m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e1befb0e50a4ba294a45bcca81260af : SUCCESS in 5m 40s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/fa8ad95f29f64c68897ab8ee81fa2497 : SUCCESS in 1h 18m 35s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/d3a9a397111c4257a3805a1ce0ff2732 : SUCCESS in 1h 41m 05s\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/496955db1d9d470e87289f762f1cad0a : SUCCESS in 1h 24m 41s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/9e3e8c8451574788ad10d5696865386c : SUCCESS in 1h 13m 10s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/6d120416f0554db885089bfb331f361b : SUCCESS in 1h 27m 13s\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/70d79c406b97419db68057ffc1277621 : SUCCESS in 1h 30m 39s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/ea0fa84490bd4cef8d55726195be68d8 : SUCCESS in 1h 36m 13s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/ebf54cb773204e52871e358b1c9977cd : SUCCESS in 1h 31m 34s\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/e99d381e1bcb42cca268831d3faeb933 : SUCCESS in 1h 24m 19s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6501d90e453e407a8aefa71061980117 : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":2},{"id":"956f8655882f21b30b63db3e8ae7c8b1c5def834","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-11-17 01:10:59.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"9fd97b31256c5a4967f98a7b4eafb89963b66332","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 01:17:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54e6c44e71a54c08b792a14d0a0e1233\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/41b5d4514caf4a48b9159b451443ed4a : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ab034a05dd23405f8c34354974cc3585 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cfbe82fcf32744f59d623a05a7c95b96 : SUCCESS in 5m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"75335fda5ecf7a4d187c30be4d49dab94b30db00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 03:05:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8c24a479afa47508f92020c5d6d8c0e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a25d74b5489340c08928938f45c32d38 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2b10ac763994bb4a07fd2cf4de24674 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/103d027d8a874b2e8842ac074e64a1ac : SUCCESS in 5m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/baa55bf871fc476fb60ec0a0150cbef8 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3419212c2ea04dc2b79e019b2e032b70 : SUCCESS in 5m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bac2535fdb624f5caf17623be75c49e2 : SUCCESS in 5m 09s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/54a4ac46462440c5b6dc111c1d4e1ae0 : SUCCESS in 1h 16m 57s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/e8715868846646be8e3bf92ff470fe84 : SUCCESS in 1h 34m 57s\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/107121ad369f45d4ba05b65334fb6f0c : SUCCESS in 1h 18m 46s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/46a84061477f437da7a053e62e5a7a83 : SUCCESS in 1h 11m 03s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/63c3a9139509421c873146258a8ce8cc : SUCCESS in 1h 25m 03s\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/e6da5756932b4ab48b30050d97b27c0d : FAILURE in 1h 24m 44s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/db6be7807b1441c187baba1e49bd9177 : FAILURE in 1h 52m 39s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/0104b20d74f74129b82974e672c2ef96 : SUCCESS in 1h 52m 57s\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/54fe465dbe504a11ab81883d34660cfb : SUCCESS in 1h 08m 04s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a6f33d9bf5fa480f8f7fc9ac13b24d82 : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":3},{"id":"bb6b0f08bdda0d6965c4ea65af11a9ea3e1d2ecf","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-11-17 06:56:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"53dc46d3c47c3243c08a2892a37c1d08cae99093","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 07:10:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21ad7e9ab2624d4fa29603a3fec0b99c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/652af45f36a542a9af04528114965993 : SUCCESS in 11m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/184d715bedf3402fb71eb4a95b19c070 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/44dd0a84ee764bf9bfa5cc9e3ceece65 : SUCCESS in 10m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ae5adad809aadd7b12ff7c447bdb4ae34189a41a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 09:08:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae32b3f5d6134616ab3404c9edaf0313\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb90778ee08f41918acd980f6bc4d1c3 : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b141cf41bcfb40f9a856e386dc964ea4 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d182117908ad4751a150dc9bba3357d3 : SUCCESS in 5m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/951a955e146d42a0b6b45c3ca4e52e4b : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e254e76125146a9be11070a074030de : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33272a2c67264d01bb84597b9f9cfcf2 : SUCCESS in 6m 47s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/eac7231c421e4b64a453e6a13694ddd4 : SUCCESS in 1h 42m 34s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/9bfc06bd2c1d4d0db19ad588b79671ee : SUCCESS in 2h 08m 34s\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/49b58b0987ad44f0b00924ee4b77b28d : SUCCESS in 1h 52m 11s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/aa11e95330b4407299ce15c38a325b28 : SUCCESS in 1h 41m 20s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/eea1a81e9bc0404abc0e4a345c0d2493 : SUCCESS in 2h 06m 41s\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/5ff69f1e676b4c7080471315ec776431 : SUCCESS in 1h 44m 37s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/ae0bef1fbda141dea522d6f072448ec2 : SUCCESS in 1h 15m 55s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/9c2c5f71225b4fd3a29e1981b02aaceb : SUCCESS in 1h 32m 31s\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/34e3e48180a344a1bd4312719bebd614 : SUCCESS in 1h 19m 07s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e530dd6beeba48959bb9cf5abd74472c : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":3},{"id":"ddb27d58f2652d11d2854c5b85a3d1fb570fc425","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-11-17 10:04:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"035706dd54900985b1552734689ab81823a20921","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-11-27 12:35:57.000000000","message":"Uploaded patch set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"f01868d784fcdb9c6028ad3e92fd1aac8351bf41","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 12:43:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e26cd7b3c4244b33a081fe276624371b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/987acaddea7d4a3594dddaafb27fb6ec : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7cca97801a8c4deaab0058a4429e0f08 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0221e35d882444f98d50443d5ec4b9b6 : SUCCESS in 5m 48s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6d8b4c8fd7db5d136e5fb23c21633063bf425965","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 14:52:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f35f6de7f0da47088ab06b3e50041c72\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07075b2054f84bbebe7cf0d6d6c68d41 : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d144496abdee4319ac83989990b25a4d : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/150922719bed44c0a432100438f5fd20 : SUCCESS in 5m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/22eb0811ebd34f909b722708ff65aab5 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bf0c537358947b8a418e753d4a9a9f8 : SUCCESS in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e44380717d8b4b6fa327566f0e0c2d9c : SUCCESS in 4m 47s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/75942c8ee2774e51bd50161a22a73c47 : SUCCESS in 1h 59m 48s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/d0d47d5b1a53420bbf84643bb220fc37 : SUCCESS in 2h 06m 14s\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/72907539057f4e70909c299e756a24a1 : SUCCESS in 1h 57m 01s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/1f08164bd4b4431fb1fc217cdd905608 : SUCCESS in 1h 36m 02s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/4b8139386c9c4aefa12ef49e94a91d10 : SUCCESS in 2h 13m 54s\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/d6cacd874aa141f3909d4f518c78c6f9 : SUCCESS in 1h 37m 19s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/c02cde2b88794c0a9cec8084b2b720e8 : SUCCESS in 1h 52m 26s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/8b94c0ee702a4daf8b390012bf48af39 : SUCCESS in 1h 30m 02s\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/5df581db9c9e417c9a8de4679b9a4fcb : SUCCESS in 1h 34m 09s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/156c5aca508a4a1c8e4f5d654d25aee2 : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":4},{"id":"d81594cc0a7ee885465d8b9c00fa6ef858255edc","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-11-29 13:16:52.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"996a9674476e877387661d9051f1be22c112a760","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 13:17:29.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"211f9bf301294d4526785448daad0f30dcb5d50a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 15:53:43.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/34cbbd0433fa41029b7aa047d81735b7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e3be77858cb41fa867cd91df12e318f : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e52d14260c346ef8a762e8305a77ac1 : SUCCESS in 6m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/752c2a72aa2d409a9a5dfa9359c7dcff : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb9bc3673c0346248ccb6774a3bca0ee : SUCCESS in 5m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64839be6780d4ebfab78963e7f8fab3e : SUCCESS in 5m 15s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/fe576f339d9242b98ed1f598e8204a11 : SUCCESS in 1h 55m 01s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/dc7ee7327a04451f93f105a8b48a9a58 : SUCCESS in 1h 52m 11s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/05fb994e34b940de8b4f2ad6e3bee898 : FAILURE in 2h 34m 51s","accounts_in_message":[],"_revision_number":4},{"id":"2f17b4a2184d6df6f7b1bb09bf06a97453bedb10","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-11-30 01:09:12.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"267b3ebc0212145f6be10d708b52006f5b6e3223","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 01:28:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5512b29b73f74306aa591015bdb6bbe4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ba2e8d7920ee421b909b2d4b609e7fc2 : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/873527dc2dc64bbaa5cd9b8ecd596a76 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cba15ebffa834130b8450735c9bcf4f9 : SUCCESS in 16m 50s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9021b1d2d504c87474b492173682b85a2fa8679e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 03:33:13.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e907c4f90c90415f9cd9a037548c36ec\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/98f27d3b206c44b7b94433c6a9a05e1e : SUCCESS in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fc227839b2447c8955b13455fe522e4 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7892f9162ee847eda27fd0440c687d2b : SUCCESS in 6m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e31a8dad4f0b456db06fce59444889fa : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b31bf8ea9d5649a8853590ec75bbfa3b : SUCCESS in 5m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d4ce635659b4c65b694dc09620123e6 : SUCCESS in 4m 49s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/6b79fecba4cd4592a435fecfcb109fee : SUCCESS in 1h 57m 43s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/472046f23dc34076a816cec632a1ee13 : FAILURE in 1h 58m 50s\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/1dd346a064a04d889acdbfe1f099256d : SUCCESS in 1h 56m 17s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/bbea3556df1e4a0e82f32ddde0486cfe : FAILURE in 2h 06m 34s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/de9c57f6609a4b6a851283e6761a3393 : SUCCESS in 2h 17m 07s\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/3845525c6738429185a61acd1939fed5 : SUCCESS in 1h 23m 31s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/209c9040f14249deb5b843926dd6888b : SUCCESS in 1h 16m 41s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/843e22df35a24adb9254407835a6cba1 : SUCCESS in 1h 32m 28s\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/13b37a1d08b6402183c55cd28ca4d3c4 : SUCCESS in 1h 16m 21s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0dbec6808c92401f9fcf920e17f6c066 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":4},{"id":"8edaebc866bac0c63649352ab7d81e2afde7b561","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-04 08:53:22.000000000","message":"Uploaded patch set 5: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":5},{"id":"ffbb808f76bae1c13c728ba73ebc37777b3dbe52","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 09:02:11.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abd6da3d1a1143bfa5e4caa88a5003e4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0b07055b5f424ce6b35f354a9b4c24cd : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3b6644237af241e6bbee2af21d4506dd : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d9b739121fd74429a65445898d55d4e4 : SUCCESS in 5m 59s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6bb2b5322a194424ea89e1649f07d1d4ed042bf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 10:51:37.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/420cbfeaa71445368928060d22673ef9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba7a72c8dee1414999469c755d038d91 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9931bd64ee0c445e90a282b670f9ea4f : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f72046cc383d4a948ccb30a0214d0ea1 : SUCCESS in 6m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/814f3b57873a464c9bd0ced7f5f1ed97 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb7e3d37f5e84fa58967b8a06e2c9070 : SUCCESS in 7m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e85d1068e394af2a164b82e77353c4c : SUCCESS in 5m 34s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/a3c638aeb84744879ba4e063eae55837 : SUCCESS in 1h 45m 32s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/e776a86ecdd34f36a84bc666d9487bc4 : FAILURE in 1h 40m 26s\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/fd5efa8f96f74192904e27e722c65c07 : SUCCESS in 1h 49m 08s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/ba54474eaf5a4ad1af03725b5bb3a652 : SUCCESS in 1h 52m 41s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/686cf74ed8194f4eb4c15e2a621656c6 : SUCCESS in 1h 33m 22s\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/9faaed7f35f04cd79d0e87e6443cc8b6 : SUCCESS in 1h 14m 05s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/be71de00d6444f058f6e854732b53976 : SUCCESS in 1h 26m 01s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/70a9f28ebced44cf8af1846279c707cf : SUCCESS in 1h 57m 09s\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/634a4a9848184330831b160ace614bb2 : SUCCESS in 1h 42m 38s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6a0dcd9dfb134eccae213140d96dde38 : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":5},{"id":"2478fc6304b62d428aeecaf347c4162086fd1b3f","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-06 03:35:57.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9d5d8b135df0c066c8f898db7046a41df42bebcf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 03:51:22.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26b45778d62b4169a4bc6fdf83f3b6b3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5e038d62b4d447d6b8708328ef949baa : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/21b2bab379354693a3381607e12344f5 : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ca952f866a2b4ae898ca1e1d98034486 : SUCCESS in 12m 26s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"eec482bb5fee40dfd80358f27a4d53d3c0db37eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 05:33:12.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bf2f6dd86844d78b1db982baba619ad\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a08d5a632b1c44cb8b88ec308cb3cfa7 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0b1210d5ec54d33b66290ad8846f402 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d947ed5b6fe94e8cb395c9f3c67fb1d1 : SUCCESS in 6m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/70e5a1f6a86a4bdda3931ebd16277192 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/475804b1a8b342a696e8ebf6b4e55944 : SUCCESS in 5m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b0194add026400882cff0c8ee94b95d : SUCCESS in 5m 44s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/aa54d2c0c75f49a08a0155ed2b35b795 : SUCCESS in 1h 20m 01s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/f80af3a86caa4607a73bcb2659435e94 : SUCCESS in 1h 55m 54s\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/9e03c558fbef4bbe8406d6df6cc8b270 : SUCCESS in 1h 24m 56s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/2a331f20ee7d49e98548a9b45da050ae : SUCCESS in 1h 12m 00s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/17ffedcf28b54c50aaa3b2333965e584 : SUCCESS in 1h 36m 39s\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/a46d78b7db704652863a6e1c02c91583 : SUCCESS in 1h 14m 13s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/58668fe466704cc0bd9153090662a09a : SUCCESS in 1h 11m 31s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/4f4e206ffa764097ae0dafa7280e3507 : SUCCESS in 1h 40m 54s\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/ed480376c26d48a4a47feb252e75a030 : SUCCESS in 1h 12m 56s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c0e4471f072348718fac36795faebc66 : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":5},{"id":"131d4eae2654985be087561a24a940f96edda07f","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-06 07:15:18.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"8663eb1340c84d648915ee7e32a41f2de17c5dd4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 07:15:58.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c21e2f3375a5f51f815b7812be6f13bf37434ece","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 09:03:27.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7aedc50787e4402b8ee102c1f3c021c3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f388f6d420da437396a395f1b05500b1 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24fdf5a7814548f7ade5c5141f72d7cd : SUCCESS in 7m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/59d45bb6d9cd47c19f5495dc9ad92fce : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/619274e639e045019a8c243e9468bc26 : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73451c8199254386a0e0dd623dbc193c : SUCCESS in 5m 17s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/534333a4437a4cbeb8118917e841eddc : SUCCESS in 1h 23m 19s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/8c042475612a4cf7a50dea0338e12cef : FAILURE in 1h 26m 41s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/50c622d6a9de47109890d9161c04a7e1 : SUCCESS in 1h 40m 08s","accounts_in_message":[],"_revision_number":5},{"id":"1adbae2e8b4fb7fc2b4e9edc6f353fb55a0afcd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-07 08:36:55.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":6},{"id":"265e39c568303fc578a548810fba3caea0900bd9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 08:52:57.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e119f9d9cc04d2fa8d63df3c75af2fe\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1c822bfb63a9495c8288d42744f4f0f1 : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b0b1b78643194c13aa1da2b0b51e43a9 : SUCCESS in 12m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ff728ecc0cc440fd9e9c880c56343c08 : SUCCESS in 5m 28s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6c79608e00941b4a262af2759b1e0e2bdede2ce0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 10:38:03.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6ac84486c44421eb85b731738d66a24\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52d194226708456a82bf5080b5bd2d99 : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5cd7595e0a8443394c6f9f152249b2a : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5412be470aa34f4eaa3d23d951700290 : SUCCESS in 5m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c91a33190ea549c0ad21f2756456c586 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78c1c4e8059144198a91b2787b51a02f : SUCCESS in 6m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/589dbdbc6a7e4a21873fef8f5402b4dc : SUCCESS in 5m 27s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/058fae70f8a34b4fb9622ed2419a2ce0 : SUCCESS in 1h 59m 15s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/f32bdcfd8a924c9abcaba051861afeaf : SUCCESS in 1h 40m 37s\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/935793d4d9264f9f8da6923d71ca7d8f : SUCCESS in 1h 23m 54s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/3268ac0405444100b37bf11e69c08a3c : SUCCESS in 1h 12m 21s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/8c5f675a5a7d4f1faf5ea6c5cf79c5a3 : SUCCESS in 1h 40m 18s\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/6150771d029441b39b87a5f038a59b7f : SUCCESS in 1h 15m 46s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/6e66383c83df4fcd8c7bfcecf7e72241 : SUCCESS in 1h 11m 27s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/a6d1932bf4c64bf4b078750cd38939d5 : SUCCESS in 1h 27m 01s\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/77f037969c7247f09184325d25ea1ced : SUCCESS in 1h 16m 26s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b1b27d1add6e406d8d11124e1b233f9b : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":6},{"id":"e1a66f5d527ce7c58bb30aca6ebf99f31222bff8","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-08 03:36:28.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"e81dc5bed66f7565d4328e5377018d2aed930af7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 03:37:04.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"f9e85aa4c7502cf87c12ab706fc52dda5e0d3e09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 05:19:26.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a552fd1f312648b19767cdfe32a2e386\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d22c884d0ab411b8d3be6eddf59c956 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/235eb8918dd8494e8470d8ce95e4304c : SUCCESS in 6m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f60b93906da7422a93de81dc7cb45688 : SUCCESS in 7m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28888afb5a534fd4abe7ee214e729740 : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0cb539c3a9546daaf6e28291f64b73f : SUCCESS in 5m 30s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/6eb6d68f458c4182be654c18e54fde33 : SUCCESS in 1h 32m 57s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/0a2125462867490582cfbfc05f5014c5 : SUCCESS in 1h 40m 55s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/c38c250240f04ca3b7b92f23db68fa63 : SUCCESS in 1h 39m 56s","accounts_in_message":[],"_revision_number":6},{"id":"5a7656c2c119bb8a9c40887ae1c43310084e01e4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 05:19:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"4d4ed7c3658817bf28a4af8ca6c1a3eefe68c54a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 05:24:03.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/002b96a354fb415caa140c0a3a4da5d8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9741d6fcfd004856a8ca07cff0497105 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7ccd1b00ef449c4b88aa32ce1b67d1d : SUCCESS in 51s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"895e59237efb17f8d46763e58210dae96f772d7c","revisions":{"b4e365be6af60452a32127355f27baf5c7be78e0":{"kind":"REWORK","_number":1,"created":"2023-11-16 14:55:44.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/69/901169/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/69/901169/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/69/901169/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/69/901169/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/69/901169/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/69/901169/1"}}},"commit":{"parents":[{"commit":"48439619b5fbce3ac70be5ae092c88fae072cde9","subject":"[CI]Add postgresql jobs in zuul","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/48439619b5fbce3ac70be5ae092c88fae072cde9"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-11-16 14:50:28.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-11-16 14:50:28.000000000","tz":480},"subject":"Fix swift can\u0027t work over TLS","message":"Fix swift can\u0027t work over TLS\n\nThis commit adds insecure\u003dTrue parameter to swiftclient.\n\nStory: #2010674\nTask: #2010674\nChange-Id: I6156b289659dc9130effb1666c3cd2e14366bffe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b4e365be6af60452a32127355f27baf5c7be78e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b4e365be6af60452a32127355f27baf5c7be78e0"}]},"branch":"refs/heads/master"},"db78582392fdccb28c7f604d240261b529e085b6":{"kind":"REWORK","_number":2,"created":"2023-11-16 14:58:13.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/69/901169/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/69/901169/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/69/901169/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/69/901169/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/69/901169/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/69/901169/2"}}},"commit":{"parents":[{"commit":"48439619b5fbce3ac70be5ae092c88fae072cde9","subject":"[CI]Add postgresql jobs in zuul","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/48439619b5fbce3ac70be5ae092c88fae072cde9"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-11-16 14:50:28.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-11-16 14:58:02.000000000","tz":480},"subject":"Fix swift can\u0027t work over TLS","message":"Fix swift can\u0027t work over TLS\n\nThis commit adds insecure\u003dTrue parameter to swiftclient.\n\nStory: #2010674\nTask: #2010674\nChange-Id: I6156b289659dc9130effb1666c3cd2e14366bffe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/db78582392fdccb28c7f604d240261b529e085b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/db78582392fdccb28c7f604d240261b529e085b6"}]},"branch":"refs/heads/master"},"4726c7f7aaf9610e316dfe82e5f4c5fd8a46850b":{"kind":"REWORK","_number":3,"created":"2023-11-17 01:10:59.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/69/901169/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/69/901169/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/69/901169/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/69/901169/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/69/901169/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/69/901169/3"}}},"commit":{"parents":[{"commit":"48439619b5fbce3ac70be5ae092c88fae072cde9","subject":"[CI]Add postgresql jobs in zuul","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/48439619b5fbce3ac70be5ae092c88fae072cde9"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-11-16 14:50:28.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-11-17 01:10:44.000000000","tz":480},"subject":"Fix swift can\u0027t work over TLS","message":"Fix swift can\u0027t work over TLS\n\nThis commit adds insecure\u003dTrue parameter to swiftclient.\n\nStory: #2010674\nTask: #2010674\nChange-Id: I6156b289659dc9130effb1666c3cd2e14366bffe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4726c7f7aaf9610e316dfe82e5f4c5fd8a46850b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4726c7f7aaf9610e316dfe82e5f4c5fd8a46850b"}]},"branch":"refs/heads/master"},"1b39060f271516653ec182b570cca041faf4b831":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-11-27 12:35:57.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/69/901169/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/69/901169/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/69/901169/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/69/901169/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/69/901169/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/69/901169/4"}}},"commit":{"parents":[{"commit":"48439619b5fbce3ac70be5ae092c88fae072cde9","subject":"[CI]Add postgresql jobs in zuul","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/48439619b5fbce3ac70be5ae092c88fae072cde9"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-11-16 14:50:28.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-11-27 12:34:58.000000000","tz":480},"subject":"Fix guest agent failed to use swiftclient over TLS","message":"Fix guest agent failed to use swiftclient over TLS\n\nThis commit adds insecure\u003dTrue parameter to swiftclient.\n\nStory: #2010674\nTask: #2010674\nChange-Id: I6156b289659dc9130effb1666c3cd2e14366bffe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1b39060f271516653ec182b570cca041faf4b831"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1b39060f271516653ec182b570cca041faf4b831"}]},"branch":"refs/heads/master"},"89e995ccf3050265f794119446c683a59553f16a":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2023-12-04 08:53:22.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/69/901169/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/69/901169/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/69/901169/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/69/901169/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/69/901169/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/69/901169/5"}}},"commit":{"parents":[{"commit":"48439619b5fbce3ac70be5ae092c88fae072cde9","subject":"[CI]Add postgresql jobs in zuul","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/48439619b5fbce3ac70be5ae092c88fae072cde9"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-11-16 14:50:28.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-04 08:52:43.000000000","tz":480},"subject":"Fix guest agent failed to use swiftclient over TLS","message":"Fix guest agent failed to use swiftclient over TLS\n\nThis commit adds insecure\u003dTrue parameter to swiftclient.\n\nStory: #2010674\nTask: #47748\nChange-Id: I6156b289659dc9130effb1666c3cd2e14366bffe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/89e995ccf3050265f794119446c683a59553f16a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/89e995ccf3050265f794119446c683a59553f16a"}]},"branch":"refs/heads/master"},"895e59237efb17f8d46763e58210dae96f772d7c":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-12-07 08:36:55.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/69/901169/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/69/901169/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/69/901169/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/69/901169/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/69/901169/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/69/901169/6"}}},"commit":{"parents":[{"commit":"aacc4a6917c86296e90dc3cc2b2b19acb7f52fb9","subject":"Merge \"Remove unnecessary ceilometer service overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/aacc4a6917c86296e90dc3cc2b2b19acb7f52fb9"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-11-16 14:50:28.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-12-07 08:36:55.000000000","tz":0},"subject":"Fix guest agent failed to use swiftclient over TLS","message":"Fix guest agent failed to use swiftclient over TLS\n\nThis commit adds insecure\u003dTrue parameter to swiftclient.\n\nStory: #2010674\nTask: #47748\nChange-Id: I6156b289659dc9130effb1666c3cd2e14366bffe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/895e59237efb17f8d46763e58210dae96f772d7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/895e59237efb17f8d46763e58210dae96f772d7c"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
