)]}'
{"id":"openstack%2Ftrove~901568","triplet_id":"openstack%2Ftrove~stable%2F2023.2~I9bd497c85c6eba67485840d07a80cca8eed27994","project":"openstack/trove","branch":"stable/2023.2","topic":"add-region-name-stable/2023.2","hashtags":[],"change_id":"I9bd497c85c6eba67485840d07a80cca8eed27994","subject":"Add region name to trove-guestagent","status":"MERGED","created":"2023-11-22 12:23:14.000000000","updated":"2023-12-12 05:29:33.000000000","submitted":"2023-12-12 05:28:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901568-add-region-name-stable/2023.2","cherry_pick_of_change":899434,"cherry_pick_of_patch_set":2,"meta_rev_id":"a7feef50c2dd72a5bef4b46973d23df4dfc659ea","_number":901568,"virtual_id_number":901568,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-12 05:28:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-12-12 02:19:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-12-12 02:19:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9426,"name":"Gabriel Adrian Samfira","email":"gsamfira@cloudbasesolutions.com","username":"gsamfira"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-22 12:23:14.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":9426,"name":"Gabriel Adrian Samfira","email":"gsamfira@cloudbasesolutions.com","username":"gsamfira"},"state":"CC"},{"updated":"2023-11-22 12:30:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-22 15:28:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4f67692f4f12343ef906a3b0c1fdb3c65898d28f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-11-22 12:23:14.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0dc8e3ebdf5d234294dd11362a16362305a2263b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 12:30:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/276e709f28ab4d4288b45e15b38f2e39\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/306f6cdc581f4a3b9f9e31fedc41dd13 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f2e16f1925bb493a97905a055d8bd5d8 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/205aec67bda04298b67329236642ca62 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c4cbc0f075b9404594b550461f94fb90 : RETRY_LIMIT in 1m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ffda8934a8472c188c9fd41d10623822ef35410d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 15:28:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33af239bd29c45f893b8c404296b7457\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1598ba04272b4298bd8e9e1f293a158c : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64ff2b36b0014c628b122e10b5546363 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d43918b513024ffbbeb87285529e81b9 : SUCCESS in 6m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65ed50a2630b4dc6ad7f3aab40befdb3 : SUCCESS in 5m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0813440533c24b2e8348b750eb8b0a4c : SUCCESS in 6m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/47ffebea1f1d4bd186bcaa689157b8a7 : RETRY_LIMIT in 1m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc141d6c55094f3691c829ddd2ac842b : SUCCESS in 5m 30s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c9f3ff50a8914edcb8e6f38ed0b98ae2 : SUCCESS in 5m 17s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/4b8aa21c950441429bfb8684263d80ea : TIMED_OUT in 3h 02m 48s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/1d2b9d7061194176b285f7abed49dd86 : SUCCESS in 2h 42m 52s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/da32c995d87946d7aee24093848a3844 : FAILURE in 2h 49m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2729ee31b42b7a4e0a1618bf15057941b27f0c4b","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-12 02:19:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3d485fec48173c6d069fc795f72b62bc76e3ee22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 02:20:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b99768f8c4bf49ad872e3c0b95f29092434e2d6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 05:28:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2a15ad01cae4f138797a97754f20bfe\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f7515327a1f43e7bbd41a4fde644f38 : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2f091adc6c348738b30067d65082aa8 : SUCCESS in 7m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/83df2427b7df416aa0f8ce487079b88a : SUCCESS in 6m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9798a835311447c3a7a243dc3e15de2d : SUCCESS in 7m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b741d81c36244d9bef4715eb49cbbea : SUCCESS in 6m 57s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/c884467ada63453b8eb1cf07c3985d33 : SUCCESS in 2h 59m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1e959cfbff96cc300ee7a3dbf841a643aebcc712","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 05:28:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a7feef50c2dd72a5bef4b46973d23df4dfc659ea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 05:29:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9eb81385741a49cebd9251ddf41a314a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6b22d97eb0d402088de99b582d97b0b : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"505a9834e5a50823f67776e1f2f0f5307a8c6013","revisions":{"505a9834e5a50823f67776e1f2f0f5307a8c6013":{"kind":"REWORK","_number":1,"created":"2023-11-22 12:23:14.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/68/901568/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/901568/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/901568/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/901568/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/901568/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/901568/1"}}},"commit":{"parents":[{"commit":"446e40e4d1e37e1fed52d3f2f5d2f5877975b3f1","subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/446e40e4d1e37e1fed52d3f2f5d2f5877975b3f1"}]}],"author":{"name":"Gabriel Adrian Samfira","email":"gsamfira@cloudbasesolutions.com","date":"2023-10-26 17:26:34.000000000","tz":180},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-11-22 12:23:14.000000000","tz":0},"subject":"Add region name to trove-guestagent","message":"Add region name to trove-guestagent\n\nIn a shared keystone multi-region setup, we must set the region_name\nwhen using any of the OpenStack clients. Without the region name,\nclients may attempt to use the API endpoints in the wrong region.\nBackups currently fail if there are multiple regions created in the\nsame keystone catalog, due to the fact that the trove guest agent\nattempts to connect to the S3 endpoint in another region.\n\nStory: 2010950\nTask: 48994\n\nChange-Id: I9bd497c85c6eba67485840d07a80cca8eed27994\nSigned-off-by: Gabriel Adrian Samfira \u003cgsamfira@cloudbasesolutions.com\u003e\n(cherry picked from commit e998b6886602575127ebe613e56cee3a5a01c6c6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/505a9834e5a50823f67776e1f2f0f5307a8c6013"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/505a9834e5a50823f67776e1f2f0f5307a8c6013"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
