)]}'
{"id":"openstack%2Ftrove~901569","triplet_id":"openstack%2Ftrove~stable%2F2023.1~I9bd497c85c6eba67485840d07a80cca8eed27994","project":"openstack/trove","branch":"stable/2023.1","topic":"add-region-name-stable/2023.1","hashtags":[],"change_id":"I9bd497c85c6eba67485840d07a80cca8eed27994","subject":"Add region name to trove-guestagent","status":"MERGED","created":"2023-11-22 12:24:14.000000000","updated":"2023-12-12 05:39:31.000000000","submitted":"2023-12-12 05:38:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901569-add-region-name-stable/2023.1","cherry_pick_of_change":899434,"cherry_pick_of_patch_set":2,"meta_rev_id":"6f6715dcc1fb3c8b62954e19f8ce9fa22fe16040","_number":901569,"virtual_id_number":901569,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-12 05:38:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-12-12 02:19:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-12-12 02:19:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9426,"name":"Gabriel Adrian Samfira","email":"gsamfira@cloudbasesolutions.com","username":"gsamfira"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-22 12:24:14.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":9426,"name":"Gabriel Adrian Samfira","email":"gsamfira@cloudbasesolutions.com","username":"gsamfira"},"state":"CC"},{"updated":"2023-11-22 12:30:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-22 14:57:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"68c17d7103664cd5e6c272f7c80fc34779b2ca14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-11-22 12:24:14.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"bc1d4f7f6007eb000b0e70cbd7ecff59e617d1ba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 12:30:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de4b348ed8304586b536b1fcee51d826\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/60fcb5bb834a4bad85cdea52152d3f1d : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1511e477cde3449da5b4517f14fe0eb1 : SUCCESS in 5m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d4ef2e498d54ad8ef457eeaf90b8cb59f772a1f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 14:57:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f8de3805bb6462daf9a9797e8eced73\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/296da5956495448da03b0b36ed133f13 : FAILURE in 6m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c22d96c692ec4a42a3cad2af31aff0e3 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f38f700923464340904cf006ce9623a9 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b863145ce7b4c1d9fe2bc8d14379929 : SUCCESS in 5m 38s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ea123eda50d1454a93016e1e44269f8d : SUCCESS in 5m 06s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/5a2e7f48ad21452eada626795badc6ca : SUCCESS in 2h 32m 12s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/fd165e2f2a094e85bf1c9c28f7fd9272 : FAILURE in 2h 20m 26s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d973a478160649c0ba65f36eed6a08a1 : TIMED_OUT in 2h 03m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"adba9d21182b1efe58d597e4309cf5eeb257c508","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-12 02:19:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3bc94b540ee1b30b2387e0114ea8d2965b1d22be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 02:20:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b5e2c8541aa590462d3c2542f97749dd6a42b6e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 05:38:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cc36080059e44b7b686634a7fc0f10e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ad85db979734cf8a8c64336ebf6facf : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e26132d54f541f099a0c73249fe88c1 : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08e4d8ffcfd44aafa9a0a46c505ccfc5 : SUCCESS in 6m 06s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/d341b8c7fd2c4e398d9e40c96bdfa247 : TIMED_OUT in 3h 03m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f2ac0ee026d7f8ddbaa121fe7660f10d642efa58","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 05:38:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6f6715dcc1fb3c8b62954e19f8ce9fa22fe16040","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 05:39:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34ad70b7bfe248fca9ee76a79343bd83\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/271dcb755d3d4e77a887bb48f1644f0b : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bb8bbbeb66779e3d811843d2246d2fae094acb7e","revisions":{"bb8bbbeb66779e3d811843d2246d2fae094acb7e":{"kind":"REWORK","_number":1,"created":"2023-11-22 12:24:14.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/69/901569/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/69/901569/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/69/901569/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/69/901569/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/69/901569/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/69/901569/1"}}},"commit":{"parents":[{"commit":"8c47360daa23ee45ac975f88d2ba93c2460136b2","subject":"Merge \"[CI]: Don\u0027t run periodic jobs on stable branchs\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8c47360daa23ee45ac975f88d2ba93c2460136b2"}]}],"author":{"name":"Gabriel Adrian Samfira","email":"gsamfira@cloudbasesolutions.com","date":"2023-10-26 17:26:34.000000000","tz":180},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-11-22 12:24:14.000000000","tz":0},"subject":"Add region name to trove-guestagent","message":"Add region name to trove-guestagent\n\nIn a shared keystone multi-region setup, we must set the region_name\nwhen using any of the OpenStack clients. Without the region name,\nclients may attempt to use the API endpoints in the wrong region.\nBackups currently fail if there are multiple regions created in the\nsame keystone catalog, due to the fact that the trove guest agent\nattempts to connect to the S3 endpoint in another region.\n\nStory: 2010950\nTask: 48994\n\nChange-Id: I9bd497c85c6eba67485840d07a80cca8eed27994\nSigned-off-by: Gabriel Adrian Samfira \u003cgsamfira@cloudbasesolutions.com\u003e\n(cherry picked from commit e998b6886602575127ebe613e56cee3a5a01c6c6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bb8bbbeb66779e3d811843d2246d2fae094acb7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bb8bbbeb66779e3d811843d2246d2fae094acb7e"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
