)]}'
{"id":"openstack%2Ftrove~901571","triplet_id":"openstack%2Ftrove~stable%2Fyoga~I9bd497c85c6eba67485840d07a80cca8eed27994","project":"openstack/trove","branch":"stable/yoga","topic":"add-region-name-stable/yoga","attention_set":{},"removed_from_attention_set":{"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2023-12-12 02:37:08.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"I9bd497c85c6eba67485840d07a80cca8eed27994","subject":"Add region name to trove-guestagent","status":"MERGED","created":"2023-11-22 12:26:59.000000000","updated":"2023-12-12 05:45:10.000000000","submitted":"2023-12-12 05:44:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901571-add-region-name-stable/yoga","cherry_pick_of_change":899434,"cherry_pick_of_patch_set":2,"meta_rev_id":"ea850b191c948e2071f6fe925460206942362d66","_number":901571,"virtual_id_number":901571,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-12 05:44:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-12-12 02:37:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-12-12 02:37:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9426,"name":"Gabriel Adrian Samfira","email":"gsamfira@cloudbasesolutions.com","username":"gsamfira"},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-22 12:26:59.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":9426,"name":"Gabriel Adrian Samfira","email":"gsamfira@cloudbasesolutions.com","username":"gsamfira"},"state":"CC"},{"updated":"2023-11-22 16:02:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-05 08:26:34.000000000","updated_by":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"CC"}],"messages":[{"id":"49ef24915e33f40ef870cb747351808c2cabbf50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-11-22 12:26:59.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"544f4e6358547e51a5a1ae967fc24e5889f711ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 16:02:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9af10076ef564b1ba0180af95b43a8d3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/313e249852e3451587f01c0982dc649b : FAILURE in 6m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/765ceddff1c848f3a96ed3bf2b12f335 : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d5e9bd657a0471497018fd93cea643d : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d382a981206d428fad46ffd077465f27 : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d4b49e96e9245e8a075438f6358aaee : SUCCESS in 6m 22s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/43a074440fd2456aaf35c6aac014e6d2 : SUCCESS in 6m 14s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/3e793529fb744152a967a6665fe14de0 : TIMED_OUT in 3h 03m 24s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/e4dbfc6d3cba4130bb7fbf98ee571e98 : TIMED_OUT in 3h 02m 20s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c37315ea31bc4bf69db4cffc2cb637a5 : FAILURE in 43m 13s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/ea586aec7da8448cade1798194402768 : FAILURE in 33m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6e8913fded97bc1612a25ab76d3187032244b5f8","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-12-05 08:26:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"474609cd0095a3c687934457f06c0bfa0317aac8","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-12 02:37:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"732452e959819b145de72932c60f888446289de4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 02:37:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8e843b573fd5a1aadbab3e81f93d547c51c53d66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 05:44:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5a75f71cd9a4df8b9d5bb008e5fd3e8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0830f8bb7d9e431bb6675093146c0922 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28296d6ffab943629a4b17c53c4fe404 : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/17b8e8e5eafe4f3582f2e4c993c2b0c7 : SUCCESS in 7m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fe585dd1e9f4df3a8452cad143514dd : SUCCESS in 7m 45s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/88b402db0061496da3ce56a760b96710 : TIMED_OUT in 3h 02m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"76be7e4fd34a4f884baf05cdaeb3ab83f20f3feb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 05:44:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ea850b191c948e2071f6fe925460206942362d66","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 05:45:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b12c1047372847d6bf43542854174d2a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33090448acd9471e8efd93f86d1d6860 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cc0e021f758617dd0a41dea2addae3bc02e6f002","revisions":{"cc0e021f758617dd0a41dea2addae3bc02e6f002":{"kind":"REWORK","_number":1,"created":"2023-11-22 12:26:59.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/71/901571/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/71/901571/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/901571/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/901571/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/901571/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/71/901571/1"}}},"commit":{"parents":[{"commit":"f07e85ded52abe31eaad6340d07f0d16c0a40c33","subject":"Fix \"create database\" failed for postgress","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f07e85ded52abe31eaad6340d07f0d16c0a40c33"}]}],"author":{"name":"Gabriel Adrian Samfira","email":"gsamfira@cloudbasesolutions.com","date":"2023-10-26 17:26:34.000000000","tz":180},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-11-22 12:26:59.000000000","tz":0},"subject":"Add region name to trove-guestagent","message":"Add region name to trove-guestagent\n\nIn a shared keystone multi-region setup, we must set the region_name\nwhen using any of the OpenStack clients. Without the region name,\nclients may attempt to use the API endpoints in the wrong region.\nBackups currently fail if there are multiple regions created in the\nsame keystone catalog, due to the fact that the trove guest agent\nattempts to connect to the S3 endpoint in another region.\n\nStory: 2010950\nTask: 48994\n\nChange-Id: I9bd497c85c6eba67485840d07a80cca8eed27994\nSigned-off-by: Gabriel Adrian Samfira \u003cgsamfira@cloudbasesolutions.com\u003e\n(cherry picked from commit e998b6886602575127ebe613e56cee3a5a01c6c6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cc0e021f758617dd0a41dea2addae3bc02e6f002"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cc0e021f758617dd0a41dea2addae3bc02e6f002"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
