)]}'
{"id":"openstack%2Ftrove~902900","triplet_id":"openstack%2Ftrove~master~Ief5a87724ccaed2f0c63cf32672a7437f384179f","project":"openstack/trove","branch":"master","topic":"ipv6-error","attention_set":{},"removed_from_attention_set":{"28691":{"account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"last_update":"2023-12-12 05:44:20.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2023-12-12 03:38:14.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"Ief5a87724ccaed2f0c63cf32672a7437f384179f","subject":"Revert \"Fix docker daemon failed to set IPv6 gateway\"","status":"MERGED","created":"2023-12-08 06:42:56.000000000","updated":"2023-12-12 05:45:35.000000000","submitted":"2023-12-12 05:44:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"revert_of":902557,"submission_id":"902900-ipv6-error","meta_rev_id":"695fb5d3ad60c206b3e21e0df3e3de6d991bd932","_number":902900,"virtual_id_number":902900,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-12 05:44:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-12-12 03:38:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-12-12 03:38:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-08 06:42:56.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"REVIEWER"},{"updated":"2023-12-08 06:42:56.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5d2c80c726727bff7741ba8d60a3c984f15e3897","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-08 06:42:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"44ae545c6c1fd9a63d7cd9f54137788701d7552b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 06:49:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/032e5733a05547dba06472c6c6082410\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b2d089ac2a9f41a2aab38be298859297 : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f0178c95158b4b80861ffa9e26293b08 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1d062c34d7cb49a3b9bb988e161d5835 : SUCCESS in 5m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a129fd38b36deea5c6a375f8342610313b577503","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 09:19:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/df2386ce8f5e4fd18f3218e156fcfc58\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/096e81d0301b4d3cb9e869965708d0bc : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b3ed14b254c496591681ba0859cafac : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc3356df371b4b5d98fb4ae7b0125d36 : SUCCESS in 6m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/40d31498a81c4d5b91c39b7da36deb06 : SUCCESS in 7m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa5562f4bcde42318f07ae7d2c0dd87f : SUCCESS in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bcef3aab55bb4e3a8cbc2d041b51ea6f : SUCCESS in 7m 12s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/1c97ec3fa7b4428582a61856345da7c5 : FAILURE in 2h 34m 23s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/76b04a320d9943de9b18762b4585edbe : FAILURE in 1h 29m 02s\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/60467f2bd6bc41c9a0297bd3002ea119 : FAILURE in 1h 17m 27s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/c3b41b67911a4197bc84b394127150c1 : SUCCESS in 1h 55m 30s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/17312311fe934e63ae42d407dd4b3832 : SUCCESS in 2h 17m 57s\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b170a60bfff24dd38b2baff7c3fb3fc3 : SUCCESS in 1h 55m 16s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/7cb7bf0781b54c0b89f49675ebab5451 : FAILURE in 2h 32m 51s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/a57e137b3ec1433fa5e7cff5441bc645 : SUCCESS in 2h 13m 05s\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/4725989d59d44ba6af3275884ef0e073 : SUCCESS in 1h 48m 51s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/797c16e2e7e24e22a4e847266ac43d24 : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":1},{"id":"52d787a2d9f4c76035f76c16743fb9ad29b72c54","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-11 03:24:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"360a63f48d6d64b495b36017a44b78bc213b02bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-11 03:37:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e83098c34bd643d1b8cc83a0987fc16a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/13c5abaf87864fd1b3450c79b6ca7889 : SUCCESS in 8m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/eb8eefc7d30c43dc8024bd87fbb1ae9b : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/105a9e865bd34521baf0e8f162a4f647 : SUCCESS in 10m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"51862be501cbb41172121617ed93776a8fee4459","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-11 05:50:25.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c218277a67b4f6aa03e714d4d2de51a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/edb9ea65839d49319962211c0ed329b6 : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8737192ec484dc4ab889f5ba41147d9 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/643d870e15944813aef57529105e964d : SUCCESS in 6m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/94ba39c6ba9448b88a6024b48d228763 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/654a1e14746d4a249b08888b6a722669 : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06fbad592eea4bf4a097fceb479d811a : SUCCESS in 6m 35s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/27a51076a8b24e61a9783a5b6f408175 : SUCCESS in 2h 01m 22s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/1fdc7b7911bd4d179c7429635cb62150 : SUCCESS in 2h 02m 58s\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/66627dbc6fda4f1f9ba3ec8cafd71985 : SUCCESS in 1h 50m 10s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/bd8aa2a7f6be4e0c9d3b6ced49e109c2 : SUCCESS in 1h 41m 40s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/988a4902c323414fb9cbb9a934134cd4 : SUCCESS in 2h 14m 03s\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/73b97dc87daa41579e243ae6011c80e8 : SUCCESS in 1h 42m 24s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/6929225fa96142fd8ef9a9c272778e5f : FAILURE in 2h 23m 19s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/124dc772d5764814a97ac6d2f7631f87 : SUCCESS in 1h 47m 13s\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/a79577175eaf4aa990e6f1b4fafc8109 : SUCCESS in 1h 15m 05s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/068312f7681f4345826081e64d0154ee : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":1},{"id":"05bec2539e45fa266becc3ae260f1a9daca9a576","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-12 01:38:54.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"013af0f23453574ed305e2152144f233d876c5ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 01:45:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d826924c50ae4dfc894cf75805b6affd\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/78f1bd489b494f00beaa41c61691a867 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/28be60a629cb4fc9be2205a6f55c546a : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5de083d2a04945d7b6bf1c90f41ebe7b : SUCCESS in 5m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c3d16b5f198cc2003ee3e419537f514a060c5333","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 03:34:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f664a11c6a3a4a4fa1442d3edae70153\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2433ab0161c449a8852b0787e7f674be : SUCCESS in 7m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0e70a69b7c14f3e9bf08e9c6577d173 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f7ca6d0c1234f52b8e9f3fd617c3878 : SUCCESS in 7m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7b520b23d44f448cb22adca8f65c87d0 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89c1131dbf8448d1993c866d6c6ac5b9 : SUCCESS in 5m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c59ce02d3f0a45b594d7ab48903414fc : SUCCESS in 5m 35s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/269d61c5ca3b439bac544e3066ba091b : SUCCESS in 1h 26m 32s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/f83e896ea8b9446aa0d77196ff162424 : SUCCESS in 1h 47m 42s\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/821436b451dc47bdaaef86d7481f325f : SUCCESS in 1h 43m 25s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b48dc4ca2f81434784e041c261177de4 : SUCCESS in 1h 18m 14s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/d813dee3227146b497212fcacedd1693 : SUCCESS in 1h 54m 22s\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/638ac466fcdb40329d364d480dff5e6b : SUCCESS in 1h 22m 18s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/b9994db07d814b0fa7c56fe5653b74c5 : SUCCESS in 1h 21m 15s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/7cfac3c5543d481eb105b720f00719b3 : SUCCESS in 1h 36m 23s\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/e2ad246fc71d414881048e603fff0c63 : SUCCESS in 1h 18m 59s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/eeb6462d8c554deca29f736d0a5206da : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":2},{"id":"03c362e52f1cef5574bc4ea53fba8e750ac78b32","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-12 03:38:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4e1449aa1b4bee1fb4114568a59865e23741374f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 03:38:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cc519010ea1172dfbfba0e072abc175e920a2f3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 05:44:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3952bd40e404e0099de3455f37b4ff4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38e37eff0e07476bacf57cbecdc5baa5 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3fcbeee22c674b49b129c810d7ae0da5 : SUCCESS in 7m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a001432389c9410184d536b0f9b98538 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d1f802138ea464181341146bb0656b6 : SUCCESS in 7m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ba6e32147b84cf1ac659c6f4f48e06a : SUCCESS in 5m 41s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/a3ccd72cd73242a88b413bd4fdf57abb : SUCCESS in 1h 57m 03s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/0b95612231114e67bf10c78af7864dea : SUCCESS in 1h 29m 00s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/8bb783c3488e42fab4a52fce93427587 : SUCCESS in 1h 44m 54s","accounts_in_message":[],"_revision_number":2},{"id":"b09e7660aea1df3fc09f8f79ccd31b4e39d864cd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 05:44:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"695fb5d3ad60c206b3e21e0df3e3de6d991bd932","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 05:45:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1bbe6f7d2dc4edfae76f7adde9260fa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adcfb34591bd451d99a5c2dbfc120378 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f91737447864236a5c8947f194693fc : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9f7b9a95e86d69fce9ada96c16677d1e1b168884","revisions":{"bd65c604de552623a78e393f439be587140f21e1":{"kind":"REWORK","_number":1,"created":"2023-12-08 06:42:56.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/00/902900/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/00/902900/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/902900/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/902900/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/902900/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/00/902900/1"}}},"commit":{"parents":[{"commit":"ae27a32a9f15e89f1f27f48fc255490c1b6dd97b","subject":"Fix docker daemon failed to set IPv6 gateway","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ae27a32a9f15e89f1f27f48fc255490c1b6dd97b"}]}],"author":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-12-08 06:42:56.000000000","tz":0},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2023-12-08 06:42:56.000000000","tz":0},"subject":"Revert \"Fix docker daemon failed to set IPv6 gateway\"","message":"Revert \"Fix docker daemon failed to set IPv6 gateway\"\n\nThis reverts commit ae27a32a9f15e89f1f27f48fc255490c1b6dd97b.\n\nReason for revert: This commit didn\u0027t fix the issue\n\nChange-Id: Ief5a87724ccaed2f0c63cf32672a7437f384179f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bd65c604de552623a78e393f439be587140f21e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bd65c604de552623a78e393f439be587140f21e1"}]},"branch":"refs/heads/master"},"9f7b9a95e86d69fce9ada96c16677d1e1b168884":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-12-12 01:38:54.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/00/902900/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/00/902900/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/902900/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/902900/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/902900/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/00/902900/2"}}},"commit":{"parents":[{"commit":"21b4cc2cf7ca3a09bdefb3654dcbb07043ea2ee6","subject":"Don\u0027t configure ip route for slaac ipv6 subnet","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/21b4cc2cf7ca3a09bdefb3654dcbb07043ea2ee6"}]}],"author":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-12-08 06:42:56.000000000","tz":0},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-12-12 01:38:54.000000000","tz":0},"subject":"Revert \"Fix docker daemon failed to set IPv6 gateway\"","message":"Revert \"Fix docker daemon failed to set IPv6 gateway\"\n\nThis reverts commit ae27a32a9f15e89f1f27f48fc255490c1b6dd97b.\n\nReason for revert: This commit didn\u0027t fix the issue\n\nChange-Id: Ief5a87724ccaed2f0c63cf32672a7437f384179f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9f7b9a95e86d69fce9ada96c16677d1e1b168884"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9f7b9a95e86d69fce9ada96c16677d1e1b168884"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
