)]}'
{"id":"openstack%2Ftrove~903448","triplet_id":"openstack%2Ftrove~stable%2Fzed~I6156b289659dc9130effb1666c3cd2e14366bffe","project":"openstack/trove","branch":"stable/zed","topic":"2023.2-stable/zed","attention_set":{},"removed_from_attention_set":{"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2023-12-14 08:47:48.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"I6156b289659dc9130effb1666c3cd2e14366bffe","subject":"Fix guest agent failed to use swiftclient over TLS","status":"MERGED","created":"2023-12-12 08:15:26.000000000","updated":"2023-12-14 11:15:05.000000000","submitted":"2023-12-14 11:11:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"903448-2023.2-stable/zed","cherry_pick_of_change":903441,"cherry_pick_of_patch_set":3,"meta_rev_id":"2becb94388dcad6e9525516b12b2722408859ad2","_number":903448,"virtual_id_number":903448,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-14 11:11:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-12-14 08:47:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-12-14 08:47:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-12 09:54:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4fc5d8476c13c1bab0775f3d77519647a78e258a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-12 08:15:26.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.1.","accounts_in_message":[],"_revision_number":1},{"id":"4307e1a8196ee9e09c16e9e7023558a8f67fcda0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 09:54:39.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b875998fde324275aebe5c26dbd6d80f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ba100d3da5247cb8b476a516480cd0b : FAILURE in 5m 37s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/484b24812cfc44f1910942aeb002aa4d : FAILURE in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/085025a7e08541ef8cce0e5429a231e9 : SUCCESS in 6m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3ac1d4e7e3b48cbb3b50de464611ab6 : SUCCESS in 6m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/175292466e3241acbabaee7afaf8979a : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ec737de2d264228ba5be3a2d0220b7b : SUCCESS in 6m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c0ededbd2c84d2ca05d56d8fe346b50 : SUCCESS in 5m 50s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fee354fecacd49838ff7637d3184c8e1 : SUCCESS in 6m 04s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/a8705efe485148c08ce8026663c2ce4d : FAILURE in 1h 37m 44s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/4220120388f24fec8b6c2b8829250ce9 : FAILURE in 1h 35m 43s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/89132013b74f4fb9b9b542383ab1b320 : FAILURE in 33m 16s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/8c8a4ab061834da296957b95234a6981 : FAILURE in 31m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b576bf2aef5fd79ad470e5585d6a7996f4fafcdb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-13 01:19:27.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"1a292d7b64df50798ed94c3637c00b7424dcf7f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 03:03:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0095afbf9de849959d29afceebe02162\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ae4b12c78a94170a83b681afa42b7e7 : FAILURE in 6m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a0f47ccb29343d5a923f28602c36d00 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7363d87d3ae46d396c5f9b372459daa : SUCCESS in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/51da995628794f32b05a7c527c649168 : SUCCESS in 8m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9423614c73214e49bf14ec39d82e20b8 : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83cc438db03a4b12ac8a1b5899437ac3 : SUCCESS in 8m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/02719af99ac14f68982ef2d25deab164 : SUCCESS in 5m 06s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c4ec631453b24fd0ab8fa629b5df7ee1 : SUCCESS in 5m 16s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/c5e8a56b863c429ea82067b0991a776a : FAILURE in 33m 25s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/b216b336f8cb476ea8d0283623bd1693 : FAILURE in 1h 42m 33s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0a1107463acc4f6c9628a5e5c2f73ed3 : FAILURE in 40m 39s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/e6f1440327b54b86b54069edfe87ba61 : FAILURE in 35m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"73f438d0fbcb5101471aff31b87182fa47335861","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-14 08:47:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"47e0b8836428043833ea76d8a68bb0b2303778d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 08:48:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"50d1e345b4a77827f9c5bd092cb439a4e3b29559","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 11:11:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fb9e187f311479ea42496c62680fb58\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0e152d23b1f4ebd81747880c92d9ad3 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83636c0d13724f3b938fc539922f0a2a : SUCCESS in 7m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/705362c96b6b4610bd6ba6ac6ee8ddf7 : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85caa339d0584a1aaa9f617636496e1e : SUCCESS in 5m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30ee97e41d944ba98b7093fd22183c64 : SUCCESS in 5m 20s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/c6fd41356d874f25bf853a81ed271611 : FAILURE in 1h 51m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cce2657ebe282b823e99df7388b619123832b147","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 11:11:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"2becb94388dcad6e9525516b12b2722408859ad2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 11:15:05.000000000","message":"Patch Set 2:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3169346c66644954821515178a7c3406\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cbf915b1b1c4effa11d944638c0cdd6 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f27101b248f846d28edd9713112f6bca : FAILURE in 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c8579f514357f83b7b786f0389780cebdb453224","revisions":{"2608fc289c536c6138a179c831ee35a8cd1d6ee3":{"kind":"REWORK","_number":1,"created":"2023-12-12 08:15:26.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/48/903448/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/48/903448/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/903448/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/903448/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/903448/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/48/903448/1"}}},"commit":{"parents":[{"commit":"e260d048da6460a1218ae4c210b601864770d393","subject":"Add region name to trove-guestagent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e260d048da6460a1218ae4c210b601864770d393"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-11-16 14:50:28.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-12-12 08:15:26.000000000","tz":0},"subject":"Fix guest agent failed to use swiftclient over TLS","message":"Fix guest agent failed to use swiftclient over TLS\n\nThis commit adds insecure\u003dTrue parameter to swiftclient.\n\nStory: #2010674\nTask: #47748\nChange-Id: I6156b289659dc9130effb1666c3cd2e14366bffe\n(cherry picked from commit 895e59237efb17f8d46763e58210dae96f772d7c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2608fc289c536c6138a179c831ee35a8cd1d6ee3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2608fc289c536c6138a179c831ee35a8cd1d6ee3"}]},"branch":"refs/heads/stable/zed"},"c8579f514357f83b7b786f0389780cebdb453224":{"kind":"REWORK","_number":2,"created":"2023-12-13 01:19:27.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/48/903448/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/48/903448/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/903448/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/903448/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/903448/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/48/903448/2"}}},"commit":{"parents":[{"commit":"e260d048da6460a1218ae4c210b601864770d393","subject":"Add region name to trove-guestagent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e260d048da6460a1218ae4c210b601864770d393"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-11-16 14:50:28.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-12-13 01:19:27.000000000","tz":0},"subject":"Fix guest agent failed to use swiftclient over TLS","message":"Fix guest agent failed to use swiftclient over TLS\n\nThis commit adds insecure\u003dTrue parameter to swiftclient.\n\nStory: #2010674\nTask: #47748\nChange-Id: I6156b289659dc9130effb1666c3cd2e14366bffe\n(cherry picked from commit 895e59237efb17f8d46763e58210dae96f772d7c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c8579f514357f83b7b786f0389780cebdb453224"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c8579f514357f83b7b786f0389780cebdb453224"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
