)]}'
{"id":"openstack%2Ftrove~903449","triplet_id":"openstack%2Ftrove~stable%2Fyoga~I6156b289659dc9130effb1666c3cd2e14366bffe","project":"openstack/trove","branch":"stable/yoga","topic":"2023.2-stable/yoga","attention_set":{},"removed_from_attention_set":{"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2023-12-14 08:47:56.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"I6156b289659dc9130effb1666c3cd2e14366bffe","subject":"Fix guest agent failed to use swiftclient over TLS","status":"MERGED","created":"2023-12-12 08:16:20.000000000","updated":"2023-12-14 11:15:03.000000000","submitted":"2023-12-14 11:11:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"903449-2023.2-stable/yoga","cherry_pick_of_change":903441,"cherry_pick_of_patch_set":3,"meta_rev_id":"72bf7ec863c61776a961fad639954c9c32e99985","_number":903449,"virtual_id_number":903449,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-14 11:11:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-12-14 08:47:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-12-14 08:47:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-12 11:20:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1f961dd3019481a6a3fba9324bc17bb913cfb272","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-12 08:16:20.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.2.","accounts_in_message":[],"_revision_number":1},{"id":"9350d9b06e43a194787bcd8317bac22e6cb0dbdf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 11:20:25.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a29cf6450564ca5b41c2e915e84c324\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c9a324b7ef24ebfb76b256b327817dd : FAILURE in 6m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0eac4da4819423aa7beabb1b2779289 : FAILURE in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4fc8c0d92dcc4514a8ad66382fa575e1 : SUCCESS in 8m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/20850826d6fd47c7a9db0627ad4938ac : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b395d2722db4412db55951f3682c1fac : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04bbc2c6d9584572a136197037282633 : SUCCESS in 5m 32s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5a085903808a4883ad173ba2420a05a9 : SUCCESS in 5m 54s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/3598588e80a74602b9d598aac7c2f0ca : TIMED_OUT in 3h 02m 37s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/30355e0fd03f4933a7a4f8828057d1ff : TIMED_OUT in 3h 02m 29s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9248d6ce3fd34f7783957b027d0f1fca : FAILURE in 41m 05s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/616fecad568c47cfb98b40443fe83356 : FAILURE in 33m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"00ee3fb8be37d1c96eb941fd4fc1c548bacc2efd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-13 02:18:41.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e2948fdfa3f5ccabbd83efa945fd6c7c81bdf4c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 03:46:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1ee05bfd7ac4f1d9b9157965d2964e1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b9738f3bd3a439aa62df070932293ad : FAILURE in 7m 08s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2a06fff68b144b2bc6e78c90f226b89 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f644d0af19554d18a2484c5f70c970c5 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bfc6ef62f23d4126a6ecc915e68cbfa1 : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/141484135d724674b730995852e4f789 : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ddc4225f2364896a07ef1f275e849fb : SUCCESS in 6m 06s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c785a5c53e5e4fea958182d169b88588 : SUCCESS in 6m 54s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/32d5656a47dd4f7cbe2e458cf82cd9c5 : FAILURE in 33m 35s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/c061db28b3af42958f0ffa149355e239 : FAILURE in 34m 01s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/be39b688a46049d599b4a07e45004fb2 : FAILURE in 33m 16s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/040a99f0068e40558db9f17b0be8a023 : FAILURE in 31m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"345d48c13d9a3bf02f1c3d4db928b5b6a38885b4","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-14 08:47:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"490880fb36281dcc1f6bb05ff6f68248088e9bf2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 08:48:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e6c10e568e3e5e5f45decf99fc6d45baf9b27b82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 11:11:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/733b5b4994f64088b3207edaa921c74d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9be4fecf7b1b4bbb8d3bc7cc74c7cf58 : SUCCESS in 7m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70f072d5384d49a59b74320cc1f5180f : SUCCESS in 6m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7014660383264b2db6f32607f0414332 : SUCCESS in 7m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0eab546eac54aa498c5cd1efbb4dd34 : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82d614f671ca4b6bb2a0349543fa9287 : SUCCESS in 5m 52s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/10f34d8136fc446e9825186f528c23ce : FAILURE in 40m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dad46be9130bf630e3a8369b795678475d156885","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 11:11:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"72bf7ec863c61776a961fad639954c9c32e99985","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 11:15:03.000000000","message":"Patch Set 2:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/93edca8f4ad242d8bf845eb293ba9f90\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efc51c0132cc454698eb9c669f688989 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a69d9e1bcabe4007be74c0c69d40e1c9 : FAILURE in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"29e54a6e1fce836fc5a44f305acdee8020f6e065","revisions":{"bb5f1429429617be428300da8474ebdf27f37d83":{"kind":"REWORK","_number":1,"created":"2023-12-12 08:16:20.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/49/903449/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/49/903449/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/49/903449/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/49/903449/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/49/903449/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/49/903449/1"}}},"commit":{"parents":[{"commit":"cc0e021f758617dd0a41dea2addae3bc02e6f002","subject":"Add region name to trove-guestagent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cc0e021f758617dd0a41dea2addae3bc02e6f002"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-11-16 14:50:28.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-12-12 08:16:20.000000000","tz":0},"subject":"Fix guest agent failed to use swiftclient over TLS","message":"Fix guest agent failed to use swiftclient over TLS\n\nThis commit adds insecure\u003dTrue parameter to swiftclient.\n\nStory: #2010674\nTask: #47748\nChange-Id: I6156b289659dc9130effb1666c3cd2e14366bffe\n(cherry picked from commit 895e59237efb17f8d46763e58210dae96f772d7c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bb5f1429429617be428300da8474ebdf27f37d83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bb5f1429429617be428300da8474ebdf27f37d83"}]},"branch":"refs/heads/stable/yoga"},"29e54a6e1fce836fc5a44f305acdee8020f6e065":{"kind":"REWORK","_number":2,"created":"2023-12-13 02:18:41.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/49/903449/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/49/903449/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/49/903449/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/49/903449/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/49/903449/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/49/903449/2"}}},"commit":{"parents":[{"commit":"cc0e021f758617dd0a41dea2addae3bc02e6f002","subject":"Add region name to trove-guestagent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cc0e021f758617dd0a41dea2addae3bc02e6f002"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-11-16 14:50:28.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2023-12-13 02:18:41.000000000","tz":0},"subject":"Fix guest agent failed to use swiftclient over TLS","message":"Fix guest agent failed to use swiftclient over TLS\n\nThis commit adds insecure\u003dTrue parameter to swiftclient.\n\nStory: #2010674\nTask: #47748\nChange-Id: I6156b289659dc9130effb1666c3cd2e14366bffe\n(cherry picked from commit 895e59237efb17f8d46763e58210dae96f772d7c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/29e54a6e1fce836fc5a44f305acdee8020f6e065"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/29e54a6e1fce836fc5a44f305acdee8020f6e065"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
