)]}'
{"id":"openstack%2Ftrove~904255","triplet_id":"openstack%2Ftrove~master~Ie021e0bd708cf8ed13853dea632312af38190205","project":"openstack/trove","branch":"master","topic":"nose-proboscis","attention_set":{},"removed_from_attention_set":{"28691":{"account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"last_update":"2024-03-29 08:26:04.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2024-03-29 05:40:39.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"Ie021e0bd708cf8ed13853dea632312af38190205","subject":"Get rid of nose and Proboscis","status":"MERGED","created":"2023-12-22 09:55:22.000000000","updated":"2024-03-29 08:27:06.000000000","submitted":"2024-03-29 08:26:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"904255-nose-proboscis","meta_rev_id":"e5e9f04ed90d5ad9635c798850f76a4b4a0c4672","_number":904255,"virtual_id_number":904255,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-29 08:26:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-03-29 01:55:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-03-29 01:55:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-22 10:08:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-12-22 11:58:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-25 08:13:29.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"REVIEWER"}],"messages":[{"id":"332a0bdd04a2d1cc984d234b8494562ebc3f8303","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-12-22 09:55:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e39c80249f57ff05c013a348a7ed9c489e3a40cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-22 10:08:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/780c2368134445cda8cf4f8a08060d83\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/32f35439f1c84cf2a576513f9b7a3c41 : FAILURE in 11m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/94642c9f7fe64e6bbc0fd49b76f4a84c : FAILURE in 7m 24s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5d6b0a0c124946b4acc68ae063f2c2a4 : FAILURE in 9m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3e81ea7bc2d5a4737d8e7e31c459e7a2e196e47c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-22 11:58:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f6f1e75d5b5437e823c567ea9f5ad7b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7b1b24b83a044088a3184a119589ad98 : SUCCESS in 5m 11s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6d33033edc5d41d485f10f9a6293adfe : SUCCESS in 3m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0bfac89e67354a43ac306c2cc05812d5 : FAILURE in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/625118fc604343f4882b265e173d8bc7 : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0542fdfa167e483a8a967a7fdc21f2d5 : FAILURE in 5m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e44da4ba6233451ba1358d320eba8a1e : FAILURE in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38160985725641d8876a94c597b7c908 : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a4977622eb346a3b7ef9074f29fda5d : SUCCESS in 5m 47s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/664a145997da48a09598dbc3856cb044 : SUCCESS in 46m 36s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/e726ecd3bb7846babd9e3cd231478fea : FAILURE in 1h 54m 32s\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/37b4393fa99e48f68a999e5e8f1e2eff : SUCCESS in 1h 22m 46s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/beeeba3b7349489fb846570284c2a41e : FAILURE in 1h 14m 07s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/2e827007a5ee48168b82f8b739080449 : SUCCESS in 1h 32m 42s\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/a84d1dadf057479a86e2d1c32b974669 : SUCCESS in 1h 38m 16s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/55f817cf49d6440e8bfdfc9899afacf4 : SUCCESS in 1h 02m 12s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/c6eb27e6e9824636825b0cd1eb0c3f51 : FAILURE in 2h 00m 21s\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/016a58dfe3ee4682a2d2e14d5bbae2a0 : SUCCESS in 1h 34m 49s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2135bfce652a4a1d92febd18f57853a3 : SUCCESS in 7m 11s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/7820207809d8436c9df0d9d5c1f10d45 : SUCCESS in 15m 02s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/85b5d6f468eb41569a5d1b65502e9d56 : SUCCESS in 22m 07s","accounts_in_message":[],"_revision_number":1},{"id":"75cc5d7f97650c50c61b03d937e74f04816f809c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-01-10 07:03:58.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"17caa16e6924d0c9a89241547bc0b485444c1eec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 07:10:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bec76a8dd37c4fe39c737ee5befaa07e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/14cfa85ca1624c2094bba6701d3d018a : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fd2ac7403f844b3ab62af0182effc543 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/328f3307241949e4a357e4aee665b5a6 : SUCCESS in 5m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"378ca7b707966123bb941352dd96d2ee18f83e1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 09:12:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7a55a9d4b394db9ae5d9253b32f9553\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/526460d6f4614cbbb3607912639fdb6a : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f481997a7ab84bd9adcb70f94be53948 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a64bbfbd5ac848fcae1a28a98e5e2b0c : SUCCESS in 5m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/795414dedf6f43ce912bfb09d04c9586 : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4b3ec5217cc446881791d0fadea393b : SUCCESS in 6m 36s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/34e9556a919948e2b0e53fc869c409a3 : SUCCESS in 1h 01m 54s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/32c620dcdc3643c2a3a50324c4da7e2b : SUCCESS in 2h 06m 42s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/4a1453fe8d834f2ab2aabcd9f7ab7c7c : SUCCESS in 1h 28m 18s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/1fd4cbd4fc0e40d798d65a349781bb51 : SUCCESS in 46m 52s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/88a92d8c8cc545c2a41bed462c1ef6ab : SUCCESS in 1h 35m 07s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/ec8d915430c24a8ebe485305642abc66 : SUCCESS in 1h 52m 11s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/a1f02208e10946c3a07e680bf1dfdca9 : SUCCESS in 52m 50s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/1051bc81a1af4728ac7b13894d8125d5 : SUCCESS in 2h 06m 35s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/119877374864406d8812c0ecf4f862f5 : SUCCESS in 1h 21m 03s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d7add2904cea4ba7a00b96914789a931 : SUCCESS in 5m 22s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/f07c2c119ee045ab914dbb107dce08fb : SUCCESS in 10m 30s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/bda88df7a3434b4d93416ccecc685e97 : SUCCESS in 9m 06s","accounts_in_message":[],"_revision_number":2},{"id":"86041d8de824621c9c3a6e560ee5ebe2319fa5fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-01-11 03:41:02.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"825af48e82379a1c84de5f57159f063dc46c8144","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 04:02:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79fbcfd65ab743249905a5f2ed41b9b0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cb3c36b9f3d54f478f6aa177a8eeda25 : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/77ea1924de4448209b143aaa7bd05e5a : SUCCESS in 18m 42s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f2384fac504e431b891e92546c8c6170 : SUCCESS in 5m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e8ffad82674a25573614a48539fdab2377b5fd42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 05:51:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ddd72c4855d44bf4984742f2e5b23752\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a3749f2278f347b3b75c0ed944e03a4c : SUCCESS in 6m 44s\n- requirements-check https://zuul.opendev.org/t/openstack/build/289d05bc7d1f4511981daf4b72abd485 : SUCCESS in 3m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/462650da447b4c34b104da8ee51a7cb9 : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df4e775deb2249b3b109a4dd48ca2e57 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67ba6cb055024802bde80ff402dcab05 : SUCCESS in 5m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/11705be738294d4f946c0794c846e1b4 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3729bb8678d449ba89504aee1328fe5 : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41830f5d1e3749ffa097b62c859e49b2 : SUCCESS in 5m 35s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/9f02622a78074d03a09060d5dfadbb4a : SUCCESS in 44m 26s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/d7c5723a302044f99c8806ec06afe791 : SUCCESS in 1h 42m 40s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/093c2d2eecb64f6e8777a04a702e0e50 : SUCCESS in 1h 21m 17s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b5b5bad0e2674c5886cc50e99e3274f9 : SUCCESS in 42m 12s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/0c21e2c5ee044f8fa48e4a44c466106c : FAILURE in 1h 26m 00s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/6e983452bb204d97a97c99c2fb1c4a68 : SUCCESS in 1h 39m 37s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/8650e00bce8d4c12b05717a316f62e2f : SUCCESS in 58m 24s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/dab2539b12024d9bbd0add36815dd2f3 : SUCCESS in 2h 08m 59s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/edeea237499c457f8d56f06d912fc760 : SUCCESS in 1h 42m 36s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/751a9b92ff974b0aba9711caa61eff17 : SUCCESS in 5m 23s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/fe60c72de45640438945fe8003f49f46 : SUCCESS in 9m 45s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/e4163d443c30482fa918b1105e59c3b2 : SUCCESS in 11m 43s","accounts_in_message":[],"_revision_number":3},{"id":"e0df3ace4067d7d55e53a45e2096ff1e55738fc2","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-01-25 08:13:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7af3a8e8300d532e9d691a8f853ed649b698b8f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-03-26 13:45:20.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"53291a0300ba86197430569d7b2d1a85d94caeb3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 13:51:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a2b45a6945143efb353cd6c2e7f27f3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/647e549d76b34664ad8acde661c7604a : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4dc737b4838d43d98c9858a8e130758a : SUCCESS in 4m 42s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9e236438e31e9d50ec5ea90ac96c49a4a08268c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 16:02:54.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d92476526324a4388d2a5326276f517\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c12d6865e6da49f5bf620606f6866344 : SUCCESS in 4m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e3797f8f45f34dd7b9f7af27c2281188 : SUCCESS in 3m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0bb04f3ec06e4bcb844cf642d83e73ae : FAILURE in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6464ff535e84c6bb9fb0d10f1706840 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/23da7775f5844fcdab12b1445e52f12e : SUCCESS in 5m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/76ef33db67814cee8cad27d9af1b2655 : SUCCESS in 5m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9393104d9a6c42e39f8688958a424add : SUCCESS in 10m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99f29a156bf9445c90917a077e3281b4 : SUCCESS in 4m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/321ce250fa764bb8b770b2c14a591b5a : SUCCESS in 6m 18s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/63d61058a1b74183b0b22bb39f1bdde9 : SUCCESS in 1h 10m 32s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/a7648762cb2c4ee4898a1ac0a863cf92 : FAILURE in 2h 03m 21s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/c8fb716c731e4128b0798e92b4b8cd09 : SUCCESS in 2h 13m 04s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/3fd3400b713e40cfad3cf34fd8c86756 : SUCCESS in 54m 03s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/01f80e5c176144d080341acfb69bb5f1 : SUCCESS in 1h 26m 37s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/eceade81ebdd4fe589364cb5d0822458 : SUCCESS in 1h 11m 57s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/4ac936be4b814c5781f23d1e16dfb7db : SUCCESS in 41m 21s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/9c36b935ca154a4eb5cf694f0a06ea56 : SUCCESS in 2h 15m 33s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/879c29ac7d3d41ceb9b8f1863e231d5e : SUCCESS in 2h 00m 43s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/140ed4ec5d2240d6bc3c014397d01602 : SUCCESS in 5m 02s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/0a1f09bbe07949af8a4bdeb2586f3543 : SUCCESS in 11m 28s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/58434b396ffc47c186d2b191cfb83d80 : SUCCESS in 12m 25s","accounts_in_message":[],"_revision_number":4},{"id":"8c9abda194ce2d839ef806f7b645559ae1c5191c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-03-27 05:30:33.000000000","message":"Uploaded patch set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"1033604f6ce4c33e7ad3658b7a800beb1589c1c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-27 05:41:24.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c004df25a584a068046da0cadba663a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/94719ee101504557a512d427321e1b9c : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b74a4812d7444dabbef041626d849d49 : SUCCESS in 9m 05s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6baa6e4e58ac732ad9b5627fc2eff7eb90bd6feb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-27 07:14:42.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a7b02ea27f64ad8bc068205575f68fe\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1580ef34b7444244bfcd15baee71cd1e : SUCCESS in 5m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f52af53eca14410c8ed3575cf1e82e1c : SUCCESS in 2m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9032ef893674d4c816ded08fba6d31e : FAILURE in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68333b26c1dc4d42ae768ede3ffe9117 : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2de3fa3479704811968d26728e6bde28 : SUCCESS in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6ef03487993549aca63aeca98a742d32 : SUCCESS in 4m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d9335acdcf9640d2a5144e889f22cbbe : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/483e07b67e9b442d9e0d07f7ab0a1e6c : SUCCESS in 4m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20ae29ebd7ff4c16ac3842cd77d76688 : SUCCESS in 4m 27s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/8e9c20471ac34f51a286aec4e4a37306 : SUCCESS in 44m 31s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/55657ec9263e45c99609e51cbcaa7979 : SUCCESS in 1h 38m 35s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/d8af60ffa2134cf193be4ae68b84fc0b : SUCCESS in 1h 36m 05s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/beecaa1db2f24115ac48c64dc9e5b952 : SUCCESS in 42m 18s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/122fbebfcbda4064bb249590b2952386 : SUCCESS in 1h 28m 05s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/d0fcf05e008743139b79dfd612918709 : SUCCESS in 1h 16m 05s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/b0580d1db9cf4b9ea0584814f5daf2a9 : SUCCESS in 46m 02s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/03a7ae169a474c559bfaaeaf0f794225 : SUCCESS in 1h 38m 43s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/3a770cbbd41d4fd99e03110e79436bd7 : SUCCESS in 1h 36m 12s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/871468022b3e4120883ba42b1ddd1b11 : SUCCESS in 5m 22s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/b7d191abb6134c699b58732f4cc5cf0b : SUCCESS in 9m 48s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/170055a40fa84552994bcee1a910bd13 : SUCCESS in 11m 33s","accounts_in_message":[],"_revision_number":5},{"id":"4ab5192e2ea29e29ac1647223d13fabf4dbde905","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-03-27 08:14:16.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"89b3d0be47a568b7716b9161ac375c33f0cb0c7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-27 08:26:51.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/843b54ad92df4bdba7bf00be9c495d99\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a4d1df6fee844ff1838c7aa513621c60 : SUCCESS in 4m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf11fcb21bfe453f867220ee2c145d96 : FAILURE in 7m 03s","accounts_in_message":[],"_revision_number":6},{"id":"a7ba0e46bf4cda58efb0cd2a335bba220464da3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-03-27 15:07:14.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"3ba9447090fd4e2615df2bcefdb57e77154994a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-27 15:14:40.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb9e86c2f48d481e97b333878c3b70fd\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/25b07e55fa8b430aa6bc4edb36931ef5 : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2ea06fc657c74289a9affa29c022cc17 : SUCCESS in 4m 45s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"39433e66eb7c19d487afe0ec822dd159169dad58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-27 18:03:52.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7cfac4b417a94cacade1103913411e89\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/77bff9a684a34ae6a5ba4a1c4b1e1208 : SUCCESS in 5m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b338ded5f4184d78bd8c997721381043 : SUCCESS in 3m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34c3918c79eb456cbd2329c0a4ae8824 : SUCCESS in 8m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/462fc1b7ecd443b5bb0a695dde2bef1e : FAILURE in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89e4a09a09834906b02a756df56d2d95 : SUCCESS in 6m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/07ac440fb7b34f5a9f49e558b783cd2b : SUCCESS in 5m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/55f17c65b9a3463997f4eb6db218e1a8 : SUCCESS in 8m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16c1d5a0159a4025803c6ccef6477f24 : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/085908d282894d349d250e623bd0fbcd : SUCCESS in 6m 55s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/c6fd016b73764aaa9258cc4df18d88df : SUCCESS in 46m 06s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/59b7108b738642a0b4ec3321013d8365 : SUCCESS in 1h 39m 34s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/faed1b38dbf441839856e0d1caf356ae : SUCCESS in 1h 16m 18s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/3017dc4a630f40babe85941c06998517 : SUCCESS in 43m 14s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/3935ee70e43d46238a32f1dfbc1cf064 : SUCCESS in 2h 24m 18s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/9797238ce35740a880e9db0d38cc2f4c : SUCCESS in 1h 44m 24s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/161bf083c8cf4814a5c0a90a32a3aab6 : SUCCESS in 1h 17m 42s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/aed2dc6f143341f58a8f340e407e4312 : SUCCESS in 2h 02m 14s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/28a2be890dd4446faebfd260bd53b16d : FAILURE in 1h 39m 55s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/435dde909ef34701a360be1d329f5d4a : SUCCESS in 5m 56s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/74730ee7de9645038e5b79af04ed3232 : SUCCESS in 10m 10s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/b9fe1156575e4d18912cbdfe12bb9135 : SUCCESS in 13m 25s\n\nWarning:\n  Comments left for invalid file trove/db/sqlalchemy/api.py","accounts_in_message":[],"_revision_number":7},{"id":"38ea521841ca069be341b70a6b0004ed98370498","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-03-28 07:07:15.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"c9cea5b4c0147cf1a1b4e2db580b7b6d06dad444","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 07:13:22.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f030b107c1f348bfaa79241411f1a1cf\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/36550826e34047c198582f41b207e630 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/909f7e821f824e589a3fe873736eee2f : SUCCESS in 4m 48s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"bc851406e7610dabd5c0011118674ee47868e906","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 08:52:48.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/74424e76dcf4424cab30c34ae0458ef8\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c30ae611730842628fc22e10c848517e : SUCCESS in 4m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7559e8a8292c466c8dc55ae7d94e51f2 : SUCCESS in 2m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c48922e56234bd0b5326026c6f1be85 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba15fb966f6d419fa72b4bb36a1d2fc2 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3a57291dbf2c4b448f8d9ddc9f2eb34a : SUCCESS in 5m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/857e901f44354b8e9b3b31e37d20e330 : SUCCESS in 4m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/40c77632eab04da5a648e1036314a2b8 : SUCCESS in 7m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e48bbabbe8d42dbb4516e2f1734f20a : SUCCESS in 4m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86fa6ff1e6104d7ca0743b131e509dbf : SUCCESS in 4m 20s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/3df9a24ade13487f868aa12d4348bc9b : SUCCESS in 50m 27s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/066f84671d3142cd8b7c241f1e13c1ac : SUCCESS in 1h 44m 23s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/9ee49878169949889bc3cb3087ba81a1 : SUCCESS in 1h 27m 37s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/e2d3fe51c4ee4dada0991dc339027110 : FAILURE in 47m 27s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/273e23780e834d01b7b4c62caa0f8611 : SUCCESS in 1h 34m 14s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/ce1e6b2560f846bebebc3d425ce749d9 : SUCCESS in 1h 26m 13s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/3fc81fb416f24ae39968797a9104c036 : SUCCESS in 44m 34s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/67a8b95f8b774ab1b88e24f699c4c4b0 : SUCCESS in 1h 43m 22s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/3a0a0a80ff734bf388393163c8c7e3f8 : SUCCESS in 1h 29m 10s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a9eea78be66e48d09d3f63c527ce690d : SUCCESS in 5m 16s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/2ac48dcae1824727b5c8e5f339cb2a7d : SUCCESS in 7m 27s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/577ea9f173e8424a958788ace2beb77c : SUCCESS in 13m 31s","accounts_in_message":[],"_revision_number":8},{"id":"6534870e81ad0ac651b14644cfccd1d51511a078","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-03-28 09:30:50.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"0bd4f363c7f391e6e4a728498ad54b0111f56568","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 09:46:07.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d9ce12ffd244b1b83e02f409c4b0d76\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0bf0f120bd23451f9c246c2ac7a7de83 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c739d8cfc355438bb60d31c9ea38ed55 : SUCCESS in 13m 24s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"7cf7ed6005145b797c66c8f8979a306f8bb80b1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 11:53:58.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae5c7f0fb1f2450892a717aca3885bec\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8098afd0973f41dd957b636fe57c0ff3 : SUCCESS in 4m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0c6d86412fbd4a888814162f2e1eae20 : SUCCESS in 3m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/edcba4deb0464f769bb7f06fd82eaaeb : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eec8ce511ea14f12b72b02e291053e2b : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/300ce2eeb49e4b9dbc5c7998c10aec2d : SUCCESS in 5m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/10ee687fbecd4734bf999e21ccb1c6f2 : SUCCESS in 4m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d78c87a923424b58b057dc3edf0574a2 : SUCCESS in 8m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c79aa0889de84287a3b621f0ff2762a6 : SUCCESS in 5m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e5a41544c534fed9aee9e67e05af2b0 : SUCCESS in 5m 26s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/bb0df18051e14bd3a64fc79160afed96 : SUCCESS in 40m 20s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/5978382dbd294ed39bc3804f7501f647 : SUCCESS in 2h 19m 21s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/34400396e4794a588e9d9c16d132ee62 : SUCCESS in 2h 08m 41s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/74d8caa0ae044351a63a324eb4d1643a : SUCCESS in 50m 22s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/e7d91a8d88b74bc8ac530baadf0736f4 : SUCCESS in 2h 20m 33s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/fb8503537eeb4a6b8ca31360bbaf6e2c : SUCCESS in 1h 57m 37s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/d92ba9f4d93f4a85bf122fc6a955af8f : SUCCESS in 1h 04m 52s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/28cbf54071394d079bdce177168f6fd1 : FAILURE in 1h 16m 28s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/0322b23d96d1413db136c9eeae3c32a9 : SUCCESS in 2h 01m 29s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5f85f889282b48cda6c36d2b0e36972a : SUCCESS in 4m 54s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/342afb0f1c2241a69390475b95571e4c : SUCCESS in 10m 10s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/b647602e362c44598a7103e258556ed3 : SUCCESS in 15m 29s","accounts_in_message":[],"_revision_number":8},{"id":"aa2fadac5b46a2cbde6322f2b74fe7c9b3bc56df","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-03-29 01:55:11.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"2ab5886bafceef6276958b8b04c08cb6e6fa7cf9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 01:55:43.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"b05129103e2630a78d2fe8c5926c55076aa24d86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 04:48:47.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/14a7c6aa47a14596810ca857dfd6b9f8\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/13775c9e61d94988be2d2404e742f2b0 : SUCCESS in 6m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/186ed0199a24488f877280706edf587d : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36963ad03f524309974665d136fdb366 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c6152cb6d6694e2db94236e1e38a90bb : SUCCESS in 5m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5e5f7cc77e4d4d38b7f9e0c0777d78bf : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17f4069538e1440b8fa26bd2571ba352 : SUCCESS in 4m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a0a19c207ba43d58f0cf669597ea9fa : SUCCESS in 7m 44s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/11135c85279c47adbe957e7b2d097c17 : FAILURE in 20m 11s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/cb022168776649979134c240a3f2b306 : SUCCESS in 1h 04m 25s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/f30ca86b8aa24aae91055f289e724531 : SUCCESS in 42m 51s","accounts_in_message":[],"_revision_number":8},{"id":"2641777f3c37396a775bcf7aba99bfcbf088190a","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-03-29 05:40:39.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"41cabfae822b99509dd0962283942db9727cdd16","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 05:47:36.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/083aaf094d184f08b93beedd6e7e7af1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0f908993b5844e988379918e75b749e7 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/828dd22c03fa4235aac87f465f1d43d4 : SUCCESS in 5m 34s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"eefed8fd2f79557cb12f5954a0c7ec5e833c4b0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 07:28:42.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f68f7b752b4a46168db0df69cfdfcda2\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c8f41c93b3f14c75b2719f982002fdfc : SUCCESS in 5m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f9a91fdac86f4bb5a2886c9161d9ef2a : SUCCESS in 2m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57ed6d1d95314ea09dbd56c084d34783 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f89abaadd4414359bd060c56d466e3ae : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37b62908706c44099dd2886ea03ab1d1 : SUCCESS in 5m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3528326246214275bc75ba2c4caf8fee : SUCCESS in 5m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b30992215c8a41dfb45da7ad9060b19a : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77b93af0868546acb7e5c8b5d4726780 : SUCCESS in 4m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97915e8fe6f54e7994996492fe757061 : SUCCESS in 5m 40s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/007ca3e440ef40c6b7c22c6f91b37b13 : SUCCESS in 55m 01s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/82e11d105f2840c6966f65b0d0ab0f3e : SUCCESS in 1h 46m 42s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/f872b5102771456cba1df0f823eb6e0c : SUCCESS in 1h 18m 53s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/5c6f235c2c284660ac0b3831eee4f22d : SUCCESS in 51m 23s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/cc6ee670280d4a44a833b2320155ec15 : SUCCESS in 1h 23m 04s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/a79260c224d84619bf6329f3ada55a1e : SUCCESS in 1h 09m 31s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/5e980c142e3e457887469649a617aaff : SUCCESS in 53m 41s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/a10a1c97a8f84b3a8eb6a46b8ac3279a : SUCCESS in 1h 36m 56s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/b2cc4722213b4388b6b72cef58b0959f : SUCCESS in 1h 22m 42s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/08d07c79582742ed9f1ccebd4d355a3d : SUCCESS in 4m 45s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/53811ff5c268444383524262d9d45091 : SUCCESS in 7m 53s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/a94940c47cc74f47bae8c7b02f0cf0b7 : SUCCESS in 9m 36s","accounts_in_message":[],"_revision_number":8},{"id":"c0d7ad84ad1aa6cfbdb1dbd3ef4b50399b21298f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 07:29:15.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"71460f7a9e4160cd3a038773cb9c6a624d6a7673","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 08:26:03.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eba0fb74463c47e5b3dbe1f2d83b6402\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f954cb64df1e44509569de7a700a3080 : SUCCESS in 5m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/eb16f8ad40cc4142963b7edd122adb7a : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49c118d1543e44d3be30ee8d2cbf5b8d : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb1c07d726b543a1938b4e5a2543ff22 : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/27dd158f1b21424f960b7e3812b29a42 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec405ef2295e4a9bafebfe76ce9480a8 : SUCCESS in 4m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3411d2477e0f4a4eacf42bcc01cb0e43 : SUCCESS in 5m 28s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/13cdd2dc1ea74266b2e9c7097eab1af1 : SUCCESS in 54m 34s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/ed961f33e8e24ea6bca492e24ab62c6c : SUCCESS in 46m 07s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/8638a364c7384a3f8cc3adbff4b532e7 : SUCCESS in 48m 29s","accounts_in_message":[],"_revision_number":8},{"id":"fe03d6e1312ace716e2606648c9ed9713f15b0ed","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 08:26:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"e5e9f04ed90d5ad9635c798850f76a4b4a0c4672","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 08:27:06.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3213ac67560c4f2c8a90f5ebbcb8a8ed\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/2c8c065ea5ec46848620bb33730f6351 : SUCCESS in 41s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/171f967147124df2b98ee9981cf04dc1 : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e84fb2b39cf497783e5f666e45f46be : SUCCESS in 43s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"93f35c7f041878bf1afa30c861926c0a9e1296f4","revisions":{"25484ed15b2b3578fdabb31472ecc79eb0fa39c2":{"kind":"REWORK","_number":1,"created":"2023-12-22 09:55:22.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/55/904255/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/55/904255/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/55/904255/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/55/904255/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/55/904255/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/55/904255/1"}}},"commit":{"parents":[{"commit":"8b6ff821a13888336c141f9b006e648bf75a1744","subject":"Cinder Support For Boot Volume","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8b6ff821a13888336c141f9b006e648bf75a1744"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-22 09:51:38.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-22 09:51:38.000000000","tz":480},"subject":"Get rid of nose and Proboscis","message":"Get rid of nose and Proboscis\n\nthey are not maintained for a long time, and trove already switch\nthe tests to templest.\nThis commit remove the usage of them. more detals see story.\n\nStory: 2010882\nTask: 48606\nChange-Id: Ie021e0bd708cf8ed13853dea632312af38190205\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/25484ed15b2b3578fdabb31472ecc79eb0fa39c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/25484ed15b2b3578fdabb31472ecc79eb0fa39c2"}]},"branch":"refs/heads/master"},"efc4c7aa5963b4c818e8f0a58b3ef59de9e5c155":{"kind":"REWORK","_number":2,"created":"2024-01-10 07:03:58.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/55/904255/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/55/904255/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/55/904255/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/55/904255/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/55/904255/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/55/904255/2"}}},"commit":{"parents":[{"commit":"ca28382ed9a97bac9e10e2333fb5328ebaa266c6","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ca28382ed9a97bac9e10e2333fb5328ebaa266c6"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-22 09:51:38.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-01-10 07:03:32.000000000","tz":480},"subject":"Get rid of nose and Proboscis [part1]","message":"Get rid of nose and Proboscis [part1]\n\nthey are not maintained for a long time, and trove already switch\nthe tests to templest.\nThis commit remove the usage of them. more detals see story.\n\nStory: 2010882\nTask: 48606\nChange-Id: Ie021e0bd708cf8ed13853dea632312af38190205\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/efc4c7aa5963b4c818e8f0a58b3ef59de9e5c155"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/efc4c7aa5963b4c818e8f0a58b3ef59de9e5c155"}]},"branch":"refs/heads/master"},"cc24d0a0e207dbd872ea65290e5a132f52f359ac":{"kind":"REWORK","_number":3,"created":"2024-01-11 03:41:02.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/55/904255/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/55/904255/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/55/904255/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/55/904255/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/55/904255/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/55/904255/3"}}},"commit":{"parents":[{"commit":"ca28382ed9a97bac9e10e2333fb5328ebaa266c6","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ca28382ed9a97bac9e10e2333fb5328ebaa266c6"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-22 09:51:38.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-01-11 03:40:16.000000000","tz":480},"subject":"Get rid of nose and Proboscis","message":"Get rid of nose and Proboscis\n\nthey are not maintained for a long time, and trove already switch\nthe tests to templest.\nThis commit removes the usage of them. more detals see story.\n\nStory: 2010882\nTask: 48606\nChange-Id: Ie021e0bd708cf8ed13853dea632312af38190205\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cc24d0a0e207dbd872ea65290e5a132f52f359ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cc24d0a0e207dbd872ea65290e5a132f52f359ac"}]},"branch":"refs/heads/master"},"70b7631f7266e0a7b966d03d600f62dab3babd56":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-03-26 13:45:20.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/55/904255/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/55/904255/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/55/904255/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/55/904255/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/55/904255/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/55/904255/4"}}},"commit":{"parents":[{"commit":"83a525067deb5a182e6d439a2acdb68695d9a2d8","subject":"Merge \"Retire TripleO: remove the TripleO jobs/ref\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/83a525067deb5a182e6d439a2acdb68695d9a2d8"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-22 09:51:38.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2024-03-26 13:45:20.000000000","tz":0},"subject":"Get rid of nose and Proboscis","message":"Get rid of nose and Proboscis\n\nthey are not maintained for a long time, and trove already switch\nthe tests to templest.\nThis commit removes the usage of them. more detals see story.\n\nStory: 2010882\nTask: 48606\nChange-Id: Ie021e0bd708cf8ed13853dea632312af38190205\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/70b7631f7266e0a7b966d03d600f62dab3babd56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/70b7631f7266e0a7b966d03d600f62dab3babd56"}]},"branch":"refs/heads/master","description":"Rebase"},"0d46486336735602f0d5f0403a8038bf029facc6":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2024-03-27 05:30:33.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/55/904255/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/55/904255/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/55/904255/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/55/904255/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/55/904255/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/55/904255/5"}}},"commit":{"parents":[{"commit":"83a525067deb5a182e6d439a2acdb68695d9a2d8","subject":"Merge \"Retire TripleO: remove the TripleO jobs/ref\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/83a525067deb5a182e6d439a2acdb68695d9a2d8"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-22 09:51:38.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-03-27 05:29:49.000000000","tz":480},"subject":"Get rid of nose and Proboscis","message":"Get rid of nose and Proboscis\n\nthey are not maintained for a long time, and trove already switch\nthe tests to templest.\nThis commit removes the usage of them. more details see story.\n\nStory: 2010882\nTask: 48606\nChange-Id: Ie021e0bd708cf8ed13853dea632312af38190205\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0d46486336735602f0d5f0403a8038bf029facc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0d46486336735602f0d5f0403a8038bf029facc6"}]},"branch":"refs/heads/master"},"fa69a419e3f38f7fe8f7535f56028f4147171b27":{"kind":"REWORK","_number":6,"created":"2024-03-27 08:14:16.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/55/904255/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/55/904255/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/55/904255/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/55/904255/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/55/904255/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/55/904255/6"}}},"commit":{"parents":[{"commit":"83a525067deb5a182e6d439a2acdb68695d9a2d8","subject":"Merge \"Retire TripleO: remove the TripleO jobs/ref\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/83a525067deb5a182e6d439a2acdb68695d9a2d8"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-22 09:51:38.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-03-27 08:14:02.000000000","tz":480},"subject":"Get rid of nose and Proboscis","message":"Get rid of nose and Proboscis\n\nthey are not maintained for a long time, and trove already switch\nthe tests to templest.\nThis commit removes the usage of them. more details see story.\n\nStory: 2010882\nTask: 48606\nChange-Id: Ie021e0bd708cf8ed13853dea632312af38190205\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fa69a419e3f38f7fe8f7535f56028f4147171b27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fa69a419e3f38f7fe8f7535f56028f4147171b27"}]},"branch":"refs/heads/master"},"77c8e3b955796fadf226f31d7554c2a6ed1f5eff":{"kind":"REWORK","_number":7,"created":"2024-03-27 15:07:14.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/55/904255/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/55/904255/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/55/904255/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/55/904255/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/55/904255/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/55/904255/7"}}},"commit":{"parents":[{"commit":"83a525067deb5a182e6d439a2acdb68695d9a2d8","subject":"Merge \"Retire TripleO: remove the TripleO jobs/ref\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/83a525067deb5a182e6d439a2acdb68695d9a2d8"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-22 09:51:38.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-03-27 15:04:02.000000000","tz":480},"subject":"Get rid of nose and Proboscis","message":"Get rid of nose and Proboscis\n\nthey are not maintained for a long time, and trove already switch\nthe tests to templest.\nThis commit removes the usage of them. more details see story.\n\nStory: 2010882\nTask: 48606\n\nDepends-On: https://review.opendev.org/c/openstack/trove/+/914228\n\nChange-Id: Ie021e0bd708cf8ed13853dea632312af38190205\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/77c8e3b955796fadf226f31d7554c2a6ed1f5eff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/77c8e3b955796fadf226f31d7554c2a6ed1f5eff"}]},"branch":"refs/heads/master"},"93f35c7f041878bf1afa30c861926c0a9e1296f4":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-03-28 07:07:15.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/55/904255/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/55/904255/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/55/904255/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/55/904255/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/55/904255/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/55/904255/8"}}},"commit":{"parents":[{"commit":"3863b81a18b2340cc71a6d9abbfec913d1193179","subject":"Merge \"db: add commit action after querying\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3863b81a18b2340cc71a6d9abbfec913d1193179"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-12-22 09:51:38.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2024-03-28 07:07:15.000000000","tz":0},"subject":"Get rid of nose and Proboscis","message":"Get rid of nose and Proboscis\n\nthey are not maintained for a long time, and trove already switch\nthe tests to templest.\nThis commit removes the usage of them. more details see story.\n\nStory: 2010882\nTask: 48606\n\nDepends-On: https://review.opendev.org/c/openstack/trove/+/914228\n\nChange-Id: Ie021e0bd708cf8ed13853dea632312af38190205\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/93f35c7f041878bf1afa30c861926c0a9e1296f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/93f35c7f041878bf1afa30c861926c0a9e1296f4"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
