)]}'
{"id":"openstack%2Ftrove~904365","triplet_id":"openstack%2Ftrove~master~Ib9d003065f943a5382e1079397bfde4292f5dabb","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-07-09 06:25:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib9d003065f943a5382e1079397bfde4292f5dabb","subject":"Use policy to determine admin context","status":"MERGED","created":"2023-12-27 01:36:01.000000000","updated":"2024-07-09 06:27:31.000000000","submitted":"2024-07-09 06:25:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"904365","meta_rev_id":"10c7498bbdf76a565deaefaa6b516189cb53fed9","_number":904365,"virtual_id_number":904365,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-09 06:25:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-07-09 05:25:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-07-09 05:25:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-27 01:42:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-12-27 02:29:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-11 06:59:01.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"CC"},{"updated":"2024-07-09 05:25:12.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"5050cb4faae8dda5dccbaddf72a48e9a208493ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-27 01:36:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"da5d985c8ba1b89e2174d15fbfa00fccfc10db3d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-27 01:42:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e154f3744ea347f0bab8ab6bfef22299\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e64ed652f786423d9042e8a2a4a04729 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bd904443aa64453c9471a0f8ef98994e : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ba803ff3725c43609439459c4d93e9ca : SUCCESS in 5m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7a88443ff819c3e39958fcfe21a0b2ec1a9fa00c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-27 02:29:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4dc370346834482dabe3c09e67797007\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ae143b31d0d42a795476c1e175fc646 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaaee5f355a04b529fb837fa160f24d2 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b77d9a3ebdc4e6f9fd738e9075f0987 : SUCCESS in 5m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6c28b6386af1425db5694024fb692fde : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eceb237eed7f4d528223d7eb5005195d : SUCCESS in 5m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26e273e38b134352aa429baa325c8551 : SUCCESS in 5m 22s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/623ffe72221047d3b6dc77deba45ca5d : FAILURE in 28m 16s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/8839663a8aa54e8db7aff820c7139742 : FAILURE in 50m 54s\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/42da263e7b0a44ef9b90c92b48b69834 : FAILURE in 40m 08s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/3ae3583572ba42d98a7c8ac5cf315a5f : FAILURE in 37m 35s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b9de5fd43cfb43349eb518e76a5ac64d : FAILURE in 42m 59s\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/6faf2972825c4e9ab457bce201f62374 : FAILURE in 50m 32s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/17f129ed88844062930302b162855b37 : FAILURE in 30m 05s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/8cbbc87954b44373ad433d976ca2cf14 : FAILURE in 51m 43s\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/0c5eed84e3d24d5d97a8122b24c83cee : FAILURE in 41m 36s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2c12f2483c1e453fb3555a5420a5da4a : SUCCESS in 6m 24s","accounts_in_message":[],"_revision_number":1},{"id":"dc8651e98f21a002267620c98c5fca4e0faadd89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-02 13:26:16.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"770da7fe765d92194d0a0f4e350eec02819241dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-02 13:40:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d13089ab657b4618b520245f7d4a80f3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ccfafc278de645bfb9e9dc099be98b54 : SUCCESS in 11m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a4ec12cb94874bd1b883d88ee5a2e17c : SUCCESS in 11m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e69b45834f9c4ce3b699c12a3173588e : SUCCESS in 9m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"968956acc7ad21dca68332773754cd9b8134190e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-02 15:49:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e2c584b98e84eb08cdf0db94609e049\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c4ea0af73054d739667735421a81dd4 : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/138c3aca5cfe49b3a31f2f636acc3ff6 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbf78b8f107342f78dd3271fdda85da0 : SUCCESS in 5m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f6d2477832d240dfaaa5ce12a719452b : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a34e2469c3434b9e9b827d31a0a8da58 : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/096b1a84f4a641ffb49135aeb70bef8f : SUCCESS in 5m 05s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/51b0f4bfeaea4474b54d4f4aaf17c7b2 : SUCCESS in 53m 42s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/a9e16b49fa224d228b424748d22b4352 : SUCCESS in 2h 02m 32s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/dfdd060e45bd419ea3f49a4a097367f8 : SUCCESS in 1h 42m 28s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/3fec6f68a8ed4b189b22112ab69c18d3 : SUCCESS in 52m 54s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/d59a74c252814094ac7b49550f2870d4 : SUCCESS in 2h 20m 05s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/65589a75e5f34a1894b0e452d9c01d05 : SUCCESS in 1h 48m 23s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/c96feabc66a1404a9854a3e62c9190b4 : SUCCESS in 58m 37s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/568b83e0ebc142a4a65e6c90033272ba : FAILURE in 1h 18m 15s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/9511359b24a648f79b4ec73caa5c3907 : SUCCESS in 1h 43m 57s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/97b3dfdb4e7d471887364d23684c17a9 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":2},{"id":"97f5a23012f895d2308a93874d5207f3111aa423","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-01-11 06:59:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0d4617e32fca8e711378a96c94774edcd116b496","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-04-02 06:39:09.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_9816\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"_revision_number":3},{"id":"299e6558695a703bb8fe901b93a52a6a0afc3eef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 06:45:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a93561828f94366aeb03dd8b8eeb143\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0b1673bc2a474c19aa299c1e8ad7e418 : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/02d5d323147b45388106880918589232 : SUCCESS in 5m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3e808ced3fefe4368d89e4ca964ee584a87426f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 09:45:21.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4f6e71bb101e437a8a0ae26f42c75c69\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a37ad7f9ff140e8ac743d6d4833dc39 : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d58c6f5b1a754247b9637a269524404d : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d59ab5eb9e5842b49bd5378193c5c1ec : SUCCESS in 5m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e1a1b5133dee42bfae7ae322e2d6d422 : SUCCESS in 7m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5b88c9e9e34c46cfadcf247abd3a1037 : SUCCESS in 10m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4e123b7c4f7477792dff6dbdf4fb238 : SUCCESS in 4m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7861cec980d466e8dda962bc28fcfde : SUCCESS in 4m 26s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/184d94ca0f9743e1892603094eb376e6 : SUCCESS in 57m 55s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/2d2edb4ba5c34fe7b69af0a98db2a44f : SUCCESS in 2h 00m 43s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/8933cf2f86d14b45ad098b7905abb608 : SUCCESS in 1h 54m 06s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/5b2cb12c88244977ac617b506a5de8ff : SUCCESS in 46m 30s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/650e73822f9f499789a59850aaf31593 : SUCCESS in 1h 56m 12s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/1e1c72bd1d6144dcb4228a1861c3d36b : POST_FAILURE in 44m 02s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/b51df641afaf41efab7258129643fbac : FAILURE in 51m 49s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/3c49ece0db1e47f9bd76aaaa90923263 : FAILURE in 55m 07s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/c0066a4a705846089241e3de93105222 : SUCCESS in 1h 12m 12s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/27121b3540ee41a0aa6f611878ab0b63 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":3},{"id":"2655a38e2caf80a00802aefd431ea4377d502dee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-04-28 01:06:14.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_9816\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"_revision_number":4},{"id":"adb52b7b354602271563b7d1790079a303e51135","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-28 01:14:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42e2896e1fd347449cfce4c03ce47943\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/10d728b4ee674f76afdd3f9c09a2cc70 : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e6dbaf25753b4030af9cf3ad2457765c : SUCCESS in 5m 12s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8676b5c2392d98712f60d720d7f44b2084428349","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-28 03:24:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c5112c012e14bb38d702bf16fac99a3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02efe9826eff4788bd50edf149dea26e : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/804065d97e2b4a4eb783f2c6735a3c62 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d30047ed4ff24f9788638a44c6097cc4 : SUCCESS in 7m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a29e1a55d5d0497d910ee7358994ae1d : SUCCESS in 5m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ec9219797f6947008ca206811df1adb2 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4f5e423ccf9400f861ff47a9af58660 : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1616317b378a400fb97f3499ac142f46 : SUCCESS in 5m 15s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/af38f5ce1ea14079ab0f2990d16519f1 : SUCCESS in 40m 57s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/76e777d39ad545b8bc36c0f7eba25b6c : SUCCESS in 1h 35m 49s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/5ea3dfbd0c2e4017bb9ea40eaa53b0f6 : FAILURE in 2h 11m 27s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/fbb8ea013341414cab3d6129c55bb518 : SUCCESS in 45m 08s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/56d4bbdaee244a2fadb2152e08037d98 : FAILURE in 1h 51m 51s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/7caf244848844fd981df3ef7e56ca849 : SUCCESS in 1h 30m 58s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/af234240ac2d4db2a7e5679c68eac4c8 : SUCCESS in 58m 33s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/d26c41a5e5e54438a98d71822138253b : FAILURE in 1h 10m 33s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/6995a5bc2226413fb0cafd12124241d0 : SUCCESS in 2h 03m 10s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9c84be19596342d9a4f32b53134a714a : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":4},{"id":"dfd953e3e02b46c29dfd841755a02902b3a6b281","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-06-16 02:36:21.000000000","message":"Patch Set 5: Patch Set 4 was rebased on behalf of \u003cGERRIT_ACCOUNT_9816\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"_revision_number":5},{"id":"4a6b3ea75ee83b7551dc278125f4c6c6a6db9a89","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-16 02:44:15.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbf7a30415d0496f84f3ebdbe06e9e29\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fc725577ba9c45b39bf6a2269934c52c : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3e03c55c6ed84bd4a79e8b9014d7ece1 : SUCCESS in 6m 53s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"71372384da99985c8b4529daa942f730f0208d6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-16 04:36:02.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cbc28794ad749adb9ee1f8b444d54e6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/751bb6c24b044c04bd8ef32ffd3531d5 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8619c879b2c34c73858632220f903667 : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/122c8abaaee0461bae9f4e1db8f17b3e : SUCCESS in 5m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/938d101a32de464cba35d801c416a5a7 : SUCCESS in 5m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a730be8667e1473d924f6372863f1766 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd94a236667948058e74102094d58c11 : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4930843982ef4249839bfb36241a5ba2 : SUCCESS in 5m 35s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/04d44b482ff14a45b0f31724b86c65e9 : SUCCESS in 57m 28s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/eba372832f4d43fa8d41ddee95d8e6a3 : FAILURE in 1h 34m 06s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/06664c2f1f8a461a8cf5a4d771a58d1f : SUCCESS in 1h 46m 13s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/bb7bb17821a9439ba718fc0ade81902b : SUCCESS in 54m 01s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/3469b63b47f84efcb2fe37ad549c4369 : SUCCESS in 1h 55m 20s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/9a608719af464a5b92035a23ea7ac777 : SUCCESS in 1h 26m 14s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/fe39c246c4364c7b8d60dff5d3863530 : SUCCESS in 49m 42s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/e3286da6048349ffb2f6270ac212dfc6 : SUCCESS in 1h 37m 39s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/367a3643889c4ea99b6f8952dc54220d : FAILURE in 1h 13m 14s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/63fcc372774649018c5a1818447a6f65 : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":5},{"id":"de1d138f9c13c0b125f16ef74e3f1c2e1811a133","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-07-08 13:58:48.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1694f8d2fc794ef45f5063a113ac88dca174deb3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-08 14:11:32.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3aa8eaecbbba44cc9d4abb42c29ce87e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/db3298bd99474f37b780f6754a47cb49 : SUCCESS in 11m 04s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8cbe1c9d0e354fbc98be69782a4b9385 : FAILURE in 2m 47s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"fe911f73fcdf2c2dcb5fde4fd44db18328d9d0a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-08 15:40:38.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5b105a665a342c2bad0e28c3883aaff\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9bdf62c96c5448f6893d942c5aa80995 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/200b28bd2d6e4ad1a203855a8516ac4e : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d164738797eb420d8583c455fd8c7362 : SUCCESS in 5m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a75fd6c9fbac465aba82fb69365cd099 : SUCCESS in 4m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fbae26b11e3c4636bb76485071ed5071 : SUCCESS in 8m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b512ddcacaa74fd3b1100f311a32bc58 : SUCCESS in 4m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9afad4c02f7a49f59a96b314b62c0a37 : SUCCESS in 6m 48s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/34800c63209b45c2898b72e1af8d923b : SUCCESS in 48m 13s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/114b689073c34c0b84ea2bfef8a6c4c7 : SUCCESS in 1h 40m 49s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/d47374600efc4c3ab66685d6da816a24 : SUCCESS in 1h 25m 36s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/53458f7ba22f4607b6921503c1926f61 : SUCCESS in 46m 06s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/59b51ad862c04ebab439cbce427ca08f : SUCCESS in 1h 33m 01s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/f4eca32a96284538bf054340547cb7bc : SUCCESS in 1h 12m 45s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/9ab80fe9c13d49788bbf27e15c6ec2e5 : SUCCESS in 49m 13s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/923659f432cb46d28b2a4cb0b8d56f79 : FAILURE in 1h 09m 34s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/2b43b908061e433292ebe4febdac4c2a : SUCCESS in 1h 10m 04s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6ffe9c809cbe4286b9043172a73f316e : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":5},{"id":"f9ce1b2d8cf0def568a57414bd9e1c26a3c7eb54","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-07-09 05:25:12.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"90deab74548fe60dca883222d382d135776f85dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 05:25:40.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"405083f0283d2fd908ecfcc10e2b0f287f68db67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 06:25:48.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d40b5732f384ddd8d511343b0ac2c0c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3bf78cfab2b48688a039ae12cb8ce85 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a6b9853787c4c3694f659114d2af974 : SUCCESS in 4m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4bec2fe320f449b29e96e30c05f230ec : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9eebc40d44f9410bac76ad8eaeea9757 : SUCCESS in 4m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/571f391e8bf9417fb04404eb44b1b910 : SUCCESS in 5m 49s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/39d54c1bccbd41539299af29c1afa582 : SUCCESS in 58m 51s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/3c61f524b6d44b16a9825455e6b8ef0c : SUCCESS in 51m 07s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/5f43bd1b158b4388ba9c7e68892a0719 : SUCCESS in 43m 44s","accounts_in_message":[],"_revision_number":5},{"id":"48d3868746b0a7e39c2fdf153658176c6ced6867","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 06:25:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"10c7498bbdf76a565deaefaa6b516189cb53fed9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 06:27:31.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9581b51961c4bf6b988c470bdf8390a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1319bdbcbe194c29bda2426535e60e1e : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d833086391e74915ae8b1dfd71da9736 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"be9bd3f3a034d9e31b32fa194db8e42ed0ee4155","revisions":{"76d1316cd40a0638ed183eb84822382db7a847bb":{"kind":"REWORK","_number":1,"created":"2023-12-27 01:36:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/65/904365/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/904365/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/904365/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/904365/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/904365/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/904365/1"}}},"commit":{"parents":[{"commit":"69cfc9b081674947e2e91026ac02298379ab7fdd","subject":"[CI] Allow to perform insecure SSL requests to swift","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/69cfc9b081674947e2e91026ac02298379ab7fdd"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-27 01:32:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-27 01:35:48.000000000","tz":540},"subject":"Use policy to determine admin context","message":"Use policy to determine admin context\n\nCurrently the logic to determine the admin context is controlled by\nthe [DEFAULT] admin_roles option but this requires that users set both\npolicy and the option properly.\n\nThis replaces the option by a new policy rule so that users can define\nall policy related settings in the policy.yaml file. This also allows\nusers to create their is_admin rule with not only roles but also with\nother elements.\n\nChange-Id: Ib9d003065f943a5382e1079397bfde4292f5dabb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/76d1316cd40a0638ed183eb84822382db7a847bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/76d1316cd40a0638ed183eb84822382db7a847bb"}]},"branch":"refs/heads/master"},"88f9f5b766b39d4a8d3225d9b4c196fe7a921503":{"kind":"REWORK","_number":2,"created":"2024-01-02 13:26:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/65/904365/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/904365/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/904365/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/904365/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/904365/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/904365/2"}}},"commit":{"parents":[{"commit":"69cfc9b081674947e2e91026ac02298379ab7fdd","subject":"[CI] Allow to perform insecure SSL requests to swift","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/69cfc9b081674947e2e91026ac02298379ab7fdd"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-27 01:32:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-02 13:26:06.000000000","tz":540},"subject":"Use policy to determine admin context","message":"Use policy to determine admin context\n\nCurrently the logic to determine the admin context is controlled by\nthe [DEFAULT] admin_roles option but this requires that users set both\npolicy and the option properly.\n\nThis replaces the option by a new policy rule so that users can define\nall policy related settings in the policy.yaml file. This also allows\nusers to create their is_admin rule with not only roles but also with\nother elements.\n\nChange-Id: Ib9d003065f943a5382e1079397bfde4292f5dabb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/88f9f5b766b39d4a8d3225d9b4c196fe7a921503"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/88f9f5b766b39d4a8d3225d9b4c196fe7a921503"}]},"branch":"refs/heads/master"},"375b808ea1445851461b17d3ccd92044baed6dc1":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-04-02 06:39:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/65/904365/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/904365/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/904365/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/904365/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/904365/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/904365/3"}}},"commit":{"parents":[{"commit":"d702a30db431e089267815a01f6cb2f3c57d0e11","subject":"Merge \"Fix error can\u0027t detach the configuration group binding\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d702a30db431e089267815a01f6cb2f3c57d0e11"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-27 01:32:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-02 06:39:09.000000000","tz":0},"subject":"Use policy to determine admin context","message":"Use policy to determine admin context\n\nCurrently the logic to determine the admin context is controlled by\nthe [DEFAULT] admin_roles option but this requires that users set both\npolicy and the option properly.\n\nThis replaces the option by a new policy rule so that users can define\nall policy related settings in the policy.yaml file. This also allows\nusers to create their is_admin rule with not only roles but also with\nother elements.\n\nChange-Id: Ib9d003065f943a5382e1079397bfde4292f5dabb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/375b808ea1445851461b17d3ccd92044baed6dc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/375b808ea1445851461b17d3ccd92044baed6dc1"}]},"branch":"refs/heads/master","description":"Rebase"},"de7b26d4975e6f9256d4480aa979f75de7290c49":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-04-28 01:06:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/65/904365/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/904365/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/904365/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/904365/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/904365/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/904365/4"}}},"commit":{"parents":[{"commit":"397d948ad575252c4e79ce524e58655a65f4933b","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/397d948ad575252c4e79ce524e58655a65f4933b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-27 01:32:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-28 01:06:14.000000000","tz":0},"subject":"Use policy to determine admin context","message":"Use policy to determine admin context\n\nCurrently the logic to determine the admin context is controlled by\nthe [DEFAULT] admin_roles option but this requires that users set both\npolicy and the option properly.\n\nThis replaces the option by a new policy rule so that users can define\nall policy related settings in the policy.yaml file. This also allows\nusers to create their is_admin rule with not only roles but also with\nother elements.\n\nChange-Id: Ib9d003065f943a5382e1079397bfde4292f5dabb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/de7b26d4975e6f9256d4480aa979f75de7290c49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/de7b26d4975e6f9256d4480aa979f75de7290c49"}]},"branch":"refs/heads/master","description":"Rebase"},"be9bd3f3a034d9e31b32fa194db8e42ed0ee4155":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-06-16 02:36:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/65/904365/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/904365/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/904365/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/904365/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/904365/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/904365/5"}}},"commit":{"parents":[{"commit":"114601edba9fc736af43a3f2705552d8c746115c","subject":"Introduces Alembic database migration tool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/114601edba9fc736af43a3f2705552d8c746115c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-27 01:32:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-06-16 02:36:21.000000000","tz":0},"subject":"Use policy to determine admin context","message":"Use policy to determine admin context\n\nCurrently the logic to determine the admin context is controlled by\nthe [DEFAULT] admin_roles option but this requires that users set both\npolicy and the option properly.\n\nThis replaces the option by a new policy rule so that users can define\nall policy related settings in the policy.yaml file. This also allows\nusers to create their is_admin rule with not only roles but also with\nother elements.\n\nChange-Id: Ib9d003065f943a5382e1079397bfde4292f5dabb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/be9bd3f3a034d9e31b32fa194db8e42ed0ee4155"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/be9bd3f3a034d9e31b32fa194db8e42ed0ee4155"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
