)]}'
{"id":"openstack%2Ftrove~904692","triplet_id":"openstack%2Ftrove~master~I9d0b62e62da38a83217eec68f0b74c6be40aae87","project":"openstack/trove","branch":"master","topic":"py3.11","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-01-05 04:18:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9d0b62e62da38a83217eec68f0b74c6be40aae87","subject":"Update python classifier in setup.cfg","status":"MERGED","created":"2024-01-04 09:09:58.000000000","updated":"2024-01-05 04:19:13.000000000","submitted":"2024-01-05 04:18:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"904692-py3.11","meta_rev_id":"39978350b8928835dfdf1a5637777d7bf274a413","_number":904692,"virtual_id_number":904692,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-05 04:18:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-05 03:19:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-05 03:19:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-04 09:23:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-04 12:06:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-05 03:19:51.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"c181176f1860a3ea5e6ef60f376650121704bfb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-01-04 09:09:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de13a2ac52533c59224716926131bde26bc239fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-04 09:23:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/088598f0c7ed415dbc1ecb8eef69eeb2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/45e1642432804e34b506da66c497a032 : SUCCESS in 10m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ea560bc2ac8048b78257bb887baf4bf2 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5f4de027c41a41df991cfc487a851262 : SUCCESS in 11m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"392a0571936f3492aee1599f71f1796c7f917dd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-04 12:06:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/deffb0dbd07c49cab355df0cf2c1acde\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/eafbc882d40646c48e7ef8a111f71c69 : SUCCESS in 1m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0c469ae23b942b38014817f99681dac : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b57f747715443a1bf25bda0b131fb9e : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42bb0a180a234913b300f4add2eba3d7 : SUCCESS in 5m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/12d60574c3a04a29a622315b312e05b7 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/454549e259ba4b3380cc63a6b9ecd1b5 : SUCCESS in 5m 45s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/e59405b7f133446198af39fc3bb2567c : SUCCESS in 1h 00m 24s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/df02b063671e462cbd38d1c47697fd15 : SUCCESS in 2h 26m 47s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/54503e72351e45998ad7fbf1a80d2f7e : SUCCESS in 1h 59m 34s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/61570136c4e2498b9bc47d7e3d723ec2 : SUCCESS in 42m 15s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/5a689c3f6fe5448d860f276f15de3fce : SUCCESS in 2h 08m 42s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/0276072013b44d1ea888d9693653d97e : SUCCESS in 1h 12m 04s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/4ba4639d1e7e476c899c6a01aa71edc4 : SUCCESS in 1h 03m 33s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/6c884fc4632b495e92c374d5e2bc0fd4 : SUCCESS in 2h 11m 26s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/d5391661f5154915ab6b00d0bbece2f5 : SUCCESS in 1h 12m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c992b387562bb560b95906484c73299fb3b46267","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-01-05 03:19:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"652d5578cae5d4f36e784a53cace4959c0375a20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 03:20:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1a5fd21a5bd6da25d9adc39a9f83f9e705f4e6a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 04:18:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f49a38e00ae43e296ab8aa2279bedae\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5b4dcb59fc374d029991d3c2a96e98d2 : SUCCESS in 2m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e69c98177ab04e73bbf2fb5a0f2db776 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49e4d1010567459d8a0aeebbe019863a : SUCCESS in 7m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/121ee996aa3b47dabdbdd619737bffa8 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95740e3b10e04562b7e63bbd94a43652 : SUCCESS in 6m 04s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/dbc1f57594534112a8b2b6d95697f0d7 : SUCCESS in 55m 33s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/f582a0ae73f24ca69299a4f04cc5785b : SUCCESS in 56m 05s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/e02c6cfdb08d44a1bc2430fc5aeb4573 : SUCCESS in 54m 06s","accounts_in_message":[],"_revision_number":1},{"id":"65da80c206c09609594f635cb39b06be5a7f268d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 04:18:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"39978350b8928835dfdf1a5637777d7bf274a413","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 04:19:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a02d3fed7d764080834eb46e974d53f1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd91eef874984eedb638978f0ff353e0 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ca28382ed9a97bac9e10e2333fb5328ebaa266c6","revisions":{"ca28382ed9a97bac9e10e2333fb5328ebaa266c6":{"kind":"REWORK","_number":1,"created":"2024-01-04 09:09:58.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/92/904692/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/92/904692/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/92/904692/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/92/904692/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/92/904692/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/92/904692/1"}}},"commit":{"parents":[{"commit":"f8150f82a31c6f86239bc8bcd1e9e5e5d3389189","subject":"Merge \"Disables shell\u003dTrue in backup and restore\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f8150f82a31c6f86239bc8bcd1e9e5e5d3389189"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-01-04 08:51:54.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-01-04 09:09:56.000000000","tz":-480},"subject":"Update python classifier in setup.cfg","message":"Update python classifier in setup.cfg\n\nAs per the current release tested runtime, we test\npython version from 3.8 to 3.11 so updating the\nsame in python classifier in setup.cfg\n\nChange-Id: I9d0b62e62da38a83217eec68f0b74c6be40aae87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ca28382ed9a97bac9e10e2333fb5328ebaa266c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ca28382ed9a97bac9e10e2333fb5328ebaa266c6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
