)]}'
{"id":"openstack%2Ftrove~904961","triplet_id":"openstack%2Ftrove~master~Idb0366ad1d73c85126e18fccd58f1e52176b63de","project":"openstack/trove","branch":"master","topic":"fix-mysql-socket-patch","attention_set":{},"removed_from_attention_set":{"28691":{"account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"last_update":"2024-01-12 07:23:23.000000000","reason":"\u003cGERRIT_ACCOUNT_28691\u003e replied on the change","reason_account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2024-01-19 01:04:18.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"Idb0366ad1d73c85126e18fccd58f1e52176b63de","subject":"Change mysql socket path to a persistent path.","status":"MERGED","created":"2024-01-08 10:05:07.000000000","updated":"2024-01-19 04:53:08.000000000","submitted":"2024-01-19 04:52:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"904961-fix-mysql-socket-patch","meta_rev_id":"dbb973da20d408d10d0a4ee5f4eabf885c25c46b","_number":904961,"virtual_id_number":904961,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-19 04:52:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"recommended":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"all":[{"value":1,"date":"2024-01-12 07:23:23.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-18 13:00:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-18 13:00:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-08 10:19:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-08 12:24:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-10 06:26:53.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"REVIEWER"}],"messages":[{"id":"4648eb95ad6e25465ddd9544d181987a4f24912d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-01-08 10:05:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"adedc7ba19ead2da676e0cbcf4a6469f8d1ff8c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-01-08 10:06:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7079146997b4a48b08d87c508c66d34185ef6454","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-08 10:19:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35eb3cb6342846beb97f41530d1aabfc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6d9d7c6561854b52903b0430652ec658 : SUCCESS in 9m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7252daf64b814672986790b67222e376 : SUCCESS in 11m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ddfb49d956a54b34b2332001101a0d67 : SUCCESS in 4m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"322d7013657ef252375c4005b22270bb97d14f0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-08 12:24:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f67635fa2b143479c414eaf2367ce3f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/15fe7129ab3b4c3691cbbae32f91e2ba : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5d2c44dd4a545aa95e223dde0aef5f8 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98ff5aed24ce4a57bbe0dcf5ad93b93e : SUCCESS in 5m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6ddfdbb81b5040129e0f1d91c730065b : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83916b38343b45f2b3474f13f00d4913 : SUCCESS in 5m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99f8bb07948b4771948ca58567f2edc2 : SUCCESS in 4m 36s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/81c4dde794b047549a31dbd7fba46c89 : SUCCESS in 1h 07m 04s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/9b373a8c686d4cb8821d6a7e6f81e6fa : FAILURE in 1h 50m 50s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/cb16c9734f1644768f7b614e3fa98419 : SUCCESS in 1h 55m 18s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/19f9b6d4555e459fb71ccab60b9c4992 : SUCCESS in 48m 57s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/a854a49f669c4c3195986a7916d78d35 : SUCCESS in 2h 15m 33s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/03af62a651a043a1bbc073cb85f96eb8 : SUCCESS in 1h 33m 21s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/336fa71761424632b091c694a0c710f5 : SUCCESS in 43m 31s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/21428659ea754c3b87ddcad17dda8210 : SUCCESS in 2h 08m 29s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/a55b9bfebacd4d4d930fea07fcae5c6c : SUCCESS in 1h 45m 58s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1aefd8dae43847a28b451a28ca5f9dd2 : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":2},{"id":"65def49cdc85c6e3245b5a5985d0c293eb4f5302","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-01-09 02:48:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"962fb9de5ecfc50ee5941271158bbbb624aa2008","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 02:57:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea2fd7b43ea142958a6ca1eca507363e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cd6813320693403d9f8cb7d6ca16536e : SUCCESS in 7m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/011d626e54e544d4bb0a0cdac05e9339 : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/174f8299ebd747499d1356eef29b139c : SUCCESS in 5m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"00a7a8d6461bf491bbefda070cf82fd387e54e54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 04:25:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db2ef5f44f5c41fb99dd4f5a5cf21601\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42d73d7d02fa497cb44501e6dd2c6c56 : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d641ba89017459f8c8588d34a599a44 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ff628d9b8dc43c495dcbfc526895f0d : SUCCESS in 6m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f6d9184e3446452aa3949817a890146a : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87f20b28bb62456c93a1dda9aec88077 : SUCCESS in 8m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ebc0588a1e74ff88fec4f0129275cda : SUCCESS in 7m 23s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/dfc4baf19a3e4e68b95eef803ac777ab : SUCCESS in 43m 55s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/59b0711cb42648a08586898c3425da63 : SUCCESS in 1h 35m 38s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/953f28b3715a4f17ad451f598b1a767f : SUCCESS in 1h 22m 52s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/7e23a8cb81ac44daa9b5dc025db44226 : SUCCESS in 42m 40s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/0c9cade31b084169849ea0505aac42a2 : SUCCESS in 1h 34m 43s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/0b993ad65e5742488236f62d366d2686 : SUCCESS in 1h 19m 50s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/debcd9936543423592fd358f46e9ec35 : SUCCESS in 44m 23s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/c1c9a5c1f47a486dac008e3fb0bb4230 : SUCCESS in 1h 27m 28s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/adaf7b729e5a40f288b0ef164ece4d02 : SUCCESS in 1h 16m 05s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/30423660f6d347f5aff50f2a1d0ac9f1 : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":2},{"id":"69c4b50ba09022b787655978b4b117571f538cc4","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-01-10 06:26:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ad85316274a256fac3d0feb82c5c9fec92c18812","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-01-11 02:18:21.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"edade7b94507fd45ac5a6710dda7aa06401029ea","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-01-11 03:20:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"33ba4497fe4a90abc2c57a758b27a64c63df9512","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-01-11 03:26:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c7570e5bd531a32f78b39613e6870bec744a19ab","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-01-12 07:23:23.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2bba4b32f163faccee187b17c4449fd23a814b09","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-01-18 13:00:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7722dde0734e8c70a19865c5628f6c0597846c97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 13:00:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a6c3bf937a58546588f8a25fa77deafaa7900f03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 14:42:28.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/806b65fa2cae4da3a54734858ce59045\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65f04582228640428700fee206ba7af3 : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b5f958cfd834178b82c51c5ae7b5ef4 : SUCCESS in 6m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3ec322c2e20040a89e81f712e18dc76e : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c713b9cd7fa943a785b32e937dff8838 : SUCCESS in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6617472a0e1849c593af49a148042dc0 : SUCCESS in 5m 29s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/ffdfdbbd76ee4f27a774a89a84c7395e : FAILURE in 1h 09m 39s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/badcffd57b3c4ffbbdd6d1a3e550fcd8 : SUCCESS in 1h 04m 17s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/abc3fc9c59b94e2690b98e54f6e89090 : FAILURE in 1h 40m 38s","accounts_in_message":[],"_revision_number":2},{"id":"a5ca9fe5f05c190f0dc5a97d3dc2a90c3081f167","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-01-19 01:04:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c962b6e80d9d5c75c37cce3c409e6c26cddffb75","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 01:16:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5d88ce762244cf1af8c886a0af13916\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c5fd143eac4041298b55ecf26b22d851 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0c11c1e58f5440c1ae8c17a0463b01d1 : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/485325ee0c3b40da8923d8f35edfc221 : SUCCESS in 10m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b77db78771b25179507261c55470c2ef2b88977b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 03:42:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f543619fd68b457baadf1cdba869ef52\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2d577cd752d467d8733fe6a354ca48b : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16e5d03f27104ab2aafc83d05fefddcb : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd59800dad2a4417a90c618484f9f6ec : SUCCESS in 5m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b6e956f9606a46528302bb6fa7c22979 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18ed6daba41f41608c080ec61ff0fb2d : SUCCESS in 8m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ee97745225640ca9dac58d9aa4858c9 : SUCCESS in 5m 31s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/9c83e8f4c2f746cc8e2be2e5e4591124 : SUCCESS in 53m 49s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/1e9cdead8d2c434ea8d803c4a595679e : SUCCESS in 2h 22m 52s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/e46d2721e35d485693dac0306f50b035 : SUCCESS in 2h 08m 46s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/8e29d068c7fb43ad8cffac4e5dd6afa5 : SUCCESS in 1h 06m 10s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/9ff88657280a4cd18c2b44f631dd73ef : SUCCESS in 2h 35m 34s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/ee0cf7053d324ddeaa3c63a04fe524e8 : SUCCESS in 2h 10m 45s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/187ef778e6bc4e4d893fe526b11faea7 : SUCCESS in 1h 10m 47s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/3ae4923e6e0344689b111596b135de3b : SUCCESS in 1h 23m 46s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/cb5b69eef5604afcb672a33504dd2a69 : SUCCESS in 1h 13m 19s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cb8af3f4ab0e4caf8da280e3033aff25 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":2},{"id":"88a74866cc256a0b2b5b12d131f1bb602f563ea6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 03:43:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2163a76665b18ec0a0ac731bc4dee2085002273a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 04:52:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc205904a9924bf484a65e49257389ca\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe87b7854a294232b8c23a1f54611900 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c562af2cf02422db9f0c64813eef1dd : SUCCESS in 6m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/426c03341e3b407997d56ca8e29cf813 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d6c3b62813f40f6bc3f55789eba7592 : SUCCESS in 5m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cec097e2fbc14b37bdbbf5f35366d316 : SUCCESS in 5m 59s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/7ef6275a6f16420dbd0f144efba13f3b : SUCCESS in 1h 01m 05s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/624ed8ee68204fe2a5044139ce1d5bb5 : SUCCESS in 53m 53s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/815cddd89eb44d939fd7e67edd6c08a4 : SUCCESS in 1h 07m 26s","accounts_in_message":[],"_revision_number":2},{"id":"7f6d7d25a015544ed399349ab71ad2293c703060","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 04:52:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"dbb973da20d408d10d0a4ee5f4eabf885c25c46b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 04:53:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4413dd63f0e44fcaab01c031439c391\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8d6192449d14c3d96830e0cc191953f : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa18a206657b4a359d7b61bab65a5b1a : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9d31e46f00e5f07ef0f603469dbda98f0196aa14","revisions":{"8148d8ffaa324555d70f5aa5fa20da3333dc92d4":{"kind":"REWORK","_number":1,"created":"2024-01-08 10:05:07.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/61/904961/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/904961/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/904961/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/904961/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/904961/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/904961/1"}}},"commit":{"parents":[{"commit":"ca28382ed9a97bac9e10e2333fb5328ebaa266c6","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ca28382ed9a97bac9e10e2333fb5328ebaa266c6"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-01-08 08:34:46.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-01-08 10:02:30.000000000","tz":480},"subject":"Change mysql socket path to a persistent path.","message":"Change mysql socket path to a persistent path.\n\nThis patch changes the default socket patch on the host from\n\"/var/run/mysqld\" to \"/var/lib/mysqld\". ensuring that the\ndatabase container starts properly after a Nova instance is restarted.\n\nStory: 2010599\nTask: 47471\n\nChange-Id: Idb0366ad1d73c85126e18fccd58f1e52176b63de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8148d8ffaa324555d70f5aa5fa20da3333dc92d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8148d8ffaa324555d70f5aa5fa20da3333dc92d4"}]},"branch":"refs/heads/master"},"9d31e46f00e5f07ef0f603469dbda98f0196aa14":{"kind":"REWORK","_number":2,"created":"2024-01-08 10:06:50.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/61/904961/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/904961/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/904961/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/904961/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/904961/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/904961/2"}}},"commit":{"parents":[{"commit":"ca28382ed9a97bac9e10e2333fb5328ebaa266c6","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ca28382ed9a97bac9e10e2333fb5328ebaa266c6"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-01-08 08:34:46.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-01-08 10:06:43.000000000","tz":480},"subject":"Change mysql socket path to a persistent path.","message":"Change mysql socket path to a persistent path.\n\nThis patch changes the default socket patch on the host from\n\"/var/run/mysqld\" to \"/var/lib/mysqld\". ensuring that the\ndatabase container starts properly after a Nova instance is restarted.\n\nStory: 2010599\nTask: 47471\n\nChange-Id: Idb0366ad1d73c85126e18fccd58f1e52176b63de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9d31e46f00e5f07ef0f603469dbda98f0196aa14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9d31e46f00e5f07ef0f603469dbda98f0196aa14"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
