)]}'
{"id":"openstack%2Ftrove~905773","triplet_id":"openstack%2Ftrove~master~I54a53cc802bbd20bb9e2c6508919e4c8dda38f2a","project":"openstack/trove","branch":"master","topic":"postgresql-socket","attention_set":{},"removed_from_attention_set":{"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2024-02-22 14:11:39.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"I54a53cc802bbd20bb9e2c6508919e4c8dda38f2a","subject":"Change postgresql socket path to a persistent path","status":"MERGED","created":"2024-01-23 14:30:48.000000000","updated":"2024-02-23 13:23:59.000000000","submitted":"2024-02-23 13:23:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905773-postgresql-socket","meta_rev_id":"c4efe98fc3c226c8a71355c550a0398522b4f3c6","_number":905773,"virtual_id_number":905773,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-23 13:23:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-02-23 12:13:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-02-23 12:13:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-23 14:40:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-23 16:29:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2284755eab55d4c8b6d223057f74aa9c0d5d18df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-01-23 14:30:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70956b80fa963ea3dcb4e8fcaaa87495a488b62d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 14:40:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e51dd5cbc97547b7b7756a451d8220ee\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/56f988ec821441959e5de27f3a5b2c2b : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dd6cc1b318a0492e9a7b80ce2014540c : SUCCESS in 8m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/dbfdc05feff047ec8b97fb699777729d : SUCCESS in 6m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d235cbc44682c46bbf3e7ff51df32c71637c8bdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 16:29:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/315e8e49e9f14ddaaab210280ddc4cc8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/290e2a83901e4558bd4826e20122b15d : SUCCESS in 7m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86acff9f7afc4407998a8c831ff52f37 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73ae0a8de02244d1b843841f44e1e1f3 : SUCCESS in 5m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d4320930aaf24f22a7e3fe96bfc8e66f : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b43881c2d1747afa0a54911fba31e18 : SUCCESS in 5m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/160a53e8a7e74467bfefde7da1e4f1cd : SUCCESS in 5m 23s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/902a0d4437ae47c887d89625598a883b : SUCCESS in 46m 31s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/b33b020edf8a473fb467acf3fd0f08a8 : FAILURE in 1h 26m 48s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/c529d2353a2542019966923f35d4905d : SUCCESS in 1h 25m 00s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/76d7cef3e1664f64b97b3459de895644 : SUCCESS in 41m 29s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/7374fd90f1454ccb8197d09f7c558b00 : SUCCESS in 1h 30m 50s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/0624c0f8abdf42d29dc95c2c519bd776 : SUCCESS in 1h 57m 25s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/3f5d5afd036b43bc8ff61a677644dfbb : SUCCESS in 1h 03m 40s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/8c22f1e10816458d904778dcc4110a7e : FAILURE in 57m 15s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/b472bc603a454093a3e15b7c11d4b7ac : FAILURE in 1h 01m 40s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/01e6052dfceb4997a3ba1da1d37881ac : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":1},{"id":"70dbb176713903eea3c4e9f8324325d549ea51c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-01-24 07:34:07.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"7666a24056c7d89bff6293448a7210d691aceacc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 07:51:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b64d79cba4bc46b281ac5392f256b5de\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a8652ef1aaea4589ab1330aa6134bfba : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5acce96324bf418ca0e582aee7fc4b3c : SUCCESS in 15m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6474000dcbf749129724197e28c474ba : SUCCESS in 5m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dfadd768ea7ffd59472abdacd0f1e29811fc38e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 08:29:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c386eafeb30b400489066326631239dd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8c74781fa6c4f7eb4564a88aa37b897 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46d073eca88044b398238af0f1aee40d : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20007f7728144db491eb0465fe2dd161 : SUCCESS in 5m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/90c92913d2044a82a5059b6899081496 : SUCCESS in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8493b1a9a2c640dcbedaef90c0de7b69 : SUCCESS in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1000f9597812402588aa7094cbfe17f2 : SUCCESS in 6m 01s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/4a3ede88245849a6830405173798dcf4 : FAILURE in 32m 03s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/bf521abc184147aabdb8204a757a81a4 : FAILURE in 53m 32s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/c4c78fc3a2a7440581171059d3960b8b : FAILURE in 49m 23s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/8b5764df737241cd8b25306204398563 : FAILURE in 29m 52s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/23742a0062bd4410b8795feb1100a0e7 : FAILURE in 50m 14s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/bbdc572082d54c2eac634570335988f4 : FAILURE in 39m 14s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/ba6990127d0649c7ac5a47e2dcd00bfc : SUCCESS in 41m 39s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/0bfba76658af49c992c77138d66f9ffb : FAILURE in 52m 10s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/3c29a9ed42444382a415e60db22c72fb : FAILURE in 39m 44s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/00388e8e15744393a158a04678e9598d : FAILURE in 5m 00s","accounts_in_message":[],"_revision_number":2},{"id":"541f2188ba63a0d926d4062a10b678aeee0f09b2","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-01-25 00:58:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"73444d70aa12694780fd84117dce453260e29d8f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 01:06:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d95d8ee74f64d789e896cb77b6acc56\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7da69cb6ffc34202bc9959f2286db92c : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/55d2bebeb72c4c50b912982be7bce605 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/76621524a12f47a38230bef19efdebcd : SUCCESS in 6m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"477089ab39845623eaf09ddb64d24a803fe1f446","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 03:17:01.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee2ce8d5a2994cbcacfd8db534f3c159\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef8e96fe8c4b40d4be04ad3f25cacf3e : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de89d70252644d028a95cf2259e47c9c : SUCCESS in 7m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb7d1db376734b349dca48fef9ddfee6 : SUCCESS in 6m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3ead30783e4848e1a7fbb55b4d5f4d59 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1af3486bc6864c9fb5927d906a85a710 : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a761069bcc2148e4aa3e5ef318ed35b9 : SUCCESS in 4m 24s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/423777931eda4cabb1c73a07a6c54b5a : SUCCESS in 1h 05m 45s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/5c315e7e78f742cda1762303c8f65f27 : FAILURE in 1h 15m 31s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/c2896fe2ba404467aada71df42cdad93 : FAILURE in 1h 20m 10s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/bfe201c82ea74c0ba7ca531cb45cada2 : SUCCESS in 53m 57s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/5df3b681d2d145cea9ad708a6ffc7096 : SUCCESS in 2h 16m 22s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/8986a1fd30fc41508c0a94034c40e93a : FAILURE in 1h 09m 36s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/7efb73571e444deebf6183cd265d8fc3 : SUCCESS in 1h 02m 50s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/ebfcafa5809d43a7b11ec5cb36d22dc3 : FAILURE in 1h 21m 01s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/bf636aa7f467454baf337d8ad0101039 : FAILURE in 1h 19m 47s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d928dc29c75f4eee95db6834e8086064 : FAILURE in 4m 07s","accounts_in_message":[],"_revision_number":2},{"id":"edf6c319f85137f3c0722d82d1bbb4bf9aba6bb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-01-25 07:20:00.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"6c31c52de110f6afccda7e7c46e598bfce12b484","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 07:36:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c33167efde04080a738099246ff0dcf\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/97241cf7ed5f4abbbc13d255dc1ba7e0 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e429080137df4b0faa3a34d093efcb20 : SUCCESS in 14m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cd0057a23c1e4d9da913a2237ace24b7 : SUCCESS in 4m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"de67a56ef6a50ae51845440d8b728cf2459e73b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 08:58:32.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e35c10d8bc74236b4771ef092b1edcf\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d45a18f5f26a43d0837839e9b01c1d24 : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0629177853c04988822f80f0303b0725 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20c9c34cdc614802a2d1915d03ca40c6 : SUCCESS in 5m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/23ff316a0a714ed0bdf05b5f0a2d781e : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be515aaa65164d20ab01f78c0fc47540 : SUCCESS in 5m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5aeb4e8248794d4bb79b2ef603dd71d3 : SUCCESS in 4m 19s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/a5eef33ad9c347bd937f03ef9168c247 : SUCCESS in 42m 42s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/2e7beccc2c19436a82de5d6266490960 : SUCCESS in 1h 33m 47s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/6aaf8956d6bd45d88332191d1edef4a0 : SUCCESS in 1h 20m 25s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/ee94ae23c0f1429892f9a47a1b180bb7 : SUCCESS in 43m 30s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/1dc11110929b4130a4cf63f948d218d1 : SUCCESS in 1h 30m 34s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/ddc0c8507d7c449fa4220d06a373d67b : SUCCESS in 1h 15m 47s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/a05f87da0db048b4984a4c47072a1a71 : SUCCESS in 41m 37s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/c466387f07d343c3bd664901754856ce : SUCCESS in 1h 25m 41s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/868c2cb1cb744a3eaca25e6f625762ae : SUCCESS in 1h 36m 36s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/52fe9ae847274ffa9fa016fb9609aa98 : FAILURE in 4m 58s","accounts_in_message":[],"_revision_number":3},{"id":"03dc2c5766bd8e5f249f99be6006d9dd597ba2c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-01-30 03:08:55.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"cb212e825c92d0549ea8f288d2c5d8d1ae8ace05","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-30 03:22:14.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35d795a924b0452a873628ededf33665\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/14d6b3d901644f1a8e7b0074fe494f92 : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9239774411354ab294b805acf1e2e4d3 : SUCCESS in 11m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c73ebd5236e948759585373ae2c4846d : SUCCESS in 10m 50s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a05ade1b6c2ce190d7c7ae05841acbe24bf73447","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-30 05:28:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c859e57694b419cbc873564672f3281\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf984dcd48214b52b800bfd3d69e2bb3 : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ae19854a56242968aaacba1ef741461 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1865236f21e4b49ba808f0e00a4deb3 : SUCCESS in 6m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8ac9cb1effb2406cbc2971f405b6e43e : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c13469dead8a44c79ab443e6dfda6d84 : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fee3f3daf0ad4c2aa6840a3549e6ddab : SUCCESS in 5m 45s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/9137f1cafdad41fa9cfe8731a8d850d1 : SUCCESS in 52m 02s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/cd893ad4c820475193188a34a699542d : SUCCESS in 1h 32m 22s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/bf50d537483f40a5b958bc7e277504c2 : SUCCESS in 1h 18m 29s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/ddf1f0ebbcd24934b510211f747fd22c : SUCCESS in 39m 08s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/11fe1c98f56e4f98bc1d28deefe14f7f : SUCCESS in 1h 47m 59s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/ca9d970da30e4770ac10e4df7819e03c : SUCCESS in 1h 15m 11s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/a7e6621571f741c3b9b83e4d4683fe15 : SUCCESS in 59m 57s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/a7cdd0b912d543c391ad0d9c4e6d56ee : SUCCESS in 1h 53m 00s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/fe4644935be64423b2d42e04cc181dae : FAILURE in 1h 15m 33s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cfb56d19c49f41d68ab94d637ec52b43 : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":4},{"id":"5672278728cb782af327890277c149c1055b49ce","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-02-22 14:11:39.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0971762242c39ca3ba2677c2df930fbd936e542f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 14:24:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4acaf525ddff45d68f8730d8723b44fe\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b4920d2bd9ca4252836af2418c92fd7e : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/67871f2ae41f4fb7b291bb73af444467 : SUCCESS in 10m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e52f01b485414e1d8bfcf50405717f41 : SUCCESS in 5m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bf25a05fdb809e179bc950eb47fd7d4a3d50ac40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 16:21:38.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07870afb39ea4a298a1d5cdd4a95fce9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf430469e90341339f0d51a419af4d5c : SUCCESS in 8m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4df7c59d3c4642f4a8ec7067c890471a : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1361208e9f04a41bdad40af1eb4bbd4 : SUCCESS in 7m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fd2e49f8471a45f096ccdd30869e9d55 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d30f375ecbb648bbaa49e823fef02970 : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd816ea3114340dc988d8114b5fdf417 : SUCCESS in 6m 54s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/1035b36c32c94a6589b429b4ac21b9e9 : SUCCESS in 55m 01s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/2304ea4e3b7c4ba9999db9a55ac7e32b : SUCCESS in 2h 09m 23s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/9f3c645e22c24055a9cea2262729ed64 : SUCCESS in 2h 01m 47s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/c32e0846b62545609727a74ca05693bb : SUCCESS in 57m 29s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/7dc9b9d612664408b1bd8dd733b2eab2 : SUCCESS in 1h 27m 30s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/34f313f0f2ac42638cb454050a9e6dd4 : SUCCESS in 1h 14m 18s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/a0baf9e6adc2470cbd068c0072081203 : SUCCESS in 46m 16s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/6a69df94e4c949d9bb53417a8af4aa00 : FAILURE in 1h 26m 24s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/5a4e7ceeab7c4892aea77e7fff586d04 : FAILURE in 1h 17m 28s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d2c12521c8bb4309b289b0cac6c4833e : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":4},{"id":"b6f7f98ad1f61ad9e429986dcb05905417db7e9a","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-02-23 02:13:51.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c736360f83f96da7a5998820c643f655d5d0d213","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 02:21:53.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc35c3a32b664b189e5d0c128a3aef91\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0385cb21f2754921be1edd7599012dbc : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a4a80e847d854a49825c7a78784ff86b : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/455f125676a44ff290d0ff0731f6c301 : SUCCESS in 5m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"889419eaf4101181a4954dfaf8170d136e2377fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 04:22:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fbc5519c1bb42d18b89f9bcf67a0bb1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b26df3cb793344529b5d79d7b2d13147 : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7aec7ecfab5c48829b83200ca58be6e6 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3ec66d021b848a994380b620aab583c : SUCCESS in 5m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cd03b80522d242f3be76819dc2dc058a : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75ad9cbc6a27463da8b82734de31de51 : SUCCESS in 10m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/471e755c5a56487ab106994d0e096c6a : SUCCESS in 6m 10s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/be54010c2b5b4bd3b1f93dd918a7d854 : SUCCESS in 43m 06s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/de54b2365b104e6788c729d7aa72ddd9 : SUCCESS in 1h 32m 45s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/3791b605129e4437af7c450041626cd0 : SUCCESS in 1h 39m 37s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/3552aece5d44492292c3117ca65d288e : SUCCESS in 57m 01s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/3af0fb19f5854ed0b404ceb77d1d8a2a : SUCCESS in 1h 31m 09s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/6152a4c70ab5425f92033427beedafd0 : SUCCESS in 1h 15m 15s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/c5ab85dbc5d443e0bbab2895c6f2e874 : SUCCESS in 1h 05m 18s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/d2d9ac4c136e4d61ac1c74f5e8df97a7 : SUCCESS in 1h 31m 37s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/78ad5748561c45548eae40813a2bcd6a : SUCCESS in 1h 30m 05s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0da009eb0fc94466bb00adfda93fab0d : SUCCESS in 6m 49s","accounts_in_message":[],"_revision_number":4},{"id":"de2a991a3b61f180dfb52a705fb1726ba7a980e8","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-02-23 12:13:57.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"345c93825f077cad2e5095054501f5c15ab281a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 12:14:28.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"79a88ca02e3430d69631d047f16f93de1fb95594","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 13:23:59.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa430dfb1e274b9c909709e8c7ceb19a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7fdf3e6223d429587414ec9ef5458dd : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af3558531b2f4422a613a32780dac3da : SUCCESS in 6m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4a855ac218fd4460a389f8ba9b4c590b : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/869b3e2917944fc8b6d1a9cb31aa37a6 : SUCCESS in 5m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/793e98d6f02a4b42977cc2603ecaba46 : SUCCESS in 5m 00s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/fec476c2817346caa33cb27645eafb93 : SUCCESS in 43m 10s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b8d250135d93472a9c42f4ce06c75126 : SUCCESS in 41m 13s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/741322f96b834e6680e18665351ea93d : SUCCESS in 43m 47s","accounts_in_message":[],"_revision_number":4},{"id":"c4efe98fc3c226c8a71355c550a0398522b4f3c6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 13:23:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f151fd91d9f1a5e82b46a00bb7e3baae2b498161","revisions":{"30b90154effdbc0e0197396ec4b3f969916e7938":{"kind":"REWORK","_number":1,"created":"2024-01-23 14:30:48.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/73/905773/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/73/905773/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/73/905773/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/73/905773/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/73/905773/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/73/905773/1"}}},"commit":{"parents":[{"commit":"40fdb7b44fb33e022b77ba8df63fde2565c70dea","subject":"Merge \"Change mysql socket path to a persistent path.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/40fdb7b44fb33e022b77ba8df63fde2565c70dea"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-01-23 13:49:35.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-01-23 14:30:37.000000000","tz":480},"subject":"Change postgresql socket path to a persistent path","message":"Change postgresql socket path to a persistent path\n\nThis patch changes the default socket patch on the host from\n\"/var/run/postgresql\" to \"/var/lib/postgresql-socket\".\nensuring that the  database container starts properly after\na Nova instance is restarted.\n\nStory: 2010599\nTask: 47471\n\nChange-Id: I54a53cc802bbd20bb9e2c6508919e4c8dda38f2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/30b90154effdbc0e0197396ec4b3f969916e7938"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/30b90154effdbc0e0197396ec4b3f969916e7938"}]},"branch":"refs/heads/master"},"cd13aefb67cf845e50efdc0c9f4706fca95987ca":{"kind":"REWORK","_number":2,"created":"2024-01-24 07:34:07.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/73/905773/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/73/905773/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/73/905773/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/73/905773/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/73/905773/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/73/905773/2"}}},"commit":{"parents":[{"commit":"40fdb7b44fb33e022b77ba8df63fde2565c70dea","subject":"Merge \"Change mysql socket path to a persistent path.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/40fdb7b44fb33e022b77ba8df63fde2565c70dea"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-01-23 13:49:35.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-01-24 07:33:40.000000000","tz":480},"subject":"Change postgresql socket path to a persistent path","message":"Change postgresql socket path to a persistent path\n\nThis patch changes the default socket patch on the host from\n\"/var/run/postgresql\" to \"/var/lib/postgresql-socket\".\nensuring that the  database container starts properly after\na Nova instance is restarted.\n\nStory: 2010599\nTask: 47471\n\nChange-Id: I54a53cc802bbd20bb9e2c6508919e4c8dda38f2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cd13aefb67cf845e50efdc0c9f4706fca95987ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cd13aefb67cf845e50efdc0c9f4706fca95987ca"}]},"branch":"refs/heads/master"},"d9ddede0ccf9fb794d2498b2b5b26e59544fa7e1":{"kind":"REWORK","_number":3,"created":"2024-01-25 07:20:00.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/73/905773/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/73/905773/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/73/905773/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/73/905773/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/73/905773/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/73/905773/3"}}},"commit":{"parents":[{"commit":"40fdb7b44fb33e022b77ba8df63fde2565c70dea","subject":"Merge \"Change mysql socket path to a persistent path.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/40fdb7b44fb33e022b77ba8df63fde2565c70dea"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-01-23 13:49:35.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-01-25 07:19:47.000000000","tz":480},"subject":"Change postgresql socket path to a persistent path","message":"Change postgresql socket path to a persistent path\n\nThis patch changes the default socket patch on the host from\n\"/var/run/postgresql\" to \"/var/lib/postgresql-socket\".\nensuring that the  database container starts properly after\na Nova instance is restarted.\n\nStory: 2010599\nTask: 47471\n\nChange-Id: I54a53cc802bbd20bb9e2c6508919e4c8dda38f2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d9ddede0ccf9fb794d2498b2b5b26e59544fa7e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d9ddede0ccf9fb794d2498b2b5b26e59544fa7e1"}]},"branch":"refs/heads/master"},"f151fd91d9f1a5e82b46a00bb7e3baae2b498161":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-01-30 03:08:55.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/73/905773/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/73/905773/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/73/905773/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/73/905773/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/73/905773/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/73/905773/4"}}},"commit":{"parents":[{"commit":"f9fa6a674ebbb1322cc10675a5c5b23f1eb1f0ef","subject":"bandit: Install baseline extra","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f9fa6a674ebbb1322cc10675a5c5b23f1eb1f0ef"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-01-23 13:49:35.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2024-01-30 03:08:55.000000000","tz":0},"subject":"Change postgresql socket path to a persistent path","message":"Change postgresql socket path to a persistent path\n\nThis patch changes the default socket patch on the host from\n\"/var/run/postgresql\" to \"/var/lib/postgresql-socket\".\nensuring that the  database container starts properly after\na Nova instance is restarted.\n\nStory: 2010599\nTask: 47471\n\nChange-Id: I54a53cc802bbd20bb9e2c6508919e4c8dda38f2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f151fd91d9f1a5e82b46a00bb7e3baae2b498161"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f151fd91d9f1a5e82b46a00bb7e3baae2b498161"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
