)]}'
{"id":"openstack%2Ftrove~905948","triplet_id":"openstack%2Ftrove~master~I7eb6ab017c48ad06b8607a71ad6a1ff2c67ca3c2","project":"openstack/trove","branch":"master","topic":"bump-eventlet","hashtags":[],"change_id":"I7eb6ab017c48ad06b8607a71ad6a1ff2c67ca3c2","subject":"bump eventlet to latest version that support python 3.12","status":"ABANDONED","created":"2024-01-17 15:34:00.000000000","updated":"2024-01-19 12:56:42.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"88b4a5a6c0d7031e022c4bf3ef8cdd1310b0dbb0","_number":905948,"virtual_id_number":905948,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-01-17 17:12:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-17 15:45:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-17 17:12:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ba3aea508758aeb47c79ed0e8db110925ac98ad8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-01-17 15:34:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8345fa70e6d500226bcd92282f71722fe6170e87","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 15:45:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/895b16471d2b4f0ba43b6ef545f59785\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/107fcb4cae1e451c8ce8a882b8d80a5c : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4ae78c1d02114401be7b0413cde2ccb5 : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8ed41095b2b84556b2cb6cc25134585c : SUCCESS in 5m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f6c5df56e25776894f7fc6a2cadeec38d46e62bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 17:12:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b87ae58abf743c49c83d2be6fc01b31\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6f5b5add781c4b0ebb8058295c29bff6 : SUCCESS in 3m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bad85c665d6f45048f4c5d6d5bfce102 : SUCCESS in 8m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d855b515808245bcb974ac62c45bd417 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ca1a70cf939481897a072ab4c0d6ef5 : SUCCESS in 6m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c91910a0fb1a4becb5511f74e07201bf : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c36c138b4d94dc98e61975afe4ac5c6 : SUCCESS in 5m 33s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/2497e222e5b44a4190b64c27b34ca343 : SUCCESS in 43m 21s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/37e9945723544571a75629663c4bfc3c : SUCCESS in 1h 36m 54s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/eef1616898dd4c778a650e63a5d29000 : SUCCESS in 1h 22m 38s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/22a23a4e96604a49b5b92fc699af47f7 : SUCCESS in 43m 34s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/0f6de91f7d6147d2a3a5ee7719f7c2a3 : SUCCESS in 1h 31m 09s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/a9dfcdfb8cf648688c62e8410702034b : SUCCESS in 1h 18m 01s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/ea4d8b54850f46f382dce09676303569 : SUCCESS in 45m 15s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/db8a4e23c42c4531971d3a81d7ee8866 : SUCCESS in 1h 30m 16s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/6d4cb713dcc841dc9dfbba5dd3683139 : SUCCESS in 1h 15m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"88b4a5a6c0d7031e022c4bf3ef8cdd1310b0dbb0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-01-19 12:56:42.000000000","message":"Abandoned\n\nThis version have been reverted from Upper Constraints. https://review.opendev.org/c/openstack/requirements/+/905905","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"13f88a8e76c789acace5e32829c696a7ed674fe4","revisions":{"13f88a8e76c789acace5e32829c696a7ed674fe4":{"kind":"REWORK","_number":1,"created":"2024-01-17 15:34:00.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/48/905948/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/48/905948/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/905948/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/905948/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/905948/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/48/905948/1"}}},"commit":{"parents":[{"commit":"422507ca425f5a1be10cfc979165e7b8d58dfffa","subject":"Merge \"Fix potential network conflict\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/422507ca425f5a1be10cfc979165e7b8d58dfffa"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2024-01-17 15:33:46.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2024-01-17 15:33:58.000000000","tz":60},"subject":"bump eventlet to latest version that support python 3.12","message":"bump eventlet to latest version that support python 3.12\n\nChange-Id: I7eb6ab017c48ad06b8607a71ad6a1ff2c67ca3c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/13f88a8e76c789acace5e32829c696a7ed674fe4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/13f88a8e76c789acace5e32829c696a7ed674fe4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
