)]}'
{"id":"openstack%2Ftrove~907266","triplet_id":"openstack%2Ftrove~stable%2F2023.1~I07ff03007e4e7247ad4fcf0dba16b87a02e3179d","project":"openstack/trove","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-02-22 16:22:35.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2024-02-22 13:18:41.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"I07ff03007e4e7247ad4fcf0dba16b87a02e3179d","subject":"stable-only: Pin bandit","status":"MERGED","created":"2024-01-31 11:11:50.000000000","updated":"2024-02-22 16:23:43.000000000","submitted":"2024-02-22 16:22:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"907266","cherry_pick_of_change":907163,"cherry_pick_of_patch_set":1,"meta_rev_id":"078e1aac4c36b5097a05067e999cc574b00770b5","_number":907266,"virtual_id_number":907266,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-22 16:22:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-02-22 13:18:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-02-22 13:18:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-31 11:30:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-31 14:02:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-31 14:21:28.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"71c25f3d81d5a263699b5e80b29c0be3dfc4c024","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-31 11:11:50.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.2.","accounts_in_message":[],"_revision_number":1},{"id":"3ed6c13cd826cbaf4169d585470c7a252698a417","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 11:30:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33d48beb94134ca6a8eda528fc590b06\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/120e5c08565b4774a36a939118bdb24b : SUCCESS in 11m 04s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/203fe96b574e43f7a6c7fc253fd5dbfb : SUCCESS in 15m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a69d9ba208572e998e3f82eacc87152fbc245194","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 14:02:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/446a8c9256764caaa997a5ffd913af90\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3c97e475102a4017b811c7994c0da205 : SUCCESS in 3m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39b56c04d9f44698b2f20c03012cc45e : FAILURE in 5m 01s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a250c52d63fa4ff6a93ca56cb55f20b1 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0804de11c41242fbbadbdcae5bc7ffce : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86dfe5b13bca471da90352ce27871c3b : SUCCESS in 5m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6744c729c0534baca21c8a2c8afeb399 : SUCCESS in 6m 15s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/17aa90efacab42b4a16a275934754fc5 : SUCCESS in 2h 27m 47s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/96bf11e6c1ec4e6785df990979dbb68e : FAILURE in 2h 22m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1454d5976440ee2cc3e614fb405151186d74d28e","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-02-22 13:18:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5090b9a847631972af3f200f28d071f1815cb9f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 13:19:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"14a09d94a8802a44f98c4efc0f2c3a51948fdef8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 16:22:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/450f1948b73748a1a0e7a8e49bfccb40\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/111f57c129a6448481bd94984224bf56 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f7e37f96dba49a39a847a9912757b0c : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3533a24289314c99802ea11beab58059 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c3600da654f4761b5adc1e59f943d85 : SUCCESS in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/032aa201694149f79fa1256de1356efa : SUCCESS in 5m 29s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/5ada40e6a6bd4638982dceb90397e077 : TIMED_OUT in 3h 03m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bbde2fed7223ce5097e57cd26ffe368daaf349ee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 16:22:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"078e1aac4c36b5097a05067e999cc574b00770b5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 16:23:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d13ba0929d8547728332e27e82b1c1b8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/892f85b7158446ca97ad5dd66daed201 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1894b397a4294470ae3d92344e60ce40 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"363c1df7c94defb21e59eb6796d353b88987a2d6","revisions":{"363c1df7c94defb21e59eb6796d353b88987a2d6":{"kind":"REWORK","_number":1,"created":"2024-01-31 11:11:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/66/907266/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/66/907266/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/907266/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/907266/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/907266/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/66/907266/1"}}},"commit":{"parents":[{"commit":"9dd308f5a09c1d0d5f9b079bb9082268c964d888","subject":"Fix guest agent failed to use swiftclient over TLS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9dd308f5a09c1d0d5f9b079bb9082268c964d888"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-30 02:13:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-31 11:11:50.000000000","tz":0},"subject":"stable-only: Pin bandit","message":"stable-only: Pin bandit\n\nSince bandit 1.7.7, we have to install the basiline extra to use\nbandit-baseline.\n\nThis was fixed in master, but for stable branches it may be better to\nuse the known good version instead, to avoid updating stable branches\nwhen any breaking change is made in bandit.\n\nChange-Id: I07ff03007e4e7247ad4fcf0dba16b87a02e3179d\n(cherry picked from commit 7a039ecb09769771dff45d767f129a3ea4b53669)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/363c1df7c94defb21e59eb6796d353b88987a2d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/363c1df7c94defb21e59eb6796d353b88987a2d6"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
