)]}'
{"id":"openstack%2Ftrove~907956","triplet_id":"openstack%2Ftrove~master~Ib06695c408061f5689c32e2e68fe64ccd3f7f756","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-03-16 12:18:25.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2026-03-16 11:13:51.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-09-08 12:41:53.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"Ib06695c408061f5689c32e2e68fe64ccd3f7f756","subject":"Remove remaining reference to removed guest agent drivers","status":"MERGED","created":"2024-02-06 07:29:01.000000000","updated":"2026-03-16 12:19:25.000000000","submitted":"2026-03-16 12:18:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"907956","meta_rev_id":"dda61b4c2463e5172be816ce8afca990423ed189","_number":907956,"virtual_id_number":907956,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-16 12:18:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-16 11:13:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":1,"date":"2025-09-08 12:42:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-16 11:13:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-06 07:35:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-06 09:27:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-08 12:41:53.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-09-08 12:42:01.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-09-08 12:42:08.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"b0c092d1dd26688a3618d8f0433a3d28b7dfa8b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-06 07:29:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b35d3752cf17848267e01ed2cbd400133594608","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 07:35:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b846999723a246f1833e7f84a58428ba\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9d2b9396013644f9bc8a8ac522d06f3f : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/50e218a949da42d1a524958ac98c938f : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/de4fc370b0824c8fadd4e12572f4ae97 : SUCCESS in 5m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"05fa63dcd7962a53a655b4c02a17be1a2ac6bd14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 09:27:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de95feb3c9a74deabe24bca4d15e73eb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b716ae510ce402ab6dd3ea8182632d8 : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/224093db67a7479c928bbf1c40373e70 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fdc82b6803d94073b9302325b7ef9986 : SUCCESS in 5m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/68c0c47b38d54cfe944a60cb1c8df8db : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4eccf96a8c59469ba4729cd9c28b562a : SUCCESS in 5m 18s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/3b3c04ff13af444f9b4efe8d3b6a9291 : SUCCESS in 1h 00m 12s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/afb64e9516aa48f8a735d9f52ea0e88e : FAILURE in 1h 56m 28s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/d25dc058b683465e9455123e06cf816d : SUCCESS in 1h 19m 11s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/4ad52d4907e14a2fb3a909db770e62dc : SUCCESS in 40m 27s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/e3b2244db0db4f3b84e12b0f45a524e1 : SUCCESS in 1h 24m 52s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/964dab99f7fb4e369ea9b21c72c0ab62 : SUCCESS in 1h 10m 51s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/b750561bac1f4e27bfee96eda522b57e : SUCCESS in 40m 07s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/72a018f8cf474b82a3e5f3cecf4ae97d : SUCCESS in 1h 23m 32s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/fb74c83bad14490293f5d38aacbeb27b : FAILURE in 1h 09m 55s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/060e668102204cfbba7790801499d2b0 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":1},{"id":"f3a408661b6503b2ee5e3ae1096c178ba83c8fcc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-08 12:38:08.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b3c0aa99d42f372ff987e91b865b31380221e40d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 12:38:26.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/trove for 907956,2","accounts_in_message":[],"_revision_number":2},{"id":"9bfa8fb90cea9141e2d390fb437977ddb93de806","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 12:38:39.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/trove for 907956,2","accounts_in_message":[],"_revision_number":2},{"id":"60b04428c7a0b9986595e31c9a18bc860866b98d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-08 12:38:48.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"546165638c84343fb646569537347bc6e19471a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-08 12:38:51.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"2edeee10fea26abfd49e1c8cd20f82de57e4e7c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-08 12:39:39.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"255f4426c0ec8ad9b341afe66306c3f9ce890ebf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-08 12:41:05.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"e4d888b8b6fe43dd377cd9ce71ac5b06024f8dfa","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-08 12:41:53.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3bf1f3a3a86ff728486c7d322f53bdf1b7dd7e14","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-08 12:42:01.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"397e247714663a21bf73dfd60cb1b8c023254f9e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 13:17:57.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce712df018b646bdbfc2630818fd5348\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/265c5c7da027456fb1347f1254d5f0d9 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cab4675812e04260b21a1b7f179e6020 : SUCCESS in 14m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e40668fe6c2b41d3a575e19be58d295a : SUCCESS in 20m 43s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d344b07b7459579451b1f267fe312e393f488b80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 14:25:40.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dccd4786310e4708971239bf9bba08e1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9448c27206d4c4ea203514ef60665ac : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0405cf57b4e64fe1ab3d22bd39281faa : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fc518abba98d48549d5f6706bd3b39a0 : SUCCESS in 4m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/496fc576f3be4003918e24706611eb0e : SUCCESS in 5m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e43ffb05c7e943afa682379ba5025ff7 : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46bbfe81cf74452f9afedf3d4d5a488d : SUCCESS in 5m 45s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/886a73483ab9427491104747b2be1de9 : SUCCESS in 51m 32s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/b37e14188fcf476eaa070c1dea20990c : SUCCESS in 1h 29m 47s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/d4d8709accfc4413b0adad9ccc237f28 : SUCCESS in 1h 22m 54s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/2de86b6c15d7466892359015bb5ef6ce : SUCCESS in 1h 43m 04s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/8c6af2b936ce457093e3083ee387044a : SUCCESS in 46m 10s (non-voting)\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/5ee01b3be58945598cc87a40718f98af : SUCCESS in 1h 25m 22s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/1ac357666a0f4ef199859ed5edd04b71 : SUCCESS in 1h 11m 42s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/592dd697762141eb8ea5854f2258754e : SUCCESS in 1h 23m 33s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/c947edff097648b085112102fdeceda2 : SUCCESS in 46m 37s (non-voting)\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/054e210cdcde4b329b579668b4ba81f1 : SUCCESS in 1h 24m 46s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/b0d17f9476774d38a34f3b64667ea49b : SUCCESS in 1h 07m 45s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/d8b956e630484249a4bc02948cc9ed82 : SUCCESS in 1h 23m 40s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d7c82011320f4484bc6deedf142b1c20 : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":6},{"id":"a31581da30e8bef5d085785f67e8525718416e73","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-03-16 11:13:51.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"323567acb2475c1ad22e808ea659fdeba3d5e6b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 11:14:13.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"87c7e1ad84454a85a6453563b9e7797aeeace56e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 12:18:25.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfa41116d23c4ba3b6139f373e034daa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a2d9bea329142fab90e5c71669868c6 : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/418578d4a4254b8b9aea1e08cb2c1c68 : SUCCESS in 3m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4710831ed4f94012be641adf7049bf8a : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f74da179fef45878dfbd22f4c65a0bc : SUCCESS in 6m 43s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/eb14d19e35ab4f7fba89746675e86ed4 : SUCCESS in 47m 04s\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/774383a5c79247bab035ae04577d2090 : SUCCESS in 46m 52s\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/9e1a561bd39b46579b4fa5648589a05d : SUCCESS in 1h 03m 45s","accounts_in_message":[],"_revision_number":6},{"id":"fa1c1c38fa290386945d665cb0ec65cdb4a72c20","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 12:18:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"dda61b4c2463e5172be816ce8afca990423ed189","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 12:19:25.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54f98a45405543d28a891721bc23e217\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2629c7a61d0641078b280fc6cc8bef27 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"60a9511b1d60080c7d6a9876e45aaa48a01e7645","revisions":{"5b9407e3d3e4a656cb523ccc7f6ce908e44f7b86":{"kind":"REWORK","_number":1,"created":"2024-02-06 07:29:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/56/907956/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/907956/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/907956/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/907956/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/907956/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/907956/1"}}},"commit":{"parents":[{"commit":"fd10a99377a6f3fd5a723b6a0d24d88fd35b00d5","subject":"Merge \"Fixes the DIB script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fd10a99377a6f3fd5a723b6a0d24d88fd35b00d5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-06 07:28:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-06 07:28:25.000000000","tz":540},"subject":"Remove remaining reference to removed guest agent drivers","message":"Remove remaining reference to removed guest agent drivers\n\nThese experimental drivers were removed by [1].\n\n[1] aa1d4d224674f44d9cd882eddb2537907adf5382\n\nChange-Id: Ib06695c408061f5689c32e2e68fe64ccd3f7f756\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5b9407e3d3e4a656cb523ccc7f6ce908e44f7b86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5b9407e3d3e4a656cb523ccc7f6ce908e44f7b86"}]},"branch":"refs/heads/master"},"c16545f46de357d46e30b9f4e77fa0ff62f818f2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-09-08 12:38:08.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/907956/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/907956/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/907956/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/907956/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/907956/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/907956/2"}}},"commit":{"parents":[{"commit":"fd10a99377a6f3fd5a723b6a0d24d88fd35b00d5","subject":"Merge \"Fixes the DIB script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fd10a99377a6f3fd5a723b6a0d24d88fd35b00d5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-06 07:28:25.000000000","tz":540},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-09-08 12:38:08.000000000","tz":0},"subject":"Remove remaining reference to removed guest agent drivers","message":"Remove remaining reference to removed guest agent drivers\n\nThese experimental drivers were removed by [1].\n\n[1] aa1d4d224674f44d9cd882eddb2537907adf5382\n\nChange-Id: Ib06695c408061f5689c32e2e68fe64ccd3f7f756\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c16545f46de357d46e30b9f4e77fa0ff62f818f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c16545f46de357d46e30b9f4e77fa0ff62f818f2"}]},"branch":"refs/heads/master","description":"Edit commit message"},"087fd577891491e476ac1306f61ee40ac813bafe":{"kind":"REWORK","_number":3,"created":"2025-09-08 12:38:48.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/907956/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/907956/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/907956/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/907956/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/907956/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/907956/3"}}},"commit":{"parents":[{"commit":"fd10a99377a6f3fd5a723b6a0d24d88fd35b00d5","subject":"Merge \"Fixes the DIB script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fd10a99377a6f3fd5a723b6a0d24d88fd35b00d5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-06 07:28:25.000000000","tz":540},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-09-08 12:38:45.000000000","tz":0},"subject":"Remove remaining reference to removed guest agent drivers","message":"Remove remaining reference to removed guest agent drivers\n\nThese experimental drivers were removed by [1].\n\n[1] aa1d4d224674f44d9cd882eddb2537907adf5382\n\nChange-Id: Ib06695c408061f5689c32e2e68fe64ccd3f7f756\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/087fd577891491e476ac1306f61ee40ac813bafe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/087fd577891491e476ac1306f61ee40ac813bafe"}]},"branch":"refs/heads/master"},"8d52328f568505034c2aecf5c17163d75648a35e":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-09-08 12:38:51.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/907956/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/907956/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/907956/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/907956/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/907956/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/907956/4"}}},"commit":{"parents":[{"commit":"6f978dafa70fcefb33e38bdde9f2bff27928b377","subject":"Merge \"Drop description about CentOS7 and RHEL7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6f978dafa70fcefb33e38bdde9f2bff27928b377"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-06 07:28:25.000000000","tz":540},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-09-08 12:38:51.000000000","tz":0},"subject":"Remove remaining reference to removed guest agent drivers","message":"Remove remaining reference to removed guest agent drivers\n\nThese experimental drivers were removed by [1].\n\n[1] aa1d4d224674f44d9cd882eddb2537907adf5382\n\nChange-Id: Ib06695c408061f5689c32e2e68fe64ccd3f7f756\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8d52328f568505034c2aecf5c17163d75648a35e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8d52328f568505034c2aecf5c17163d75648a35e"}]},"branch":"refs/heads/master","description":"Rebase"},"af1066f5ba0b13714500e3ce7b6852dd6f0b5af5":{"kind":"REWORK","_number":5,"created":"2025-09-08 12:39:39.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/907956/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/907956/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/907956/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/907956/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/907956/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/907956/5"}}},"commit":{"parents":[{"commit":"6f978dafa70fcefb33e38bdde9f2bff27928b377","subject":"Merge \"Drop description about CentOS7 and RHEL7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6f978dafa70fcefb33e38bdde9f2bff27928b377"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-06 07:28:25.000000000","tz":540},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-09-08 12:39:39.000000000","tz":0},"subject":"Remove remaining reference to removed guest agent drivers","message":"Remove remaining reference to removed guest agent drivers\n\nThese experimental drivers were removed by [1].\n\n[1] aa1d4d224674f44d9cd882eddb2537907adf5382\n\nChange-Id: Ib06695c408061f5689c32e2e68fe64ccd3f7f756\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/af1066f5ba0b13714500e3ce7b6852dd6f0b5af5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/af1066f5ba0b13714500e3ce7b6852dd6f0b5af5"}]},"branch":"refs/heads/master"},"60a9511b1d60080c7d6a9876e45aaa48a01e7645":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-09-08 12:41:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/56/907956/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/56/907956/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/907956/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/907956/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/56/907956/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/56/907956/6"}}},"commit":{"parents":[{"commit":"4beb14657745bac15c745e955936e6dbd8f42e63","subject":"Remove pylint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4beb14657745bac15c745e955936e6dbd8f42e63"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-06 07:28:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-08 12:40:46.000000000","tz":540},"subject":"Remove remaining reference to removed guest agent drivers","message":"Remove remaining reference to removed guest agent drivers\n\nThese experimental drivers were removed by [1].\n\n[1] aa1d4d224674f44d9cd882eddb2537907adf5382\n\nChange-Id: Ib06695c408061f5689c32e2e68fe64ccd3f7f756\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/60a9511b1d60080c7d6a9876e45aaa48a01e7645"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/60a9511b1d60080c7d6a9876e45aaa48a01e7645"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
