)]}'
{"id":"openstack%2Ftrove~909337","triplet_id":"openstack%2Ftrove~master~I8778a7decc6669b4d95d6886c971433e7c34c5c8","project":"openstack/trove","branch":"master","topic":"bump-hacking","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-04-02 07:33:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8778a7decc6669b4d95d6886c971433e7c34c5c8","subject":"Bump hacking","status":"MERGED","created":"2024-02-18 17:45:33.000000000","updated":"2024-04-02 07:42:33.000000000","submitted":"2024-04-02 07:33:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909337-bump-hacking","meta_rev_id":"70c68a102b4b407bcfcb8a51bc06d5b946b61122","_number":909337,"virtual_id_number":909337,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-02 07:33:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"all":[{"value":2,"date":"2024-04-02 03:11:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":1,"date":"2024-03-26 18:58:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-02 06:35:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-04-02 06:35:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-18 17:58:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-18 19:50:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-26 18:58:14.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2024-04-02 03:11:14.000000000","updated_by":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"REVIEWER"},{"updated":"2024-04-02 06:35:02.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"d7471d738c99ed2b795c075ab1488d1afbdee084","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-18 17:45:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48fe601efdc4c25cd0223ad8f6b76dab53e3167d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-18 17:58:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/890dc8f62ac14cd5ac71e291047e88e4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/450385fbc5a742329c1b697226c90f90 : FAILURE in 10m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d7788d2d8b7a4e4382fe5198302e115e : FAILURE in 4m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7ba256d50dfe423daa1c837f989cab39 : FAILURE in 8m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"90027537ec5add4d184985fbe7461c75127e3050","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-18 19:50:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8cab8531d83c4aa6b541ddea4b0ea168\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cbd908d23b074ce2a3fcb3e0f3048360 : FAILURE in 5m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bad26211b9224b4b8ab561cb34a8b03d : SUCCESS in 3m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3530acf13ba74b4f98aa086aa030a947 : FAILURE in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d59c46c66e3342a69174640aa9d3dc1b : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb5102d1beba4cce976a8f64e5cc43ab : FAILURE in 3m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6e18365547df4698a1fdc7881024019e : FAILURE in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1d0866eea434163b0fc368d68e5e9a9 : FAILURE in 4m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43d4947a5c3c4943a063502372c387b3 : SUCCESS in 6m 14s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/713d36fc7f914f1fa51195d05e17a6f5 : SUCCESS in 1h 02m 40s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/40c1c06673b044e59a8295ea0066604f : SUCCESS in 1h 54m 39s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/0eedc2b8a0694aeba3c26f5e67ff40fa : SUCCESS in 1h 53m 12s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/dc824414de7f4b7781e971da84930d74 : SUCCESS in 44m 13s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/d433701295c44897913d1c99b10d58ac : SUCCESS in 2h 02m 56s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/a87ca3e140ba4eca9d1ab44e403c5e55 : SUCCESS in 1h 39m 57s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/6b0537e38693498892e087e5f4f745e1 : SUCCESS in 52m 42s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/bbfdad04e24d40879d3d81df47ef83cf : SUCCESS in 1h 58m 41s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/34d27b2b7ce1465b940bef5eb2c4ef3c : SUCCESS in 1h 30m 20s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d341c258213940648a7e7943243a6e9f : FAILURE in 3m 54s","accounts_in_message":[],"_revision_number":1},{"id":"0f3f8db6d85644736a003994137100aaf054005a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-26 01:37:27.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e5ebc55d9d16f22cff9f342c21576cea331fb09b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 01:50:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab397fbc9e1a40bf90b6348677b7f358\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/77c1d1ec4b80444eb3373a352c2a5a7f : FAILURE in 4m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c792eaa6a13f4f7ea70354dbf26ababc : FAILURE in 11m 17s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/233008367ca54869a2658fac1a3f3ec1 : FAILURE in 4m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"432c1906ee631c6801dfe79b7afafe0435bc52ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 03:33:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/568f3575d9b7473db555eb41d4930157\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1932eb66de464521b69db65d6a95e460 : SUCCESS in 4m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/681af3141ea342649191e32e278a6f2f : SUCCESS in 3m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2eaa76e5ad884a688b1f8c1bb771aae1 : FAILURE in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c073e59cdbe43d0b87b4c74ac568c25 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cbecb799a31b448381e89d93f8c135a6 : FAILURE in 5m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8a1b76be87704095b003c363a854e27a : FAILURE in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f3323bd0ea44dc8b3bc1a956868bf85 : SUCCESS in 5m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f954365c8750448c9847d7e807b19d09 : SUCCESS in 4m 35s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/6568cceeed8e4fb8a69a8f3f8da0a16b : SUCCESS in 44m 52s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/b48a47f8d9974cdeb5905ff1d84dcb10 : SUCCESS in 1h 55m 11s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/e38609563daf43119060e794ce91de6e : SUCCESS in 1h 32m 46s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/2815cda08f094e0fa80b91f30163250c : SUCCESS in 44m 16s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/ccf8e47f64df427f834d460065fbfa32 : SUCCESS in 1h 42m 23s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/57d12cfb16a34580b36ee0a4c12f6e5e : SUCCESS in 1h 26m 26s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/cbd59fe8e5e74053b3822bd182757c6f : SUCCESS in 45m 33s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/4cf2d06daa3f4338844955bcd058e7e8 : SUCCESS in 1h 29m 57s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/1c217b52497740218fdce3b7670f23cc : SUCCESS in 1h 14m 41s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e5eb8ca362db4018bbc4ba30ec626b02 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":2},{"id":"6e2171928955e59dac2a840ed4f7c12fa1bf6d4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-20 09:56:19.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d61cd293c09db7cf078eab4f27e258b4e1a315cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 10:08:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb4395075c964b1faf9cb75890b60afa\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3f396e09290045ae8a769517ab9555e8 : FAILURE in 8m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cbed2a6d14c542f4970aaa10a4bb4bf2 : FAILURE in 4m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b40b9c6be4964fd59def7e7ba5142229 : FAILURE in 10m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"514b6fc3ea07189f3a10e10e44bf9632a91032d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 11:55:57.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b318b6c302704e63b569abed64c57d7a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0148592efc2c4816afae1d6f661fd888 : SUCCESS in 6m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6bfc79c8db194534a3ce13895a5c6d6a : SUCCESS in 3m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c885ea8cdf574097a025a52c1aacd1d6 : FAILURE in 8m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98c485d522fc46048517a4de18b407d9 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc7b29262fa94fdaa7ab4d1a79c5776b : FAILURE in 7m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/590e0d91fc3142ffafd3273759868d42 : FAILURE in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06086c9a60494dc7bc561be1e2ae30cf : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a832795bf6c642e3b07bee9f2db5fc83 : SUCCESS in 5m 14s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/afd9333e1ea04aaa8012f2e523457788 : SUCCESS in 44m 13s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/f4bfaa719bc942c98ffd320c50e6fbdc : SUCCESS in 1h 35m 55s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/fdba31991fb5469f944ce19ff86a10a2 : SUCCESS in 1h 45m 14s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/a46fe6f4134b411898b6408408ee6e9b : SUCCESS in 43m 14s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/89608093a29e4cc98a69cc6cd4af107e : SUCCESS in 1h 57m 16s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/5ba05da0ee5b4eae9e3e4739b605bdb6 : SUCCESS in 1h 12m 55s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/6cbe5c6302e34a2b9f20669a35c7e16b : SUCCESS in 58m 01s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/f949fcac82b64ce8aa34a406b17978ac : SUCCESS in 1h 25m 51s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/bfabd218bbcb49b7ae9624e1f860f0f3 : SUCCESS in 1h 35m 21s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7984d1d23a41485e8bf626ab7a8f3f0e : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":3},{"id":"cef15933d7d486c2a763367968746c2fa8ee3c1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-21 15:03:45.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"666ba69d683f285aae09c0a226b20c588f237391","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 15:18:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/121945987960418c984447079a3776be\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1f81d79041c4402ab37112b93b950e83 : SUCCESS in 10m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/74679daf1bcf48ca8edb97de409f74b4 : SUCCESS in 12m 27s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/25b997fce968433ea06125f63452192f : SUCCESS in 10m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b49df31aa303f3f4bb3fc9b0a20b91760c535fa7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 17:06:00.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0004969d43ef45808994229119d124d3\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/dfbdcf6aa8c7411f9a72d7f94a9a27cd : SUCCESS in 4m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ed4eb3dd5dfc4f85872a6b29e05cab7e : SUCCESS in 4m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47804e7d9fc04706a346b6b7cf5a3850 : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a882dac9518b466185f6801311aa8f9f : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/48d55db5cf9a4cdfa2c43472dbf0ea8a : SUCCESS in 6m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f58e2e793faa4610b671a8896512c325 : SUCCESS in 7m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a9ef37f4f1f94c1ebbc9339b469ce48d : FAILURE in 7m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c8c3004a3874038a761b9c50b4823e9 : SUCCESS in 5m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3310f315dbf944fdb620abcd0f0ec11d : SUCCESS in 7m 31s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/10c642741b7b411e9eb7edc6d3592907 : SUCCESS in 1h 02m 26s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/ebb94e6f44d84e04b880c0e3c39e5e1c : SUCCESS in 2h 00m 48s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/ab8177c7044944bb98d7f4376d4a1e8d : SUCCESS in 1h 44m 43s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/aa91efe9dce94ff08434b4e80aedbf81 : SUCCESS in 59m 19s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/46f1233fb5714d9b804c8a985c1eba70 : SUCCESS in 1h 40m 41s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/447b9a502f7a436c8906bb0570050949 : SUCCESS in 1h 37m 08s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/5c0c26b22ca44d9d99f6d97f571823af : SUCCESS in 1h 02m 57s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/a0d1b069df6648e4b90c222ce9a1cc04 : FAILURE in 1h 29m 02s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/bb62c585f41d40ccbbc156455a73d4bb : SUCCESS in 1h 26m 17s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/125a13d1095945ceb15d89119c0bebb8 : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":4},{"id":"4b7fa5bbd4097956a803fcb95c125cec21a238c3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-03-26 18:58:14.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ed3ceecf392faedf8ddcb3a9176267db2bfd92ac","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-04-02 03:11:14.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7f8d45fa1a87b08956ec817a21890dffebb65ffd","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-04-02 06:35:02.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9e31b9a981bb78aee1071ec5c2a1b9591256830a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 06:35:32.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"70ce34319ead71f01d0a6f896ebf66594b35cdfd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 07:33:35.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc4c8493a2cc48bb85479def7783860b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5e577eccd8924d63878f4242da00a7c1 : SUCCESS in 4m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5caa915c2bcf49a99e485e8b7ac71900 : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b568296f6c6540c6a92f9ad4d820913c : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d842d1dde6c64c1da76e851f047e8aed : SUCCESS in 5m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b22fa74928f944358b3d29d71a05d054 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea07db1f2c4842c49d5754b6484c894e : SUCCESS in 4m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ff7df5b7b8c492c88bba4b6ad1b74b1 : SUCCESS in 5m 50s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/df2d561294f545278759895c3883fdce : SUCCESS in 54m 16s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/2d0dc5a661014b73aacd9fa16095c258 : SUCCESS in 48m 37s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/72af7be8b8674de0b1f7f8f3ff5f527c : SUCCESS in 50m 39s","accounts_in_message":[],"_revision_number":4},{"id":"6d8d3e871ce92bb740ba360d7d8ae64e3541fb26","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 07:33:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"70c68a102b4b407bcfcb8a51bc06d5b946b61122","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 07:42:33.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/789eae83dc43411f9d791fadd8f39b85\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/386ff35cbdd040d1bc3c716ccd08d066 : SUCCESS in 37s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b2fb9fe9bb64bd9b3e977ebbd4a9ee5 : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ea8f49cd04a4e60a0f2b54364d1c0b4 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"bb260949d475bb0343a35fe3213408bc28a204a4","revisions":{"3cff2459690ae48d5d4f7a8ecff97912ac3fb1ea":{"kind":"REWORK","_number":1,"created":"2024-02-18 17:45:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/37/909337/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/37/909337/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/37/909337/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/37/909337/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/37/909337/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/37/909337/1"}}},"commit":{"parents":[{"commit":"fd10a99377a6f3fd5a723b6a0d24d88fd35b00d5","subject":"Merge \"Fixes the DIB script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fd10a99377a6f3fd5a723b6a0d24d88fd35b00d5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-18 17:41:07.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-18 17:41:07.000000000","tz":540},"subject":"Bump hacking","message":"Bump hacking\n\nhacking 3.1.0 is too old.\n\nChange-Id: I8778a7decc6669b4d95d6886c971433e7c34c5c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3cff2459690ae48d5d4f7a8ecff97912ac3fb1ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3cff2459690ae48d5d4f7a8ecff97912ac3fb1ea"}]},"branch":"refs/heads/master"},"a6eff2d25e2c695eb4f5705a7b6face1600d8901":{"kind":"REWORK","_number":2,"created":"2024-02-26 01:37:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/37/909337/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/37/909337/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/37/909337/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/37/909337/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/37/909337/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/37/909337/2"}}},"commit":{"parents":[{"commit":"fd10a99377a6f3fd5a723b6a0d24d88fd35b00d5","subject":"Merge \"Fixes the DIB script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fd10a99377a6f3fd5a723b6a0d24d88fd35b00d5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-18 17:41:07.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-26 01:37:04.000000000","tz":540},"subject":"Bump hacking","message":"Bump hacking\n\nhacking 3.1.0 is too old.\n\nNote:\nWe can\u0027t directly bump hacking to 6.x.0 (which is the latest major\nversion) because of the existing cap by diskimage-builder. The cap is\nnow being updated by [1].\n\n[1] https://review.opendev.org/c/openstack/diskimage-builder/+/909336\n\nChange-Id: I8778a7decc6669b4d95d6886c971433e7c34c5c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a6eff2d25e2c695eb4f5705a7b6face1600d8901"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a6eff2d25e2c695eb4f5705a7b6face1600d8901"}]},"branch":"refs/heads/master"},"0de8beab6fe367e2fae1edf64b97ffdf6dbda2f9":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-03-20 09:56:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/37/909337/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/37/909337/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/37/909337/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/37/909337/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/37/909337/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/37/909337/3"}}},"commit":{"parents":[{"commit":"6822a76ad9fad71be7875899a6a879439547f523","subject":"Merge \"reno: Update master for unmaintained/xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6822a76ad9fad71be7875899a6a879439547f523"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-18 17:41:07.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-20 09:56:19.000000000","tz":0},"subject":"Bump hacking","message":"Bump hacking\n\nhacking 3.1.0 is too old.\n\nNote:\nWe can\u0027t directly bump hacking to 6.x.0 (which is the latest major\nversion) because of the existing cap by diskimage-builder. The cap is\nnow being updated by [1].\n\n[1] https://review.opendev.org/c/openstack/diskimage-builder/+/909336\n\nChange-Id: I8778a7decc6669b4d95d6886c971433e7c34c5c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0de8beab6fe367e2fae1edf64b97ffdf6dbda2f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0de8beab6fe367e2fae1edf64b97ffdf6dbda2f9"}]},"branch":"refs/heads/master","description":"Rebase"},"bb260949d475bb0343a35fe3213408bc28a204a4":{"kind":"REWORK","_number":4,"created":"2024-03-21 15:03:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/37/909337/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/37/909337/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/37/909337/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/37/909337/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/37/909337/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/37/909337/4"}}},"commit":{"parents":[{"commit":"6822a76ad9fad71be7875899a6a879439547f523","subject":"Merge \"reno: Update master for unmaintained/xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6822a76ad9fad71be7875899a6a879439547f523"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-18 17:41:07.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-21 15:03:05.000000000","tz":540},"subject":"Bump hacking","message":"Bump hacking\n\nhacking 3.1.0 is too old.\n\nNote:\nWe can\u0027t directly bump hacking to 6.x.0 (which is the latest major\nversion) because of the existing cap by diskimage-builder. The cap is\nnow being updated by [1].\n\n[1] https://review.opendev.org/c/openstack/diskimage-builder/+/909336\n\nChange-Id: I8778a7decc6669b4d95d6886c971433e7c34c5c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bb260949d475bb0343a35fe3213408bc28a204a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bb260949d475bb0343a35fe3213408bc28a204a4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
