)]}'
{"id":"openstack%2Ftrove~910118","triplet_id":"openstack%2Ftrove~master~I6ab051fd1c8034d9697b1caab9c1fda64a3275c3","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-02-26 02:29:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6ab051fd1c8034d9697b1caab9c1fda64a3275c3","subject":"Fix missing oslo.db options","status":"MERGED","created":"2024-02-25 16:44:32.000000000","updated":"2024-02-26 02:31:26.000000000","submitted":"2024-02-26 02:29:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"910118","meta_rev_id":"d9b9ef75aa42e7d8becfe3439ecd85214f297d32","_number":910118,"virtual_id_number":910118,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-26 02:29:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-02-26 01:45:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-02-26 01:45:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-25 17:07:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-25 19:01:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-26 01:45:40.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"fc793eacd18734d8440e7bbee5615b231e6de7a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-25 16:44:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"919f80d42867ac04096169bc9265f5f52e593023","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-25 16:45:17.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"236ce7839fec55d4f6354ffdad9e19876f6040d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-25 17:07:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a42543abf054465ac443ce63d761af5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ac6f02b13a38410ea9b917ce54ad87f2 : SUCCESS in 20m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7f29b7d6d99a4fc5aa7723b0384145a1 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/055825a9aa25431a9f12519518120969 : SUCCESS in 19m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4f28d5908d0c2a5e563eff9be24a5e971bf3833a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-25 19:01:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a10207430c3147ac991341604a24ee66\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28d5bb6241ad484eab1f990f874752c8 : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c46b004c847a4d18ba7d62a4179c26da : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b7f463e684f4bf7b2926c3041c385d9 : SUCCESS in 5m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9493758433064a638dfb0781f43bcca4 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc86b54d362d490b97e6b6c1619f6870 : SUCCESS in 5m 35s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/44cd1945bf9848a19891d43fb2bd6e8a : SUCCESS in 58m 02s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/00844b1a907f4bc8b47e34a92596723c : SUCCESS in 2h 13m 35s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/77408afb774e4a83a3dee2f35b4e4b44 : SUCCESS in 1h 48m 14s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/223554ff6be142c2b4c586b7628cb865 : SUCCESS in 1h 00m 04s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/d1e345c4eed1426bbddf6c547642fe4b : SUCCESS in 2h 01m 08s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/83f1d2222ead48858fe28950b62bb61e : SUCCESS in 1h 46m 08s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/fcaffb55c7c04362abb8ec9296b58d1a : SUCCESS in 47m 00s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/4edcd23c9ddd4056a61c165d181dcbe2 : SUCCESS in 1h 29m 09s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/52a93b8d6ea546f39c685da3703e568f : SUCCESS in 1h 15m 50s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4f1b66527cb84dec904f3af9c3d74bbd : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":2},{"id":"b6f84916b7a4d731de32503aed42466896c0d46d","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-02-26 01:45:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8030300c990da3e8e34734895eb64cb530cd6b27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 01:46:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"715eb72c22eb82c8498cd292a7a38135a9bc11ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 02:29:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8c72833388343b687d8f3f1125104de\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0fbd530eb7f4b65afc82c6bbb52e3f8 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3c6243f633f42ecb83390d73c3a493b : SUCCESS in 5m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e6d953e91bfa42c4a377396fcefbe608 : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f723f7e1eabb47228343ba9eb6d32240 : SUCCESS in 5m 11s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/5e678ee07cdc4e7283ba8a89eda655a8 : SUCCESS in 42m 28s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/671212f3900b4f0a806632184bc5225a : SUCCESS in 39m 45s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/2cabdbba13e34b7e964c30a7d2675b06 : SUCCESS in 40m 50s","accounts_in_message":[],"_revision_number":2},{"id":"590977307ce62c8cf2b67c58a5d84167573eba12","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 02:29:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d9b9ef75aa42e7d8becfe3439ecd85214f297d32","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 02:31:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84839f0242cd468ab665339192481fce\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed6d15d36977445e868476175ff3eb38 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1298374998efaa87a490e524d93e50661bba6885","revisions":{"051d1c7840e6056e6d91aa3a2cdda54b832857b6":{"kind":"REWORK","_number":1,"created":"2024-02-25 16:44:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/18/910118/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/18/910118/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/18/910118/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/18/910118/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/18/910118/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/18/910118/1"}}},"commit":{"parents":[{"commit":"fd10a99377a6f3fd5a723b6a0d24d88fd35b00d5","subject":"Merge \"Fixes the DIB script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fd10a99377a6f3fd5a723b6a0d24d88fd35b00d5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-25 16:43:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-25 16:44:22.000000000","tz":540},"subject":"Fix missing oslo.db options","message":"Fix missing oslo.db options\n\nThis ensures that the options from the oslo.db library are picked up\nby oslo-config-generator.\n\nChange-Id: I6ab051fd1c8034d9697b1caab9c1fda64a3275c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/051d1c7840e6056e6d91aa3a2cdda54b832857b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/051d1c7840e6056e6d91aa3a2cdda54b832857b6"}]},"branch":"refs/heads/master"},"1298374998efaa87a490e524d93e50661bba6885":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-02-25 16:45:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/18/910118/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/18/910118/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/18/910118/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/18/910118/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/18/910118/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/18/910118/2"}}},"commit":{"parents":[{"commit":"fd10a99377a6f3fd5a723b6a0d24d88fd35b00d5","subject":"Merge \"Fixes the DIB script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fd10a99377a6f3fd5a723b6a0d24d88fd35b00d5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-25 16:43:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-25 16:45:01.000000000","tz":540},"subject":"Fix missing oslo.db options","message":"Fix missing oslo.db options\n\nThis ensures that the options from the oslo.db library are picked up\nby oslo-config-generator.\n\nAlso, update .gitignore so that the sample config file is ignored by\ngit.\n\nChange-Id: I6ab051fd1c8034d9697b1caab9c1fda64a3275c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1298374998efaa87a490e524d93e50661bba6885"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1298374998efaa87a490e524d93e50661bba6885"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
