)]}'
{"id":"openstack%2Ftrove~911388","triplet_id":"openstack%2Ftrove~master~I0bf90212ec9c8b1fa36ada39192b9fc2a345a436","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"28691":{"account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"last_update":"2024-03-20 09:56:47.000000000","reason":"Change was abandoned"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-03-20 09:56:47.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I0bf90212ec9c8b1fa36ada39192b9fc2a345a436","subject":"wip: Fix compabitility with oslo.db 15.0.0","status":"ABANDONED","created":"2024-03-06 01:47:29.000000000","updated":"2024-03-20 09:56:47.000000000","total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"848a51de7f68517cf6e51b103d6325357059d9ab","_number":911388,"virtual_id_number":911388,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-03-06 03:34:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-06 01:59:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-06 02:45:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-06 13:55:33.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"CC"},{"updated":"2024-03-16 01:35:36.000000000","updated_by":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"REVIEWER"}],"messages":[{"id":"310f9b41f78b9add8113cffe6d962467ddaaae12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-06 01:47:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7bf30a544a6b6a31d32fceeb306e66d6c6006451","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 01:59:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e24248467b2422788e8e18253de7f7d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/93947260ac8947428da41165f710e6a1 : FAILURE in 4m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/31e585314fc04c4fa56fbd91092addac : FAILURE in 10m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ab498f03dd534b3d914fd680dd674a5d : FAILURE in 4m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d14b964a658dbb0bbcbe4b824623f22ad7fa6fd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 02:45:04.000000000","message":"Patch Set 1: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b53883995adb4e91942a8e67ab0d5347\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/688bbad9a51c412d9046d2242d121681 : FAILURE in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bac6fa47a0141458408c092dc1bfcea : FAILURE in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b25bc2058f34aefa61336b3d360ee74 : FAILURE in 5m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/624388b8233f42b4aed54cbcf4696c57 : FAILURE in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1fc08b262294e66860a66732226ba30 : SUCCESS in 5m 43s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/bf51602d56fe4e1e8bb4b971d50c0e78 : FAILURE in 31m 57s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/860b42bb16324fe886c25d3da0c834cd : FAILURE in 26m 39s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/bee68d50352e473e80bfd7b6282aedb0 : FAILURE in 26m 27s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/fca117a29b714c2d8f848a2372074735 : FAILURE in 25m 31s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/958704ceffee44d5b2e74a701ef261f1 : FAILURE in 28m 10s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/a02292e170cb4108a50f10ad14401a9b : FAILURE in 31m 14s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/1fa964def5fa400b80986cbd9228aaa9 : FAILURE in 32m 48s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/ecf69200137e41c18d9e9c9aca9722e1 : FAILURE in 35m 51s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/e4e452bf3f2b41fcb6ae68dcd6d1d8bd : FAILURE in 37m 13s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ba3a5e21282a43368629b9a567cf3e4b : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":1},{"id":"97fc50efa7472634e03942e783cfc5480bc08a43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-06 02:49:54.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"04957799d24bd7c037859f99e48e5810aabd967f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 02:57:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e803af3c802d4a009cc80c06214a9fff\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5eafffc6abc4425ca24acbb9d4726f23 : FAILURE in 6m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/18ab343c77b146388ed3ab8dd70e5a2a : FAILURE in 6m 04s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2803a614d1b149ff9acbf520d5a6ec8e : FAILURE in 5m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e9092f8353705a7d453e0b4f7eea9618d403ae17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-06 02:59:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"19e226ba213b0e56059e31f9a06e881227d44514","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-06 03:02:59.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5213c410ddcfcc2c8c0e7a2658e6d47a4a6e9a3d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 03:10:01.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c6cdd7a371246488744c9bc666f65e8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/98aa2b7cbfc544dbafe281a0216b2a3a : FAILURE in 5m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9cb2c53713f442df90cb322c021c336e : FAILURE in 5m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fa90338d51374957a85d0b0e0c41f6bc : FAILURE in 5m 23s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"eca99e7908e06edeadcb7de6d2b89e5f87b4aa8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 03:34:35.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/11ac17a28f804738b1d66211ac41d635\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b979087d793a42ac83ab1ee835b4c953 : FAILURE in 7m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c64e9f5a9924703bbabe9641a60d474 : FAILURE in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/640f199fda83458eb6776e11a2e04482 : FAILURE in 5m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/941ae6e5f33946fc97c4b01028a6f5d2 : FAILURE in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecc92d34fbe44507b33987b746967785 : SUCCESS in 6m 04s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/c2b51d94c31949f38dce4309bd06f755 : FAILURE in 25m 05s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/5721b4f076e949acb5bb3d339099f5bd : FAILURE in 25m 06s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/36c96f493afc480fb4c62db487c5d900 : FAILURE in 29m 33s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/8f1c089a822c410c8efeb3973e13f63d : FAILURE in 29m 39s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/d87f9e909e7a4ec3b81546a35473942c : FAILURE in 30m 28s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/6327974fbaf34410bb7b3148596f2ccb : FAILURE in 29m 39s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/22057b9ebe544a7d87b374912df5a610 : FAILURE in 29m 49s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/452e7886db59496691c9cd60619c15d3 : FAILURE in 25m 20s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/cc69aaec4f584385b3fb906bbcfd6edf : FAILURE in 29m 58s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fa4b05d2cb2b454f89769a035a95efec : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":4},{"id":"f355d9c76997b03d0e33389b58094cf429821e7d","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-03-06 13:55:33.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"efd148cce6aef575a43f5a644a5792881e2beea0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-06 14:58:44.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d60bf207707d00e37c442b0c9102b24193c7cd92","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-03-06 15:22:12.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"848a51de7f68517cf6e51b103d6325357059d9ab","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-20 09:56:47.000000000","message":"Abandoned\n\nthis was fixed by https://review.opendev.org/c/openstack/trove/+/912054","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5417bddae4670b607c0f154dd192ddc64b4147d7","revisions":{"b1f72ab800fcd89f4e6e7843232b547314601d75":{"kind":"REWORK","_number":1,"created":"2024-03-06 01:47:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/88/911388/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/88/911388/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/911388/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/911388/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/911388/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/88/911388/1"}}},"commit":{"parents":[{"commit":"fd10a99377a6f3fd5a723b6a0d24d88fd35b00d5","subject":"Merge \"Fixes the DIB script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fd10a99377a6f3fd5a723b6a0d24d88fd35b00d5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-06 01:46:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-06 01:46:35.000000000","tz":540},"subject":"wip: db: Use oslo_db.sqlalchemy.enginefacade","message":"wip: db: Use oslo_db.sqlalchemy.enginefacade\n\nChange-Id: I0bf90212ec9c8b1fa36ada39192b9fc2a345a436\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b1f72ab800fcd89f4e6e7843232b547314601d75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b1f72ab800fcd89f4e6e7843232b547314601d75"}]},"branch":"refs/heads/master"},"188c8e10e965e5abbb5ebf064e9fb8a49694c670":{"kind":"REWORK","_number":2,"created":"2024-03-06 02:49:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/88/911388/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/88/911388/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/911388/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/911388/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/911388/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/88/911388/2"}}},"commit":{"parents":[{"commit":"fd10a99377a6f3fd5a723b6a0d24d88fd35b00d5","subject":"Merge \"Fixes the DIB script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fd10a99377a6f3fd5a723b6a0d24d88fd35b00d5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-06 02:49:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-06 02:49:26.000000000","tz":540},"subject":"wip: Fix compabitility with oslo.db 15.0.0","message":"wip: Fix compabitility with oslo.db 15.0.0\n\nChange-Id: I0bf90212ec9c8b1fa36ada39192b9fc2a345a436\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/188c8e10e965e5abbb5ebf064e9fb8a49694c670"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/188c8e10e965e5abbb5ebf064e9fb8a49694c670"}]},"branch":"refs/heads/master"},"483495eab24a918c1cfe1a1df1030abb324be634":{"kind":"REWORK","_number":3,"created":"2024-03-06 02:59:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/88/911388/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/88/911388/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/911388/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/911388/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/911388/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/88/911388/3"}}},"commit":{"parents":[{"commit":"fd10a99377a6f3fd5a723b6a0d24d88fd35b00d5","subject":"Merge \"Fixes the DIB script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fd10a99377a6f3fd5a723b6a0d24d88fd35b00d5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-06 02:49:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-06 02:59:14.000000000","tz":540},"subject":"wip: Fix compabitility with oslo.db 15.0.0","message":"wip: Fix compabitility with oslo.db 15.0.0\n\nChange-Id: I0bf90212ec9c8b1fa36ada39192b9fc2a345a436\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/483495eab24a918c1cfe1a1df1030abb324be634"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/483495eab24a918c1cfe1a1df1030abb324be634"}]},"branch":"refs/heads/master"},"5417bddae4670b607c0f154dd192ddc64b4147d7":{"kind":"REWORK","_number":4,"created":"2024-03-06 03:02:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/88/911388/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/88/911388/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/911388/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/911388/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/911388/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/88/911388/4"}}},"commit":{"parents":[{"commit":"fd10a99377a6f3fd5a723b6a0d24d88fd35b00d5","subject":"Merge \"Fixes the DIB script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fd10a99377a6f3fd5a723b6a0d24d88fd35b00d5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-06 02:49:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-06 03:02:48.000000000","tz":540},"subject":"wip: Fix compabitility with oslo.db 15.0.0","message":"wip: Fix compabitility with oslo.db 15.0.0\n\nChange-Id: I0bf90212ec9c8b1fa36ada39192b9fc2a345a436\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5417bddae4670b607c0f154dd192ddc64b4147d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5417bddae4670b607c0f154dd192ddc64b4147d7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
