)]}'
{"id":"openstack%2Ftrove~912054","triplet_id":"openstack%2Ftrove~master~I3f0d384f353069f1eb186bbfa2aa3d8afa969e05","project":"openstack/trove","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-03-20 06:18:33.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2024-03-20 01:30:23.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"I3f0d384f353069f1eb186bbfa2aa3d8afa969e05","subject":"db: Remove use of autocommit","status":"MERGED","created":"2024-03-13 14:42:30.000000000","updated":"2024-03-20 06:19:36.000000000","submitted":"2024-03-20 06:18:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"912054-sqlalchemy-20","meta_rev_id":"d89f9c05db92ff0412bd1d5a09d0c1d9eb3caf08","_number":912054,"virtual_id_number":912054,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-20 06:18:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-03-20 05:29:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-03-20 05:29:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-13 14:48:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-13 15:19:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-14 05:29:02.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2024-03-20 01:30:46.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"CC"}],"messages":[{"id":"b48ae30c35a3f1ba74b63882ddd58b4f6a56af9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-03-13 14:42:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3936b92b0fa17cc83f9d9057ac91de1bbb2534c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 14:48:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6978c857093349ecb00a905aad03bfa4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d40fb40506a140348b26eda46306324f : FAILURE in 4m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2c3590aa5f0d4d94bb2658efac0e13b9 : FAILURE in 5m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7585013662d343f39fc75af8115bc0e0 : FAILURE in 4m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4b61a5bf2eb6623e134f9cf8b71ac9cc8a303c3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 15:19:28.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e1c430367fb74584a9a47a7600d1890a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe897398ab1e47a6ab82db4b2adb5c6d : FAILURE in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73f43a86d9894c4383838e1b6f4b413a : FAILURE in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41518856d47b41198d6a22b68abb904d : FAILURE in 5m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6761e3b39e7941dcb5db3ab45d0408e9 : FAILURE in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/143d7c97c93c46ff9f5bfb5bce795593 : SUCCESS in 6m 40s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/2f5bc459691a4b71876a3c9df8fae49a : FAILURE in 30m 00s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/0c9a83f1a1a640bc9ba2c237ea2fdf2b : FAILURE in 34m 31s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/de8159ea53244d8c97f5ab432dd8cdba : FAILURE in 34m 57s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/85d7600da5d14c4da65210e5e768b92a : FAILURE in 33m 31s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/5626cc83d3f3415c92a86d65820e9820 : FAILURE in 25m 15s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/2d89d7c592244b7bb93aa98b62756f53 : FAILURE in 25m 08s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/42e33f4ec02c44a8ab469d18414d8280 : FAILURE in 28m 34s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/970619e7af8f48de87340fc0c07fdae5 : FAILURE in 25m 37s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/0df9fba1eb924a2fb15c22a23c389890 : FAILURE in 32m 56s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cac2a65e579443ca9dce27aaacc47901 : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":1},{"id":"07006d1f1327ccf88474bcb7ecc68f00176f0dbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-03-14 05:23:23.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"5fc233bc5e906d8408044b041a64bb91e21fed29","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-14 05:29:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"27128e2e2fa8fbe2d979ba82ff7db4b2efc4f110","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 05:34:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fc404b4a1ae48fd8da530b65e3be0eb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bf4f7aca0180490e9db6d0784dfb21a0 : FAILURE in 5m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/25c03f66210249bbb06640b4484bc674 : FAILURE in 5m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0dd42af75f1548eab07e51aa92107aa4 : FAILURE in 9m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6e05455aa4c02f265076899908ea26791b5aac2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 05:54:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/690ef2d139344021ae4806579bd122fb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd2e481156fd486281327b1382b8bcee : FAILURE in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2dbcfc960af4a96b998df78d955c0f2 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e774cc17e5149258e5919312ed5d7e0 : FAILURE in 5m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/776faf6a4d0649108bdd73090391250c : FAILURE in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8e7878367fc4a97a6f28fbc974aa53e : SUCCESS in 4m 57s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/8527fd39d0f746c9800604b5192acfbf : FAILURE in 23m 34s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/dcc5dc6c76fd4ab8ac657ef849abc1f5 : FAILURE in 26m 13s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/f6173eb25a064a94bff7de94503ebec3 : FAILURE in 27m 05s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/bfa69a3712c14a629f3410dc183a595b : FAILURE in 26m 46s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/07b6ffec47d5413dbf00ce81df4e7d3c : FAILURE in 26m 59s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/da973d37feb44f249f1fdb01e943c2de : FAILURE in 28m 01s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/28b30d3bd9194c66bef0db17449e073a : FAILURE in 26m 07s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/173481e17bfa47a99c8996b2eef8fa69 : FAILURE in 26m 49s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/415206e6685b4465b5bf94c7067b6bcd : FAILURE in 24m 29s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e0dab5de60334f1697e0068812d321ff : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":2},{"id":"f10174f3eb02604268298e54d8eb5c4ceeb6c7f3","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-03-14 07:21:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"22159daf170755bbdb81114eee6a388c9cec23d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-03-14 14:48:21.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"1af8b4d57b38885ce9cd2258ca6d53cdd5bbaefe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 14:48:44.000000000","message":"Patch Set 3:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/trove on branch master.  The\nproblem was:\n\n  Unknown projects: openstack/tripleo-image-elements\n\nThe problem appears in the the \"trove-devstack-base\" job stanza:\n\n  job:\n      name: trove-devstack-base\n      parent: devstack\n      abstract: true\n      timeout: 10800\n      required-projects:\n        - openstack/tempest\n        - openstack/diskimage-builder\n        - openstack/tripleo-image-elements\n  ...\n\n  in \"openstack/trove/zuul.d/deprecated_jobs.yaml@master\", line 1","accounts_in_message":[],"_revision_number":3},{"id":"9a1838765105986c454d39c12d69e890d5eb296b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 14:48:53.000000000","message":"Patch Set 3: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/trove on branch master.  The\nproblem was:\n\n  Unknown projects: openstack/tripleo-image-elements\n\nThe problem appears in the the \"trove-devstack-base\" job stanza:\n\n  job:\n      name: trove-devstack-base\n      parent: devstack\n      abstract: true\n      timeout: 10800\n      required-projects:\n        - openstack/tempest\n        - openstack/diskimage-builder\n        - openstack/tripleo-image-elements\n  ...\n\n  in \"openstack/trove/zuul.d/deprecated_jobs.yaml@master\", line 1","accounts_in_message":[],"_revision_number":3},{"id":"6d28eb774a5831c3f1608d4d7e094e7d73a5c829","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-03-14 15:47:43.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"97054e9a423853b85497e5721c8be8aebbc1c9b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 15:48:07.000000000","message":"Patch Set 4:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/trove on branch master.  The\nproblem was:\n\n  Unknown projects: openstack/tripleo-image-elements\n\nThe problem appears in the the \"publish-trove-guest-image\" job stanza:\n\n  job:\n      name: publish-trove-guest-image\n      parent: publish-openstack-artifacts\n      run: playbooks/image-build/run.yaml\n      post-run: playbooks/image-build/post.yaml\n      required-projects:\n        - openstack/diskimage-builder\n        - openstack/trove\n        - openstack/tripleo-image-elements\n  ...\n\n  in \"openstack/trove/zuul.d/jobs.yaml@master\", line 425","accounts_in_message":[],"_revision_number":4},{"id":"557cdb45b1cd0105c6922cb94c6d33d38a10f389","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 15:48:17.000000000","message":"Patch Set 4: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/trove on branch master.  The\nproblem was:\n\n  Unknown projects: openstack/tripleo-image-elements\n\nThe problem appears in the the \"publish-trove-guest-image\" job stanza:\n\n  job:\n      name: publish-trove-guest-image\n      parent: publish-openstack-artifacts\n      run: playbooks/image-build/run.yaml\n      post-run: playbooks/image-build/post.yaml\n      required-projects:\n        - openstack/diskimage-builder\n        - openstack/trove\n        - openstack/tripleo-image-elements\n  ...\n\n  in \"openstack/trove/zuul.d/jobs.yaml@master\", line 425","accounts_in_message":[],"_revision_number":4},{"id":"c46574e47894aa85e6554a782e7da28d0d7d2efd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-03-14 15:50:45.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"17a3ccd5efddcfab0707ae53d7ca1be46a597e9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 16:31:26.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5d0c9e846a884df996da5c34305824f9\n\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/33fd46f64b5c417699ec7b67be9ca20d : FAILURE in 37m 37s","accounts_in_message":[],"_revision_number":5},{"id":"6aac634ca8d54cbbe2357fd147185f49f687a31d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-03-15 01:46:48.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"a47b2510fb7b8c600738f53ac04076be18f9c96c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 02:31:08.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55803f211ab24939bd9723cb07dcc66f\n\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/fc9ef3bfc3644bdcb09cec577054ab5f : SUCCESS in 43m 09s","accounts_in_message":[],"_revision_number":6},{"id":"8ed7395d60c220d9bf71cf6d0fe96fea5581bd25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-03-19 14:04:20.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"643188507c20e35c4b7c6e693e4de855c828f49a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-19 14:11:45.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed46ee01c3b84acf9ee0c6a08c3edef5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/adb840bd0aa4457192e04b8c53b29401 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3bb1816519c1422795932d61feeaab7b : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1e4d2a266b2f4764b4434437a8d5643d : SUCCESS in 4m 54s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"4c5f259e05441f8967d466c6d321d5f47aa33be3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-19 16:36:36.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7923476d41043bdb5b88ea739d3211f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f660c36a7624b3d897b4928abf5e8f3 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/beefd8ece95c4959b50cea1f36c0217e : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11b0347bd0704d7ca2411399b767ab26 : SUCCESS in 6m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9a498b463fd340f6ae015d0b2d2e55a0 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79c80715847e4ef3ae25ad14717aff3e : SUCCESS in 4m 56s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/a0bc8e6d1a014a7f9b5d1023120dddb1 : SUCCESS in 59m 23s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/8d7cf4807f5c4725904e7326192ab244 : SUCCESS in 2h 30m 31s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/707b9bc170eb4feebac4bb232d0e0171 : SUCCESS in 1h 57m 34s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/47afb44293d14834abbada838ba88eb0 : SUCCESS in 37m 05s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/16568330fddb4764b79f5166112ed346 : SUCCESS in 1h 25m 00s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/17d9f01b6e544cdbad63d2a3dd71e337 : SUCCESS in 1h 17m 33s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/e5379f8a63784e27bbbedb262ff151af : SUCCESS in 42m 10s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/0dfda06df714405dac821f40cf6d86e9 : SUCCESS in 1h 31m 44s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/712419b625e64901889af4d85e69d661 : SUCCESS in 1h 16m 27s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d7f84955f614496286a8eca86467226e : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":7},{"id":"26d5141bf2de242d4cee78c959e068df70c778a5","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-03-20 01:30:23.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f780e1196fcebd5af61ebaac07de3c122bd7b490","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-03-20 05:29:16.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d00ae85d4d2ad153e8ecbfa0cb4fcb03b24bd52e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 05:29:45.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"163eceafe24777528a1de78e725c6d0f992a8e15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 06:18:32.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ce632739e6447afbfa6dede423d10c1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5dff10aea7c4cbb91103ca51cfb7948 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08ebd33f790f40ad934f1866c0f9fcb8 : SUCCESS in 5m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/517d80856f3f43809e0baa848d953214 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf6f5679e3a34541ae1e520d73b86cd0 : SUCCESS in 4m 54s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/7cb027d0b8de46688438a28c32c92fbd : SUCCESS in 48m 19s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/d9fc6a4362a643e7aa419530382fcca1 : SUCCESS in 45m 07s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/9f17bf4d896643a9ae3b8dcefb3d87e5 : SUCCESS in 48m 08s","accounts_in_message":[],"_revision_number":7},{"id":"7cb7f8bd514c00617d565722a4166d2615baf6c1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 06:18:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"d89f9c05db92ff0412bd1d5a09d0c1d9eb3caf08","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 06:19:36.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/396073364b19415bb21ef58dd1537436\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4016d75f4fa04e568230727b450c6d0a : SUCCESS in 45s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"ab83ad7c8567dc33664f9bbcd521beda225a181a","revisions":{"a4b4013e3774422c92132cb1239361aed804be57":{"kind":"REWORK","_number":1,"created":"2024-03-13 14:42:30.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/54/912054/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/54/912054/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/912054/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/912054/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/912054/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/54/912054/1"}}},"commit":{"parents":[{"commit":"860772e1b0007aa08d422aa5464f6b5e6c59ac8f","subject":"Merge \"Enable healthcheck middleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/860772e1b0007aa08d422aa5464f6b5e6c59ac8f"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-03-13 14:42:00.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-03-13 14:42:00.000000000","tz":480},"subject":"Remove use of autocommit","message":"Remove use of autocommit\n\nChange-Id: I3f0d384f353069f1eb186bbfa2aa3d8afa969e05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a4b4013e3774422c92132cb1239361aed804be57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a4b4013e3774422c92132cb1239361aed804be57"}]},"branch":"refs/heads/master"},"e8b0f57bc0548be73cb2f6c711adf56e344b3584":{"kind":"REWORK","_number":2,"created":"2024-03-14 05:23:23.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/54/912054/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/54/912054/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/912054/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/912054/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/912054/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/54/912054/2"}}},"commit":{"parents":[{"commit":"860772e1b0007aa08d422aa5464f6b5e6c59ac8f","subject":"Merge \"Enable healthcheck middleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/860772e1b0007aa08d422aa5464f6b5e6c59ac8f"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-03-13 14:42:00.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-03-14 05:23:13.000000000","tz":480},"subject":"Remove use of autocommit","message":"Remove use of autocommit\n\nChange-Id: I3f0d384f353069f1eb186bbfa2aa3d8afa969e05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e8b0f57bc0548be73cb2f6c711adf56e344b3584"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e8b0f57bc0548be73cb2f6c711adf56e344b3584"}]},"branch":"refs/heads/master"},"bec4e6092b17ae7b843fc23ccb5e8488ee8211e0":{"kind":"REWORK","_number":3,"created":"2024-03-14 14:48:21.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/54/912054/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/54/912054/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/912054/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/912054/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/912054/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/54/912054/3"}}},"commit":{"parents":[{"commit":"860772e1b0007aa08d422aa5464f6b5e6c59ac8f","subject":"Merge \"Enable healthcheck middleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/860772e1b0007aa08d422aa5464f6b5e6c59ac8f"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-03-13 14:42:00.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-03-14 14:48:13.000000000","tz":480},"subject":"Remove use of autocommit","message":"Remove use of autocommit\n\nChange-Id: I3f0d384f353069f1eb186bbfa2aa3d8afa969e05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bec4e6092b17ae7b843fc23ccb5e8488ee8211e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bec4e6092b17ae7b843fc23ccb5e8488ee8211e0"}]},"branch":"refs/heads/master"},"af8b4e3cea351e98d29bbbc9afd80c926263d8fb":{"kind":"REWORK","_number":4,"created":"2024-03-14 15:47:43.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/54/912054/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/54/912054/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/912054/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/912054/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/912054/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/54/912054/4"}}},"commit":{"parents":[{"commit":"860772e1b0007aa08d422aa5464f6b5e6c59ac8f","subject":"Merge \"Enable healthcheck middleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/860772e1b0007aa08d422aa5464f6b5e6c59ac8f"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-03-13 14:42:00.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-03-14 15:47:32.000000000","tz":480},"subject":"Remove use of autocommit","message":"Remove use of autocommit\n\nChange-Id: I3f0d384f353069f1eb186bbfa2aa3d8afa969e05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/af8b4e3cea351e98d29bbbc9afd80c926263d8fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/af8b4e3cea351e98d29bbbc9afd80c926263d8fb"}]},"branch":"refs/heads/master"},"091e587c356556146c2de843b6a439cfd983ddcf":{"kind":"REWORK","_number":5,"created":"2024-03-14 15:50:45.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/54/912054/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/54/912054/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/912054/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/912054/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/912054/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/54/912054/5"}}},"commit":{"parents":[{"commit":"860772e1b0007aa08d422aa5464f6b5e6c59ac8f","subject":"Merge \"Enable healthcheck middleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/860772e1b0007aa08d422aa5464f6b5e6c59ac8f"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-03-13 14:42:00.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-03-14 15:50:38.000000000","tz":480},"subject":"Remove use of autocommit","message":"Remove use of autocommit\n\nChange-Id: I3f0d384f353069f1eb186bbfa2aa3d8afa969e05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/091e587c356556146c2de843b6a439cfd983ddcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/091e587c356556146c2de843b6a439cfd983ddcf"}]},"branch":"refs/heads/master"},"2f7f855a97521b092f0b74bef522adb35bcaf1c0":{"kind":"REWORK","_number":6,"created":"2024-03-15 01:46:48.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/54/912054/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/54/912054/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/912054/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/912054/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/912054/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/54/912054/6"}}},"commit":{"parents":[{"commit":"860772e1b0007aa08d422aa5464f6b5e6c59ac8f","subject":"Merge \"Enable healthcheck middleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/860772e1b0007aa08d422aa5464f6b5e6c59ac8f"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-03-13 14:42:00.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-03-15 01:46:40.000000000","tz":480},"subject":"Remove use of autocommit","message":"Remove use of autocommit\n\nChange-Id: I3f0d384f353069f1eb186bbfa2aa3d8afa969e05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2f7f855a97521b092f0b74bef522adb35bcaf1c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2f7f855a97521b092f0b74bef522adb35bcaf1c0"}]},"branch":"refs/heads/master"},"ab83ad7c8567dc33664f9bbcd521beda225a181a":{"kind":"REWORK","_number":7,"created":"2024-03-19 14:04:20.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/54/912054/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/54/912054/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/912054/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/912054/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/912054/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/54/912054/7"}}},"commit":{"parents":[{"commit":"860772e1b0007aa08d422aa5464f6b5e6c59ac8f","subject":"Merge \"Enable healthcheck middleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/860772e1b0007aa08d422aa5464f6b5e6c59ac8f"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-03-13 14:42:00.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-03-19 14:03:56.000000000","tz":480},"subject":"db: Remove use of autocommit","message":"db: Remove use of autocommit\n\nChange-Id: I3f0d384f353069f1eb186bbfa2aa3d8afa969e05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ab83ad7c8567dc33664f9bbcd521beda225a181a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ab83ad7c8567dc33664f9bbcd521beda225a181a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
