)]}'
{"id":"openstack%2Ftrove~912342","triplet_id":"openstack%2Ftrove~master~I275bd77736020f02678fc34332edd277f5e2a6b6","project":"openstack/trove","branch":"master","topic":"fix_error_detach_cfg","attention_set":{},"removed_from_attention_set":{"28691":{"account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"last_update":"2024-04-02 04:23:19.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2024-04-02 02:38:41.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"I275bd77736020f02678fc34332edd277f5e2a6b6","subject":"Fix error can\u0027t detach the configuration group binding","status":"MERGED","created":"2024-03-10 13:08:01.000000000","updated":"2024-04-02 04:24:47.000000000","submitted":"2024-04-02 04:23:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"912342-fix_error_detach_cfg","meta_rev_id":"400f20ec87ffb371da455e2286debe38166627d8","_number":912342,"virtual_id_number":912342,"owner":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-02 04:23:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-02 03:20:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-04-02 03:20:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-10 13:15:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-10 14:09:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-02 02:30:09.000000000","updated_by":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"0170956eadd6e31925ae3629e61d6407c682a7ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-03-10 13:08:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46328fad7771b9912afd57c2d626572cbefb140b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-10 13:15:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/991d5b7063f940df81e066905135ba35\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1d56252d61d34a5f889ff150a513bfcc : FAILURE in 6m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/db07a6ca41414c18824be2527da835a4 : FAILURE in 5m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/68ba9b4f66de42e8b4679ffcc439151c : FAILURE in 5m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"989789b24de4a64540aa3015921060387643ed64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-03-10 13:15:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c5cfacbaa3970364292f1ab038fa9bba8f0e26e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-10 13:29:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ead2f0da47414b2f822ce4699861c224\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ecac62d96f7a421c851c20fb92d88eb8 : FAILURE in 5m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2e5f1d6cfbcc48449f767ba08b71d594 : FAILURE in 6m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fe3a9e67604a4a3f950b3c4166c2b838 : FAILURE in 11m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"36901af5be5a2ba84ecd5a0ac71dc1f31b0b777b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-10 14:09:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d0abf4f58e854dd2a8c803928c1f8c37\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/548d76a5174d4aac8d03a4c6d28ce814 : FAILURE in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28b197f78e5540e5853fa577a2f1747b : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb77f5ce18434426b2c8d79254b9b268 : FAILURE in 5m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f23140d95a074ed1afd21248a8db902e : FAILURE in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/249025ef0df34a55a3e594b1344e2686 : SUCCESS in 4m 55s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/e00395a352584b12b27661713bb4868d : FAILURE in 31m 53s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/6f5d3f2fec9442bb909c808ea1d1495c : FAILURE in 33m 40s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/e33b3b750b2b48adba01e56b2d683ace : FAILURE in 28m 35s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/a8e7b61c9e6a4fd1965947c626931271 : FAILURE in 31m 06s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/af36612014e34ae5bfdcc325285cfab7 : FAILURE in 28m 19s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/d225f9d5c967447f80126ef2a33a3cfd : FAILURE in 33m 22s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/d12e8279065948aea449c3b985fb8fea : FAILURE in 24m 14s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/df137b72040545e299205d76c4718e36 : FAILURE in 25m 47s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/4b81ed26fe8f46e292532110ff69feb2 : FAILURE in 32m 58s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/34bdbe1b02d140849d8b00748441c5ff : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":2},{"id":"ad580a0083d22a1f2034c32c9f71212569e95611","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-03-11 03:11:51.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"2ae50da7efb86159200ddfb4d1698418f771b55b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-11 03:18:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba5c974208524281961d707d80a5fdb7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/11fa540912104679acae88f6fa7b15e3 : FAILURE in 4m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/94ff0845a9514149af604f869f762c75 : FAILURE in 5m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/929125a2e18c488882b2638a258e1b91 : FAILURE in 4m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"210f6b595c8aecee5259cd54160bd876da11d7a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-11 03:52:41.000000000","message":"Patch Set 3: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3fc6b2e569324f84a1f99518600a2be0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df8acf0fc01d4a64b5b98e35ddb8cfcf : FAILURE in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6431e9e19e8047f7b73a52b193a31034 : FAILURE in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f7f2af11b6347638f45b7b350129a49 : FAILURE in 5m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/10a33821c3cb42078abc807bb8946a68 : FAILURE in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed901985e1db49eeaf397b8ffd560431 : SUCCESS in 5m 45s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/dbf9ad5c09504572a8298c2874557d8e : FAILURE in 34m 30s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/4e3374fa178645b3b1dca811bc946d7f : FAILURE in 37m 11s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/d4421eb2cf624f24a82f45340f6af1b5 : FAILURE in 35m 52s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/f0925ba57c454ddca18df08f3bccd358 : FAILURE in 34m 15s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/954bee6a4473419a8dfddc3fde16b062 : FAILURE in 37m 56s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/1600653824794dd48ce446704419b50e : FAILURE in 36m 59s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/d761c163a6624994bb282cb385b72d3f : FAILURE in 36m 20s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/b165eaaa5b224739aab61d5d54781e38 : FAILURE in 25m 07s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/3fe4caae29db49818097e91bd1f6ecb4 : FAILURE in 38m 02s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ba02a655cc9c43dc9c44076ad884ba37 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":3},{"id":"a2e50a63e8588161a2b756df4e18914e2fab0008","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"real_author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-03-20 08:23:31.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_28691\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}],"_revision_number":4},{"id":"599ba8201689ffd46bee3cacc413009441cd3ca9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 08:30:05.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc0657dfb46d4fd49b603a8109005529\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e9d5a27a437a409484de26e7ab77469a : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/970d5703d55d4a8388e5dc137a5a9f48 : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6f0c8357927b4fb0bb16ee8485208329 : SUCCESS in 4m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8dac8744518badfe29a648781897c2856c8fd22c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 10:57:44.000000000","message":"Patch Set 4: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7471dc62283a48b795211bf1b6280d06\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34af1cda3f024b9f9a616cd6c912ee53 : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6903f5aa431452cb60c885a661d62c5 : FAILURE in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64c4b9e4f3e846dc969f6c42a8535c6b : SUCCESS in 5m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/88c0a9bab59b46d28b00cba82d148031 : SUCCESS in 7m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0233a88bdf464bdaa7eb1b798813fc95 : SUCCESS in 4m 59s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/af0dbb48b47149a4bdda879af0792ec9 : SUCCESS in 1h 05m 48s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/ec9f4a7a23a340e097cd183e2c1faf37 : FAILURE in 1h 47m 12s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/eb29f78ad9174dc2813a631d559e5230 : SUCCESS in 2h 11m 18s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/85343ec8d5c7410aae882a2fa7985dcd : SUCCESS in 44m 44s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/fa9428b5d8f44a32ae1ab1852ca6a573 : SUCCESS in 2h 27m 34s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/558683dc09b24ccca5239159b5f2ab1e : SUCCESS in 1h 40m 01s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/5be7df3c940a41d99cf4513f523b611e : SUCCESS in 1h 03m 59s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/fabd156bf2eb4bed90870854f43ec589 : FAILURE in 1h 39m 08s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/ee4842f1083c4e43ae0137d7ab46792e : SUCCESS in 2h 16m 34s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c2143927cc554f68a016cf6e5346c571 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":4},{"id":"6c63bb1945b08dcba37779a48923accdf4f4aac4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-03-25 04:43:15.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"f801a53c4e7974b3b90be480f4b144bb0b2c4129","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-25 04:51:02.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1339e3fa96149858f7f95f6443663e5\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/82dd04af51314f739cb522fc986caa26 : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b0f38e66630144aab5cb17d7425767eb : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/09a164ccc0ec42539097188b466cf8c7 : SUCCESS in 4m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9cd6f3e628951c92b1805819bd43968ff4db1028","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-25 07:10:46.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e089863aa5b4435bcae3f34091e4fc5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b809938dfbb4b04aad360bcb7c40483 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee1008a10de04ca49df03b8b128ce207 : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac7e0f5d336d40249bf73cc5e454088b : SUCCESS in 5m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6f5651f3c2e541638eb77f9a2351f73e : SUCCESS in 7m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/887a4db1ce7b4454bb5422838c349c2c : FAILURE in 8m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60870896e17446fe9bce780a51a5c57b : SUCCESS in 6m 48s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/42b3d3b403ca4c46a0ea4aa1580e83cc : SUCCESS in 1h 08m 17s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/af594e75730e4e63836b2ee6b49e92e9 : SUCCESS in 2h 19m 21s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/d27befa0c4324b5dbf5c47b28479a203 : SUCCESS in 1h 50m 44s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/87ccce91dc164f0b803cdb75d45a03fb : SUCCESS in 48m 02s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b6e8311fa17c49439923319868f71c0d : SUCCESS in 1h 44m 09s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/3ec0d937cb6243c6b196ead052eb8679 : SUCCESS in 1h 42m 17s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/90cab8ec370e4df5bf270d4dc2a621ca : SUCCESS in 54m 01s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/d09f50b456664b11bcc91fbdc37abda9 : SUCCESS in 1h 49m 44s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/c9057dc8bd884f28aae52e6ac37d03a2 : SUCCESS in 1h 28m 59s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/da1b9d46e6324e5d86b7e1f4935299a0 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":5},{"id":"2acbab088489741ab72849d69ca54253ed0651bc","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-04-02 02:30:09.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e4161665890a5797ea4963199844e817abb08b91","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-04-02 02:38:41.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"01d1bcf371384592f3ba9edcfdba050aee2494f9","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-04-02 03:20:25.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b8c2312b6ef8f10ba715a1d1dcce234e4b4c88b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 03:23:52.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"6e466521d1125422f3922a178b487e3b3600c0a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 04:23:19.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71dac51713214c8f8de4488413d0d1bf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e577773b44f475fb238ee41c67ed8ad : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3e8ed23c25341b49c794f3414d0ef80 : SUCCESS in 5m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/09430a4e29234521bf65f97c12afc7d1 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15d3e35b2cbf4d1fa8473c8c38a0fc98 : SUCCESS in 5m 08s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/93b3ca3972174f97a2817e5e3725f6e0 : SUCCESS in 49m 52s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/37444eaad6e5446f84224d5a7ed7d710 : SUCCESS in 55m 52s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/3e9e6e4d58c848cbb472167159eeffad : SUCCESS in 57m 52s","accounts_in_message":[],"_revision_number":5},{"id":"0a62d87fc0faf7cff542fa688c9ba23e5a858609","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 04:23:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"400f20ec87ffb371da455e2286debe38166627d8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 04:24:47.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d444d43174648e08649115da08f8bd6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24cd3a939daa45a5a958ef7202b84eb8 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"dbb18bea1022e6c1f6bb5da721c3e791aa1f69b8","revisions":{"645300cf82a1349387c373356ae88d2513e49b6e":{"kind":"REWORK","_number":1,"created":"2024-03-10 13:08:01.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/42/912342/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/42/912342/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/42/912342/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/42/912342/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/42/912342/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/42/912342/1"}}},"commit":{"parents":[{"commit":"860772e1b0007aa08d422aa5464f6b5e6c59ac8f","subject":"Merge \"Enable healthcheck middleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/860772e1b0007aa08d422aa5464f6b5e6c59ac8f"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2024-03-10 13:06:07.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2024-03-10 13:06:07.000000000","tz":420},"subject":"Fix error can\u0027t detach the configuration group binding","message":"Fix error can\u0027t detach the configuration group binding\n\nStory: #2011063\nTask: #49697\nChange-Id: I275bd77736020f02678fc34332edd277f5e2a6b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/645300cf82a1349387c373356ae88d2513e49b6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/645300cf82a1349387c373356ae88d2513e49b6e"}]},"branch":"refs/heads/master"},"53915cc093107f5718b718e985b5c6ff6a674000":{"kind":"REWORK","_number":2,"created":"2024-03-10 13:15:55.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/42/912342/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/42/912342/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/42/912342/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/42/912342/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/42/912342/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/42/912342/2"}}},"commit":{"parents":[{"commit":"860772e1b0007aa08d422aa5464f6b5e6c59ac8f","subject":"Merge \"Enable healthcheck middleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/860772e1b0007aa08d422aa5464f6b5e6c59ac8f"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2024-03-10 13:06:07.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2024-03-10 13:15:43.000000000","tz":420},"subject":"Fix error can\u0027t detach the configuration group binding","message":"Fix error can\u0027t detach the configuration group binding\n\nStory: #2011063\nTask: #49697\nChange-Id: I275bd77736020f02678fc34332edd277f5e2a6b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/53915cc093107f5718b718e985b5c6ff6a674000"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/53915cc093107f5718b718e985b5c6ff6a674000"}]},"branch":"refs/heads/master"},"8f4296ecad9b3d587372fdadf423a3cd4f57b3c0":{"kind":"REWORK","_number":3,"created":"2024-03-11 03:11:51.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/42/912342/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/42/912342/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/42/912342/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/42/912342/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/42/912342/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/42/912342/3"}}},"commit":{"parents":[{"commit":"860772e1b0007aa08d422aa5464f6b5e6c59ac8f","subject":"Merge \"Enable healthcheck middleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/860772e1b0007aa08d422aa5464f6b5e6c59ac8f"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2024-03-10 13:06:07.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2024-03-11 03:11:39.000000000","tz":420},"subject":"Fix error can\u0027t detach the configuration group binding","message":"Fix error can\u0027t detach the configuration group binding\n\nStory: #2011063\nTask: #49697\nChange-Id: I275bd77736020f02678fc34332edd277f5e2a6b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8f4296ecad9b3d587372fdadf423a3cd4f57b3c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8f4296ecad9b3d587372fdadf423a3cd4f57b3c0"}]},"branch":"refs/heads/master"},"b0813a7b618040852d12072e347bd0b68a8f17d1":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-03-20 08:23:31.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"real_uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/42/912342/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/42/912342/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/42/912342/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/42/912342/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/42/912342/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/42/912342/4"}}},"commit":{"parents":[{"commit":"6822a76ad9fad71be7875899a6a879439547f523","subject":"Merge \"reno: Update master for unmaintained/xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6822a76ad9fad71be7875899a6a879439547f523"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2024-03-10 13:06:07.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2024-03-20 08:23:31.000000000","tz":0},"subject":"Fix error can\u0027t detach the configuration group binding","message":"Fix error can\u0027t detach the configuration group binding\n\nStory: #2011063\nTask: #49697\nChange-Id: I275bd77736020f02678fc34332edd277f5e2a6b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b0813a7b618040852d12072e347bd0b68a8f17d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b0813a7b618040852d12072e347bd0b68a8f17d1"}]},"branch":"refs/heads/master","description":"Rebase"},"dbb18bea1022e6c1f6bb5da721c3e791aa1f69b8":{"kind":"REWORK","_number":5,"created":"2024-03-25 04:43:15.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/42/912342/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/42/912342/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/42/912342/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/42/912342/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/42/912342/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/42/912342/5"}}},"commit":{"parents":[{"commit":"860772e1b0007aa08d422aa5464f6b5e6c59ac8f","subject":"Merge \"Enable healthcheck middleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/860772e1b0007aa08d422aa5464f6b5e6c59ac8f"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2024-03-10 13:06:07.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2024-03-25 04:43:04.000000000","tz":420},"subject":"Fix error can\u0027t detach the configuration group binding","message":"Fix error can\u0027t detach the configuration group binding\n\nStory: #2011063\nTask: #49697\nChange-Id: I275bd77736020f02678fc34332edd277f5e2a6b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dbb18bea1022e6c1f6bb5da721c3e791aa1f69b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dbb18bea1022e6c1f6bb5da721c3e791aa1f69b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
